)]}'
{"id":"openstack%2Fswift~954379","triplet_id":"openstack%2Fswift~master~I8a66c97f12a279bbed55d2a631f2498aaebe8429","project":"openstack/swift","branch":"master","attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-09-23 17:27:55.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-08-07 03:26:03.000000000","reason":"\u003cGERRIT_ACCOUNT_30073\u003e replied on the change","reason_account":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}}},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-07-30 18:31:28.000000000","reason":"Change was marked work in progress"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-07-30 18:31:28.000000000","reason":"Change was marked work in progress"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-09-24 19:07:31.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I8a66c97f12a279bbed55d2a631f2498aaebe8429","subject":"add minimal trace to get_hashes","status":"NEW","created":"2025-07-08 23:02:07.000000000","updated":"2025-09-24 19:34:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":37,"unresolved_comment_count":11,"has_review_started":true,"meta_rev_id":"180c86aad40b195a35b7431e4752f9db4c64a1ec","_number":954379,"virtual_id_number":954379,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-24 19:18:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":1,"date":"2025-09-24 19:07:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-09 00:10:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-09 06:07:53.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2025-07-09 18:51:22.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2025-07-14 01:46:40.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2025-07-28 11:51:02.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-07-30 18:45:20.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"},{"updated":"2025-08-07 03:26:03.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"CC"}],"messages":[{"id":"eff71d46e69a3446b42f66f7d9af6dea4ebbc40f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-08 23:02:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53b0fb93cb2112ca0e5a079400760bad50907c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 00:10:22.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/813ac076bc91451fadd3c84c7d89b876\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72cd69b8f2274e72a34fccf8e4d1eeb3 : SUCCESS in 6m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/fc3b6f4406e54c32b7880da961c485ba : SUCCESS in 58m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d0b15327ee494745a296636800e673ee : SUCCESS in 1h 02m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1c449034b5504417a6273a4fda42a90e : SUCCESS in 36m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f630671a96d943348bfafc2f0a73158b : SUCCESS in 32m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85e8081182e3452e9fd891e7426571e0 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38434acc6e2149beadb6df5b1240cc8a : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88bfa46d9095400ca42206c3dd61c194 : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1fc6612a6e67457e9edd60beec4d1216 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8cb7102472204487b78e093d367ce842 : SUCCESS in 14m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ade44c6b54f94fa9bd446eaebd4de83c : SUCCESS in 13m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1be66f9dfff5480d92a307dcb8847625 : SUCCESS in 13m 27s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/32457cb56a2045f6847ad5a73ea57d76 : SUCCESS in 10m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0698b3684ccd4cc385696881e3fcee8d : SUCCESS in 7m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8705e1aaf71840c48c635a0c0077b3ed : SUCCESS in 14m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7bd34e90a47a422da820697534524dc5 : SUCCESS in 11m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e1d03618fe3e47a1acfd8f0150e0e65b : SUCCESS in 5m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e2d43dca2ee404d80f9f24196ed360b : SUCCESS in 7m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8963da9107e64b0688b56d9cc8eee24c : SUCCESS in 6m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4396e5f734f6406baf4a76aa1942e11f : SUCCESS in 40m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f74a3b3bc214e1fa180348fc33b3bf5 : SUCCESS in 53m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aa5939d61dfe4d43a8ab9f212f766683 : SUCCESS in 55m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/952b1460aeee42349551ef2aef107d37 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bed93cd7430840dab9521d6d4dd59c69 : FAILURE in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5442adc6c0a844039e2bb4dcdd54afa8 : SUCCESS in 17m 53s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61dbc64473c94bc59941a84b403b8f03 : SUCCESS in 47m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4379da08c5ab42e1af071991fbfc8851 : SUCCESS in 42m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d56e055374a392ca8a1668b482aa6d23132659b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 00:40:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/510354fc968643718d295e4337e1fa5b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1958c94f86304e18b0f59f8a9f00474a : SUCCESS in 1h 36m 20s","accounts_in_message":[],"_revision_number":1},{"id":"de0c59b4e3365aa1e6c3fb62ebfa0b39538cdadf","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-09 06:07:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fecf8326f6429e9ce000202e24e422e457124c1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-09 18:01:50.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3c2524da4693b818c4d21d3b8839e3495834d213","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-09 18:51:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b5cb5775dc9d29ed337894e39f55b37e33790d51","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-09 19:23:35.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9ad88e2be948992275abb81b725ac0d3e501f377","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-11 23:33:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"03fb6fa1e506a3dc88823a104fd20607ae3bd838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 00:42:29.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ef4813d17e0446cb8dc8ae817fc5f1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/419e656b2eb64bd8907b2fc773c3ad7c : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/b6e722d9dd054a3c87d26e8fad70d521 : SUCCESS in 58m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dddec1e5dc7d419d83aab2691e9b97c3 : SUCCESS in 58m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ebfca515ef64427f9a1a201956d54b84 : SUCCESS in 58m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98e8348e76224704835895c506579ff0 : SUCCESS in 1h 03m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47721bed65a849ee8c1f6823786d2cc9 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bb44fa6f16b4a3aa884565949da2e6e : SUCCESS in 21m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f71d9f069a1472ebf2fce090312b4fc : SUCCESS in 20m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d8344bb4b5734c71a5187b9191ac12ba : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e23aabae88fb4e879897166d64cbe6dd : SUCCESS in 20m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dc523afa461a431d871b70dbee5aef97 : SUCCESS in 19m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cb233957b56e4336a6c402ef2887d247 : SUCCESS in 26m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c1137cfe560443abb78ce7d5a2397c05 : SUCCESS in 29m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/017428d6b73747cb8cceaea8aefc972e : SUCCESS in 7m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ff2f2b0745af4cf18c0bcaf20d22959d : SUCCESS in 18m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3ba2e0fce36c452fa6b4911257982cd7 : SUCCESS in 16m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3f985422f07c4f1d8101206d49b63296 : SUCCESS in 3m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c7168656cfbe462c973ece7cfbb14cd9 : SUCCESS in 15m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5848ea98f77c4e859a431bf6aa1131ce : SUCCESS in 7m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f405f7778a24002a19a73ad0404a97a : SUCCESS in 58m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d65e648bd1d48e99a4061dd0af3ec8e : SUCCESS in 35m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b8484bf4dc5d4d8096426f1a7506f08c : SUCCESS in 54m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ac6cc960e7f4fa9b39eef58953233f4 : SUCCESS in 16m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38c2371e1330443fafebeb1daebce012 : FAILURE in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5f20cf159ad442a5a2d9d13781d3f21a : SUCCESS in 15m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/896d963bb9bd4d9e83733f0bdbd96bd1 : SUCCESS in 41m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/734caca0e8674fbc8b018d018d9326e5 : SUCCESS in 30m 27s","accounts_in_message":[],"_revision_number":2},{"id":"ff216df57b1c3b09f2a4d6e32dba4987413076d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 01:03:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7781584ceaec4bae833e17ce866678bc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5558705119964af4908e181939fb3158 : SUCCESS in 1h 27m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ec1636d3cfac675a7622dd08488d7bb0d3d163ea","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-07-14 01:46:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc722a771976560af283c10f65d56b299605aacc","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-07-14 07:28:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6b32b8a23178813993a924c71380c56ec77dd51","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-14 16:48:07.000000000","message":"Patch Set 2: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7951596d28c7640fcab975d49d18796fe60f4099","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-14 17:53:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2907d3c21edb1ff3c45741a39b99c71a08d156c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-14 20:21:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7f350d1893b2e8cad29fa6cfc77438121a4525ca","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-14 20:27:55.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b73f634b3a7c81e55262bccb66d77d2708a3b8c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-14 20:35:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5e225b78a97004508611fd3e330972b575e93f7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-14 20:58:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"825438cfcd55b6a983f47ab42dd2ea8a1f4dd968","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 21:51:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cdae0cc0ae84751836d997ccf8fbe5c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/01b5cdf9732f48feb8a65e7d73cb1ab5 : SUCCESS in 50m 30s","accounts_in_message":[],"_revision_number":5},{"id":"a2d942e033153ace1585749b2b2f4ad1bca666d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 22:30:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/decc35c0ba794f46bdfcd6ee370a0ca4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f812112e0c74f71b150e2647ffa5d90 : SUCCESS in 5m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/8f73d7da680b4f8399a61aa752911a1e : SUCCESS in 1h 04m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92ce4c84a42c4fc48e92eb42c2563ce8 : SUCCESS in 44m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/27bf009ac4f4458f8a1ff990f65ddb6b : SUCCESS in 1h 00m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70549623fd8a445981ec234678cb2550 : SUCCESS in 1h 08m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce628d36dcf74c0d8c55dfdafcd1453c : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a22da37150c1472ab7b748277e174e14 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1ef77e39ae248bb9824590321e985f4 : SUCCESS in 20m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3e4d4fba91047599fa7ab1036173759 : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/270cec7346884b2fb9a747e682587cb1 : SUCCESS in 21m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/53eed07aa51b4c9680c2f1eed595b3d6 : SUCCESS in 19m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/91f643cfa2f54ba6b77a7d8b993f3476 : SUCCESS in 12m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b6df03b36c2c4b998e25616878bf2a03 : SUCCESS in 32m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0a627e908a434e48b4cd63d62d4004e8 : SUCCESS in 8m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c5dad2d100494ce1bd04d2a7e5957cb0 : SUCCESS in 14m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dfa42df9d54e4e759d7de82d600e0029 : SUCCESS in 12m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1eb55dd6ba7426aa3b7554560b6c53e : SUCCESS in 4m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7a6a4ed5f62040df800d5b4803fa9e7c : SUCCESS in 13m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9284ae5e0654499f8678204fade6263c : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e58d3730b34442a19b44187c8d8a6b71 : SUCCESS in 55m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4f4291b2727457188bdf4d9058c2cfc : SUCCESS in 23m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a229c1cb328946458e71ee35628a41e1 : SUCCESS in 39m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/489d83d8840b46c99ba8abad611b3403 : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ed7934f2e274cc0bfb0b6622330a23a : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e344d603b2f1461e9c3aa039e3852604 : SUCCESS in 1h 10m 52s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb11a4f49e8c46688ae8db51dc50e523 : SUCCESS in 33m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/709c72dfbb934938b026b412bdbf3130 : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":5},{"id":"83120ca74faba5cfb031fe264891dec249668d9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-25 21:14:45.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"60391be5a15eb3a3ba8a0b687a06c5541eeb51f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 22:09:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97ad292cc4244572a71d99eee546d184\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d7c00bdba9684621abfc03328355d276 : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":6},{"id":"22e5d64b0e88c0450e4df59cc0921a9affccec1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 22:25:03.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a3a0e5f1f224d50a804f3cbbd18002a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f1517387584372bf7684f7efffa97d : SUCCESS in 4m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/d03a918d960e41c88974a758a43abe35 : SUCCESS in 42m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ff66bfc96c04d7f891b703b82f35c59 : SUCCESS in 1h 05m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/68464fc4440b489d82a6fb1aae5fb7a0 : SUCCESS in 36m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86fff0ed8ec94ad38f9bfbe795725045 : SUCCESS in 45m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09e6334f8c3f4e36a238718af1646f71 : SUCCESS in 16m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c69194bee10442008faf8d86c30073ff : SUCCESS in 9m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85c1c48f65934e3d9e5296f75b287a45 : SUCCESS in 18m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f7ade670c5b848bea65d6001883adc46 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/46dbcc8e3faa47b0ad53f483fb17f929 : SUCCESS in 15m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0742e36fa9be4ca3b8d8a66dd8228999 : SUCCESS in 15m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dfb0f2f53c334863992c626ac61df99b : SUCCESS in 18m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4775791797fa4a52a6a436d0fd55aa21 : SUCCESS in 10m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b6907118999f4441a710e53fa96147be : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d1285d28e9c642a9adff2ad6d29a0d03 : SUCCESS in 15m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5846143fd3cd4d7598248f405965381d : SUCCESS in 7m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ea41a897629d4d148f08f9824b5acb0a : SUCCESS in 6m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d2c7edaa67014c15b405b0f04a2c8b28 : SUCCESS in 14m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4f5d49be064a47b1a92a3da222fd2ed5 : SUCCESS in 11m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/de2c6f425b434342a055a66b1db88cc2 : SUCCESS in 48m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8805394c307d4f7084f55e4d1ecca225 : SUCCESS in 52m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/66dd10419ca4462c89d7a94fd5557e95 : SUCCESS in 21m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2435f43b2c2c403a92bbd9c5cc4a33e8 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25ad062197934399acb4a5ecfcbc06a3 : FAILURE in 2m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a30d2d0faac94c5cbc9921e81d6bb9c8 : SUCCESS in 17m 33s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62a18f84f4e64f57a4bf5ba2d6ce3ca1 : SUCCESS in 51m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/045a97eefa7c49879059abe3636ff5bd : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":6},{"id":"0d1aa72a9ac19a9b933108a18c004092d4072c01","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-28 11:51:02.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"9bae7e30a69fb610d20b25c19f269d72f828cb24","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-28 15:48:36.000000000","message":"Patch Set 6: Workflow-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0c2478e8a494af266cf0c84002d7f495526be31a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-28 20:27:30.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"6aea8a2602691797171192739443346a7e957923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 21:41:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbeb2dd051eb49c79f0d0dbd012e441c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e0983d4afa4763a945026bb7d8966c : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/dee57d1ec6564a9d95ecf0fe17ccabc1 : SUCCESS in 56m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0a75462308f43ffb70fff6b16f30abf : SUCCESS in 1h 08m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d836bf005ed481f849b70b84e9c901b : SUCCESS in 48m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f5afbc591a14fbd8bab28e6a8432c99 : SUCCESS in 1h 01m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1212c581a424027828a82d276e9f71b : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3558023a273041f199f4e31138fc7d5f : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/62c3a2fbf5224eb69cb62b24f2d5bf12 : SUCCESS in 22m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/962f0dfd4c70439787b711eea3e2b33e : FAILURE in 3m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7004328ec08247019148016be3279edb : SUCCESS in 20m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7597f32ea4e74624bb2485d123bcc6c5 : SUCCESS in 19m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/25d7384d04044265a6e3b1e6f429d94d : SUCCESS in 32m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ce1c2b5317724fd88bc9d4f459445e83 : SUCCESS in 30m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1d3c18f3fb93450db2663f4e575d8096 : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ef83ed2732f94dffa2f7091bed1155e4 : SUCCESS in 15m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b98ff787a8f8430a8ef7fab0b6728d98 : SUCCESS in 16m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5350a03c70f84f0ab713aaa5eb439e2e : SUCCESS in 6m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d3b1eeee257d4d3188e33f05bb55403d : SUCCESS in 13m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ef5014b14ec94be5bc83148b7a34bf94 : SUCCESS in 7m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/237a285b626f40fcbc917118a84d9e15 : SUCCESS in 55m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4288c6e8c92e452cbcb5293dc230fc8e : SUCCESS in 44m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cbe3722e70ad4acfa7f884c04a6e2aaf : SUCCESS in 52m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da41c9d0d9f9467294677f02f7a2dfed : SUCCESS in 19m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9529fd94832c4707801a28ffa511c813 : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/960d8f68c6674963bd9324285489fdfe : SUCCESS in 18m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd1e5f2188484c53938864bcd94b29d7 : SUCCESS in 52m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0f31a6e005a4a0ea6d14f63b8f8ab1f : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":7},{"id":"7c6bc99b2c5eeefb9ee3774097ab886831bfee14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 21:48:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c35b6ddf0785410783344417dbe208fb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6312bb43a9024efcb1695c482feb1b3f : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":7},{"id":"0663413a18f846da588a015a54701993ceb2b690","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-07-30 18:31:28.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nThe following files contain Git conflicts:\n* swift/obj/replicator.py\n* test/unit/obj/test_diskfile.py\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"c3a1a7573bf47c16d317e20fdf5a2647ec8503c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 18:32:28.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 954379,8","accounts_in_message":[],"_revision_number":8},{"id":"a779d39a6980b4e5c601caea2a80a3327237bc2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 18:32:30.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 954379,8","accounts_in_message":[],"_revision_number":8},{"id":"6ce050226e58e31b87e9d35e4a7bcdb22f415d50","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-07-30 18:45:20.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"cbe735b54ec02af6aacc7143b816ea6f66bcd425","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 19:51:43.000000000","message":"Patch Set 9: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18c313c30d6f4d2f8e7ec9e6ab38335b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d446d86b69c4099bc9c8868063304d8 : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/10c3201388374a6cb7a4de1439224474 : SUCCESS in 35m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4d57d7e3cc6142b987719a25763cb52f : SUCCESS in 46m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4593fa06ee2b4780896fda0252e80d2d : SUCCESS in 39m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3eb6a12e729845c591f461dc3de474b4 : SUCCESS in 44m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/854ee7fca2324072a566be3310ada741 : SUCCESS in 10m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc56cfbdba6b46bb844cda32eea3423d : SUCCESS in 10m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4adf886ffa604e7882a44933b2d23861 : SUCCESS in 10m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb33ff82eeda480e8acb1d13bc1c2a61 : SUCCESS in 4m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a2b128cc413245c1a962b5a3d3a7273a : SUCCESS in 15m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d93b29a247a14b908001d99f487a484b : SUCCESS in 8m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/68b30bcd7d674a25bb31fe9bfe2e66bb : SUCCESS in 17m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6c4fb1b794c64c4aa6a031e24a57e863 : SUCCESS in 19m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/78090284fd31442e8d8867a25f46d8f8 : SUCCESS in 8m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/85063c01885041698090e0b31761e328 : SUCCESS in 12m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/517642cc86bd429da996c0ab088f050d : SUCCESS in 14m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7d59cbba91864620ba735ee6e594e1a6 : SUCCESS in 9m 04s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/79951ce1b15a4f629c6eff2cfba7e642 : SUCCESS in 7m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a23ff1b6c53e41c7983881778b34d0d7 : SUCCESS in 4m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c49a163c25b4788a4e2cf46e7504f4d : SUCCESS in 58m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/016af30d04db4b7987bf08d4947c97a9 : SUCCESS in 50m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/79509d27f1ed489eae7fd73dcfe25666 : SUCCESS in 55m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d1ce3e6cc504bb3a36cfb14735ad739 : SUCCESS in 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65a4c7835f4144689f2470e6b0028e32 : FAILURE in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ab5454de45f43e78a2556317e7d307e : SUCCESS in 17m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/936405428f464c3a878d97ce5698ba6b : SUCCESS in 41m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3dea7c760cda49bcbf946c6e5577693b : SUCCESS in 33m 48s","accounts_in_message":[],"_revision_number":9},{"id":"7aed847a006f5d94e48e689ef98d96f00eb4e9c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 20:27:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2734f05e0b24ca8b69b7f2105c8bdb6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f916660c9f649669d46ff92320f9e23 : SUCCESS in 1h 40m 19s","accounts_in_message":[],"_revision_number":9},{"id":"06293c36ebcb1353bf2a8dabf5baead562891adc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-07-31 16:08:50.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"b5882133320c39e84c69ea62945fd8b62eda53b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 17:25:42.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a75205bac38346cab2aa3e2317f3fb75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85864d53f20b41cea5bcc9bf4ebc5ac2 : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/d40e92ff006d45bb8bf61e11ab6ca5d0 : SUCCESS in 1h 03m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9d2c3149d134237a6df702228b13186 : SUCCESS in 1h 00m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bab195580dde4a35b0616a1627a29d64 : SUCCESS in 52m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/067f2c5c4d34416f913c3d6dddc12eb1 : SUCCESS in 1h 06m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0cdfd8bd6874b709f0fbeed27e7403b : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4063e4befa0d4cc49f21fa26d48f4c66 : SUCCESS in 22m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adb0091998e54c93ad9cf797ebb51e3f : SUCCESS in 19m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1e904c8affdc47309854149be83290be : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7a3e2427a3fa4d87b263c2bf1ead4823 : SUCCESS in 18m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ef438927104644449646e5a06a4bcf33 : SUCCESS in 18m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/67d4a77a2d114852b36c0333f40bd329 : SUCCESS in 32m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/81277ad48e5f4dc1a12a3cddea00099f : SUCCESS in 24m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/10802162549d4035bb54f33a95201954 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/afd2ca915e994d20bb550b4369638e3d : SUCCESS in 14m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3fa8d5782cf845de9ad4fcdb52063a92 : SUCCESS in 13m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2a05521464704e7e85ccc5b4daecb95a : SUCCESS in 6m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b27c6b811586428a8d48b2c50258bde1 : SUCCESS in 7m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7207545b74e246b199f5bdd8528abebc : SUCCESS in 4m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b2ce20dd1b4417b9404ee1aaefff0fe : SUCCESS in 53m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c90660b9d463492db2ed278f0136d08c : SUCCESS in 53m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/392f82efba0b45a7bd8f4bb585b54fbd : SUCCESS in 21m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e51dd08618a4d3da49ddf7a8d0cbe04 : SUCCESS in 21m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b0456b5bcc447cb79eedb0d8c8b2a7 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c9de605d8140459f821411526f1953d9 : SUCCESS in 18m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/898e80c989a245d684c26b9e6704c944 : SUCCESS in 1h 01m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58f9a18ebf3d4eadb169f44530fc13ef : SUCCESS in 28m 44s","accounts_in_message":[],"_revision_number":10},{"id":"60d98207f2014598c0488b68d993ef8862b10a3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 18:50:27.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/905145823e884491b2a3f4f1461ef362\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e7bef8576cc4f688d728406fa67e800 : SUCCESS in 1h 39m 03s","accounts_in_message":[],"_revision_number":10},{"id":"36915101a999455caf739af6cae2c0fcde67ca86","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2025-08-07 03:26:03.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"fd98e04c7b5fb94869568097efd9ba72145fe817","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-23 17:27:55.000000000","message":"Patch Set 10: Workflow-1\n\n(3 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":10},{"id":"b4ccb2124071c345d152f508f6555af2db33adad","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-23 22:59:51.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a2339a9dc1887515531ef0a290a39f2115a3c378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-24 18:18:18.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":11},{"id":"99159c6ca31ee9f8274fef5b8509b5128fab03dc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-24 19:07:31.000000000","message":"Patch Set 11: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"51486385cf8b12fa2e9ea50c74122164f2f1fcf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 19:18:46.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3b0e8d944274918a1e1880278a485d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c69831ebe38f4c9c8f7cb0919d07d1cf : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/61efe55030eb467fbd84156a19d9f6b1 : SUCCESS in 46m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c80d3156390b422aa61093dad1932faf : SUCCESS in 46m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c5e3f2bb75874ceb9606fe7f528df084 : SUCCESS in 40m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc1ef3c4218f4b37bb31df8f9ec38491 : SUCCESS in 43m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1163289acb124f75b6a043422ea35008 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/849c10f0602f472daca2c3e1d82ab94b : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c8a73065c644c80a1af5367f9d13eb2 : SUCCESS in 12m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cd76c481164842a6a8938afc581c693c : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ad28edd7a4954284b9cb35d4f578d3e9 : SUCCESS in 10m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/276092c4b3d24a3d8eca83323e857aa4 : SUCCESS in 8m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/779e54594bb94e9c9c5c52d32cfc897f : SUCCESS in 27m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2fc835fc7aaf49ecbc978d648e2a5c80 : SUCCESS in 30m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5013e5c55fca4fc6abdd39f4526573ba : SUCCESS in 11m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5433d19bf7cf437497d6956473616020 : SUCCESS in 14m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a87ff57e4fc0498a8e1933b7d07bbcd2 : SUCCESS in 13m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6ea03a7f52214dcdbd97e0512b5eb3e5 : SUCCESS in 8m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5a716f3778bd48feb50bdf9f76ed130e : SUCCESS in 14m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b313b1c589a1406b81ff9b3f9aa9a926 : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4b3dabf434944bc927f3617439b04bf : SUCCESS in 47m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8dedf5c9a1bc4b648f103b19b2fb5771 : SUCCESS in 35m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/76eb40a2f55c4c4b8fae331650dcbd55 : SUCCESS in 26m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cff273aa3d774944879570dd88373887 : SUCCESS in 21m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe18fd3901e14cdebc75ea9e4219b033 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8936124403ef4cf2bef87483818d80f6 : SUCCESS in 18m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f294206d78154dfb8775e290b2136c57 : SUCCESS in 46m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19a5ac4968f541d7a83b9cb07c86f402 : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":11},{"id":"180c86aad40b195a35b7431e4752f9db4c64a1ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 19:34:34.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50a36986bc414b99854c0c5b69843d5f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e210f6b9c5fe47748bd96be4704806df : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"d6ba532d3c8f614207934e734ade92b665c638b5","revisions":{"521481f2c9d678cd91650e143cf4f21c919daaed":{"kind":"REWORK","_number":1,"created":"2025-07-08 23:02:07.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/79/954379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/954379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/954379/1"}}},"commit":{"parents":[{"commit":"1428eb3b58e3b397250f19ac076afa0058487a47","subject":"Merge \"Fix traceback in invalidate_hash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1428eb3b58e3b397250f19ac076afa0058487a47"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-08 23:01:22.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-08 23:01:56.000000000","tz":-300},"subject":"add minimal trace to reconstructor get_hashes","message":"add minimal trace to reconstructor get_hashes\n\nChange-Id: I8a66c97f12a279bbed55d2a631f2498aaebe8429\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/521481f2c9d678cd91650e143cf4f21c919daaed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/521481f2c9d678cd91650e143cf4f21c919daaed"}]},"branch":"refs/heads/master"},"2f1dbb0e5aaa2cfbebc88656aa06fa326b73c24f":{"kind":"REWORK","_number":2,"created":"2025-07-11 23:33:41.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/79/954379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/954379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/954379/2"}}},"commit":{"parents":[{"commit":"1428eb3b58e3b397250f19ac076afa0058487a47","subject":"Merge \"Fix traceback in invalidate_hash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1428eb3b58e3b397250f19ac076afa0058487a47"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-08 23:01:22.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-11 23:31:33.000000000","tz":-300},"subject":"add minimal trace to reconstructor get_hashes","message":"add minimal trace to reconstructor get_hashes\n\nChange-Id: I8a66c97f12a279bbed55d2a631f2498aaebe8429\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f1dbb0e5aaa2cfbebc88656aa06fa326b73c24f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f1dbb0e5aaa2cfbebc88656aa06fa326b73c24f"}]},"branch":"refs/heads/master"},"8c1a22aceeedabe07223665a1c1dac1001ac70e4":{"kind":"REWORK","_number":3,"created":"2025-07-14 20:21:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/79/954379/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/954379/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/954379/3"}}},"commit":{"parents":[{"commit":"1428eb3b58e3b397250f19ac076afa0058487a47","subject":"Merge \"Fix traceback in invalidate_hash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1428eb3b58e3b397250f19ac076afa0058487a47"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-08 23:01:22.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-14 20:21:34.000000000","tz":-300},"subject":"add minimal trace to reconstructor get_hashes","message":"add minimal trace to reconstructor get_hashes\n\nChange-Id: I8a66c97f12a279bbed55d2a631f2498aaebe8429\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c1a22aceeedabe07223665a1c1dac1001ac70e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c1a22aceeedabe07223665a1c1dac1001ac70e4"}]},"branch":"refs/heads/master"},"a46749fe1ba49fdc22cd3429a7003ba0724ae9ae":{"kind":"REWORK","_number":4,"created":"2025-07-14 20:35:30.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/79/954379/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/954379/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/954379/4"}}},"commit":{"parents":[{"commit":"1428eb3b58e3b397250f19ac076afa0058487a47","subject":"Merge \"Fix traceback in invalidate_hash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1428eb3b58e3b397250f19ac076afa0058487a47"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-08 23:01:22.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-14 20:35:24.000000000","tz":-300},"subject":"add minimal trace to reconstructor get_hashes","message":"add minimal trace to reconstructor get_hashes\n\nChange-Id: I8a66c97f12a279bbed55d2a631f2498aaebe8429\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a46749fe1ba49fdc22cd3429a7003ba0724ae9ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a46749fe1ba49fdc22cd3429a7003ba0724ae9ae"}]},"branch":"refs/heads/master"},"2fc3d2cd4bfa1101f8fc977dbaa4da6ed631a0f6":{"kind":"REWORK","_number":5,"created":"2025-07-14 20:58:53.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/79/954379/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/954379/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/954379/5"}}},"commit":{"parents":[{"commit":"1428eb3b58e3b397250f19ac076afa0058487a47","subject":"Merge \"Fix traceback in invalidate_hash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1428eb3b58e3b397250f19ac076afa0058487a47"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-08 23:01:22.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-14 20:58:36.000000000","tz":-300},"subject":"add minimal trace to reconstructor get_hashes","message":"add minimal trace to reconstructor get_hashes\n\nChange-Id: I8a66c97f12a279bbed55d2a631f2498aaebe8429\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fc3d2cd4bfa1101f8fc977dbaa4da6ed631a0f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fc3d2cd4bfa1101f8fc977dbaa4da6ed631a0f6"}]},"branch":"refs/heads/master"},"809fc8df1030873788deb2dbc2762be578c65c7b":{"kind":"REWORK","_number":6,"created":"2025-07-25 21:14:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/79/954379/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/954379/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/954379/6"}}},"commit":{"parents":[{"commit":"1428eb3b58e3b397250f19ac076afa0058487a47","subject":"Merge \"Fix traceback in invalidate_hash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1428eb3b58e3b397250f19ac076afa0058487a47"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-08 23:01:22.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-25 21:11:29.000000000","tz":-300},"subject":"add minimal trace to get_hashes","message":"add minimal trace to get_hashes\n\nThis will output DEBUG tracing in the reconstructor, replicator and\nobject-server\u0027s use of get_hashes.\n\nN.B. the replicator likes to create all it\u0027s policies jobs for all the\nparts on all the disks and farm them out to a single GreenPool, so we\nhad to be careful not to leak a trace instance onto their threadlocal\nwhen those greenthreads stay around for a lot longer than you might have\nexpected or hoped for.\n\nChange-Id: I8a66c97f12a279bbed55d2a631f2498aaebe8429\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/809fc8df1030873788deb2dbc2762be578c65c7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/809fc8df1030873788deb2dbc2762be578c65c7b"}]},"branch":"refs/heads/master"},"1c617dddd8a3472180c873c486e390f86ecd3319":{"kind":"REWORK","_number":7,"created":"2025-07-28 20:27:30.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/79/954379/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/954379/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/954379/7"}}},"commit":{"parents":[{"commit":"1428eb3b58e3b397250f19ac076afa0058487a47","subject":"Merge \"Fix traceback in invalidate_hash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1428eb3b58e3b397250f19ac076afa0058487a47"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-08 23:01:22.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-28 20:23:42.000000000","tz":-300},"subject":"add minimal trace to get_hashes","message":"add minimal trace to get_hashes\n\nThis will output DEBUG tracing in the reconstructor, replicator and\nobject-server\u0027s use of get_hashes.\n\nN.B. eventlet tpool workers are long lived, since we\u0027re going to use\ntheir thread-local for message passing, we have to be careful not to\npull a stale trace from a previous context.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I8a66c97f12a279bbed55d2a631f2498aaebe8429\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c617dddd8a3472180c873c486e390f86ecd3319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c617dddd8a3472180c873c486e390f86ecd3319"}]},"branch":"refs/heads/master"},"345f81ee41a0d7ac654ab69b052f4717c1c0e010":{"kind":"REWORK","_number":8,"created":"2025-07-30 18:31:28.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/954379/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/954379/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/954379/8"}}},"commit":{"parents":[{"commit":"c04061894f81ad52e6c210ba2f21bf773c23cec3","subject":"replicator: Add sync_batches_per_revert option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c04061894f81ad52e6c210ba2f21bf773c23cec3"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-08 23:01:22.000000000","tz":-300},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2025-07-30 18:31:28.000000000","tz":0},"subject":"add minimal trace to get_hashes","message":"add minimal trace to get_hashes\n\nThis will output DEBUG tracing in the reconstructor, replicator and\nobject-server\u0027s use of get_hashes.\n\nN.B. eventlet tpool workers are long lived, since we\u0027re going to use\ntheir thread-local for message passing, we have to be careful not to\npull a stale trace from a previous context.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I8a66c97f12a279bbed55d2a631f2498aaebe8429\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/345f81ee41a0d7ac654ab69b052f4717c1c0e010"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/345f81ee41a0d7ac654ab69b052f4717c1c0e010"}]},"branch":"refs/heads/master","description":"Rebase"},"e9b7a9897565bf80838f0503fe9ed1edf6abbf63":{"kind":"REWORK","_number":9,"created":"2025-07-30 18:45:20.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/954379/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/954379/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/954379/9"}}},"commit":{"parents":[{"commit":"e5cdd0ff4813867c647d947659bebfc673069b59","subject":"replicator: Add sync_batches_per_revert option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5cdd0ff4813867c647d947659bebfc673069b59"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-08 23:01:22.000000000","tz":-300},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-07-30 18:45:08.000000000","tz":-420},"subject":"add minimal trace to get_hashes","message":"add minimal trace to get_hashes\n\nThis will output DEBUG tracing in the reconstructor, replicator and\nobject-server\u0027s use of get_hashes.\n\nN.B. eventlet tpool workers are long lived, since we\u0027re going to use\ntheir thread-local for message passing, we have to be careful not to\npull a stale trace from a previous context.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I8a66c97f12a279bbed55d2a631f2498aaebe8429\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: indianwhocodes \u003cnairashwin952013@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9b7a9897565bf80838f0503fe9ed1edf6abbf63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9b7a9897565bf80838f0503fe9ed1edf6abbf63"}]},"branch":"refs/heads/master"},"9689669b47b92d1c3c3f19365276dfa8db6eb967":{"kind":"REWORK","_number":10,"created":"2025-07-31 16:08:50.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/954379/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/954379/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/954379/10"}}},"commit":{"parents":[{"commit":"e5cdd0ff4813867c647d947659bebfc673069b59","subject":"replicator: Add sync_batches_per_revert option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5cdd0ff4813867c647d947659bebfc673069b59"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-08 23:01:22.000000000","tz":-300},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-07-31 16:08:19.000000000","tz":-420},"subject":"add minimal trace to get_hashes","message":"add minimal trace to get_hashes\n\nThis will output DEBUG tracing in the reconstructor, replicator and\nobject-server\u0027s use of get_hashes.\n\nN.B. eventlet tpool workers are long lived, since we\u0027re going to use\ntheir thread-local for message passing, we have to be careful not to\npull a stale trace from a previous context.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I8a66c97f12a279bbed55d2a631f2498aaebe8429\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: indianwhocodes \u003cnairashwin952013@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9689669b47b92d1c3c3f19365276dfa8db6eb967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9689669b47b92d1c3c3f19365276dfa8db6eb967"}]},"branch":"refs/heads/master"},"d6ba532d3c8f614207934e734ade92b665c638b5":{"kind":"REWORK","_number":11,"created":"2025-09-24 18:18:18.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/79/954379/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/954379/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/954379/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/954379/11"}}},"commit":{"parents":[{"commit":"e5cdd0ff4813867c647d947659bebfc673069b59","subject":"replicator: Add sync_batches_per_revert option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5cdd0ff4813867c647d947659bebfc673069b59"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-07-08 23:01:22.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-24 18:16:41.000000000","tz":-300},"subject":"add minimal trace to get_hashes","message":"add minimal trace to get_hashes\n\nThis will output DEBUG tracing in the reconstructor, replicator and\nobject-server\u0027s use of get_hashes.\n\nN.B. eventlet tpool workers are long lived, since we\u0027re going to use\ntheir thread-local for message passing, we have to be careful not to\npull a stale trace from a previous context.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I8a66c97f12a279bbed55d2a631f2498aaebe8429\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: indianwhocodes \u003cnairashwin952013@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6ba532d3c8f614207934e734ade92b665c638b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6ba532d3c8f614207934e734ade92b665c638b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
