)]}'
{"id":"openstack%2Fswift~954640","triplet_id":"openstack%2Fswift~master~I8f198298a817be47223e2f45fbc48a6f393b3bef","project":"openstack/swift","branch":"master","topic":"checksums","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-07-16 14:06:45.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-07-11 13:43:20.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I8f198298a817be47223e2f45fbc48a6f393b3bef","subject":"s3api: add compat test sending too much body with checksum","status":"MERGED","created":"2025-07-10 16:56:10.000000000","updated":"2025-07-16 16:25:53.000000000","submitted":"2025-07-16 16:24:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"954640-checksums","meta_rev_id":"86cccaae2d370e281a8c72ef5d9c736b1a9af1c4","_number":954640,"virtual_id_number":954640,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-16 16:24:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-07-16 06:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-11 13:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2025-07-16 06:32:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 17:19:06.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-07-10 18:26:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-16 06:32:01.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"ec26088dda806ba8de1a92e7dcd036e2e8749f47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-10 16:56:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83078c10016060ff3a60c500af1d9d74508790ce","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-10 17:19:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d62bad9d979ac76b5b68622497cd5e5280936d91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 18:26:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e9ab919d8564a9383598e335c242e97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b49d4b93d96d4a98af6c49af112d90aa : SUCCESS in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/ab8e712a6fe74deeb74a74e62948ec48 : SUCCESS in 1h 22m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5d4dbfafc0b04a299f942c6e8a3fa092 : SUCCESS in 1h 03m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/07f921ef58014b0d8984c14d675bd03e : SUCCESS in 53m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aae2e9f18ef0472b8112df4cd643d490 : SUCCESS in 1h 03m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af3a12ff426741e0a84a2f5da539efc7 : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd7e4c4624ac477cb089399c44109d40 : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b923e0515518474587be1351e58050ac : SUCCESS in 20m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/941f5a8d5c46468280b8cfcbfee3b0db : SUCCESS in 7m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a85e837b001e45778fa268acc6c101e2 : SUCCESS in 19m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/876d349a414b4efca4809e4606803ad9 : SUCCESS in 18m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/686fdb70e3fc43cbb7ec6fa897a8c577 : SUCCESS in 22m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/09aa18853dde43d2a528b7a1714aa2fa : SUCCESS in 29m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/160b9c54c86548f5bd1f0af90f10f45f : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7dbc4b01f1124438859c0394a4c14f6e : SUCCESS in 15m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ea2966e0b8044127bd1594ee415daef8 : SUCCESS in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/32d03e796230427184732588f043b9c8 : SUCCESS in 5m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b7067dc169c49e3ab5812d18ec48d8e : FAILURE in 19m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ae054d5eec18414fb7dec314fe486806 : SUCCESS in 8m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/06173a70f7da4aa9bbcb8f226f259030 : SUCCESS in 1h 14m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cfbb75d982a54458a71bb7ea876c8a2e : SUCCESS in 50m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a1c0cf4a14d439da8bf2f3d63c5b005 : SUCCESS in 53m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a62c09b4a5e148b5ad1340dbda167089 : SUCCESS in 20m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67618e68b8e547c7ba3b5a01c53e8c4f : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55f991f238a346dba2ba54aedd41ff4e : SUCCESS in 18m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66eea29128d24a4dabc38fd44f292252 : SUCCESS in 56m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14730dbf719e4d2cb446c43f10328ea9 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2d01ee8ad7e25493a1e560175a143a9020a187ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 21:29:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e73323c5f611458e93279e4c938ad755\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cc3a1e9a2eb64c7e8912ab47cf78046f : SUCCESS in 1h 32m 56s","accounts_in_message":[],"_revision_number":1},{"id":"710356c5ba315320722f41bb68a05d0b9498e5dc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-11 08:37:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f107e7ac0990e55d6163f2a4f7f55c2880efed6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-11 09:34:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"60b1b85825ea7d06988acae348409beea9f81aa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-11 09:48:11.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"570ab1b9e45e686534f3111f4cca6e0da17af3c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 11:06:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe0ab0015b8449dc8f2842c2be568a29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c95f8b567f413e9f622a6396f55b4a : SUCCESS in 11m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/e8328eb7847e4248b7adabfc05da53a1 : SUCCESS in 48m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eac2d925d73c480596a9dc4c06b77e71 : SUCCESS in 59m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db2af600991240d18bc8fadc2c830493 : SUCCESS in 47m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a30f0ef6b7af43f9bd2336bdcf4c7290 : SUCCESS in 1h 02m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e161dfa49e57424e964521176be6e325 : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35597e3590d64c109053efa1d6d5075c : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97d1f82acbb14755aeb2236a3e1f9135 : SUCCESS in 10m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/354d7a5a1cea4258b3a2868eaa1da08a : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/058310eb73b748748ca79a7f764cba29 : SUCCESS in 20m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f5acb9b298eb4b23827570eb4f54eed4 : SUCCESS in 19m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ece27a98ec684ae0a1b54d7e31b7bd20 : SUCCESS in 26m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/85644ad800664d47b0b418292aee19ac : SUCCESS in 28m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f6e6755e5b534e53a94471d330e74155 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6b9a4a8e8f454f989caa2179eefad10f : SUCCESS in 13m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7b03624dda0b4ce68b8b78f7b35113df : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f7b8ac76e86b472982a75ad6b8ac4de2 : SUCCESS in 3m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69249af08b3d43beb81b6e99a4a46cb4 : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/897a2c765c0c487ab969a7ea0899ba48 : SUCCESS in 9m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f51a444fecc450596b3470ac4bca327 : SUCCESS in 50m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f32f5dd198c4a6ca2a8a7b998eec3b4 : SUCCESS in 49m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5cfce2b6368c4c5a98cc9611cc85cccc : SUCCESS in 42m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c07247eea91d4ec48eefaf718a2578b2 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63b4fd7877b347ca8e14e1c3cff039ef : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/df27004a116047119aa0ad63ef2de3bc : SUCCESS in 18m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/922b438fcf1c4c1cba78e04fe2895f8c : SUCCESS in 45m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75708345c7b8423299fa94d0c525cb7b : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":3},{"id":"58d1ba4dfe7b95684bd2be2fe49ffd284231d0d6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-11 13:43:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ebdc8ff82ed037af5aa87f1c6a4b3c8c76a8e69d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 14:13:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03c1e367e65746efa96f38c03f4f6360\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1906b8a1f2a5492599341028a60f6f9a : SUCCESS in 1h 31m 43s","accounts_in_message":[],"_revision_number":3},{"id":"56cb4d7d30027ca0741819e39b1db9d8fb6ebe30","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-07-16 06:32:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9bac7d1ad33ef3c972c449a2e5e61a0ce6f969f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 06:32:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3a6ab8bb4413e8e8e5ab8519aacb063b8a191b05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 07:40:05.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/121d777cf37847a08166449ee1db7207\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d255cc20b0c40198df1939fde1925a6 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/f2d6c96f606146c4a60f1b1b714530cb : SUCCESS in 1h 02m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/599983732f4c4f8c9529a771aeaf0117 : SUCCESS in 1h 03m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/98121eb86b0c4e1cbb5312edba29167c : SUCCESS in 38m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8b259380baf41a090a9e9d12a391766 : SUCCESS in 44m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d1af1a5bcef1488c883286326d02c8b6 : SUCCESS in 7m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e742cf11abb94233819e7a288f58bece : SUCCESS in 17m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a0a0105046094980b39c490f195e0182 : SUCCESS in 16m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f2ca9cc4db9940adacc4cee596583d04 : SUCCESS in 23m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7dbe53bc819f4fe0abeed4b255ed938c : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/90195e754cd54565ab099ae863e399dc : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fa61d4461ae74b3ab7c562ba7a2e22ac : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e8baeb59436c4ee2a0aa5d51a1450940 : SUCCESS in 8m 37s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e53c2be737e3422f81089af9dbd0b334 : SUCCESS in 8m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/99301557b43840a4a494abb7baf9d87e : SUCCESS in 58m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8defc24ac06b419aa1b0b8dfe6511963 : SUCCESS in 34m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3f6e874aa32548debf105712aa621075 : POST_FAILURE in 40m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31611cee38d34ef680661e32866935d9 : SUCCESS in 19m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37e9a9d6b91d4a08be05ccf1cc0e46b6 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/14b8a5c467f9468f93bfe684d05aec37 : SUCCESS in 18m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a36a2d5387847d495513f21526fca4e : SUCCESS in 45m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1683b4d4a2f54fc582281f945f135d05 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":3},{"id":"cdc6fee00ef156977dad5f263e0725d376080401","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-16 10:51:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"428326e82ac1a4541d5d97045ebd863157d30d7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 12:49:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efc73fa07e3a4bd79d3162e037eb3818\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/39f734ed5f2c4431ab5f5243ed96798a : SUCCESS in 1h 41m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b916b7402c1b10e6cd4b410319d997d7b1ef4084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 13:55:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/614fa3d1e0ae41368bcbc96ee16ffaca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffc470b862b04103b097372db836ffaa : SUCCESS in 8m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/93f45b1f88314766bb91c5a6fa4aa791 : SUCCESS in 30m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f223fed09394498a96a11d79bf2b560 : TIMED_OUT in 3h 02m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6b82c3c5432a40f784e2258adaf9fcf7 : SUCCESS in 34m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2d6268fbed24d1d89e2b553406c83c6 : SUCCESS in 1h 01m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82540ee9107a457ead4b33affb531184 : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48f90c2bb92240f3ab2d96d3c5689089 : SUCCESS in 20m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86a56af238d24a3bb3ec78c244e746f1 : SUCCESS in 25m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/637312cfa496458da8b41582d4f3874d : SUCCESS in 3m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6c5cd55430874dc7a35d6860ecfaa0e2 : SUCCESS in 17m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/33f380fec29547a086d13fee64ed214d : SUCCESS in 7m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4c7ce3c68e6a4defabed1ff1306ca801 : SUCCESS in 19m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a89005d851f8411abd62e8fadfb2bf8e : SUCCESS in 20m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d584fee1199b4f7ca7f740299f465261 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0f9fe37b0a274c2ba22b0334a7eedf23 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/680ef522e8224f949fb515c70c876299 : SUCCESS in 9m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7b3dbd1ea00e45e2ae499c54e8e6af85 : SUCCESS in 9m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d257cc7ffb6b4f158bfe320ec689f826 : SUCCESS in 13m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0e51b13b144748228e287a1f191e2bd4 : SUCCESS in 9m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e327bd1f1b3f4d158a8fc6135b1e2c6a : SUCCESS in 34m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/152ea8dee34044a3ac4ad1be1e84d4cb : SUCCESS in 52m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/67caf6799c704acf93e116cae1240dba : SUCCESS in 53m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/232d0c1d933a4687915a75eaf14b0cb5 : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79b0806220874b1d8f7714572869364d : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc6a7ee1e10b416796f8c3c94560b312 : SUCCESS in 22m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a26c63ad7894acf8b3b46a9750ec064 : SUCCESS in 44m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/998ecfa60ff841f0bc0e1acfacd4e231 : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":3},{"id":"be733fa9510dfa94c2b1822851e6d92c623c4a5d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-16 14:06:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"187e79ca1c2abf1aca8e13541ebbc7694b4e9e58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 15:10:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/796b27031f7845398293782512a9c5f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e625ee4f762464ca465a2d50d8af60a : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/392cbc44ca684307b63bf9062e06e0af : SUCCESS in 47m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/811211e4e0d74ba6800a8db4a2390cbb : SUCCESS in 50m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c22573156d5749f9addfc788a4393a57 : SUCCESS in 50m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86414ff3f56640f2bbc16f7aaaab507c : SUCCESS in 49m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d15d78bdd36e4929a1dc2a341021293d : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c15fcaefceea45d3adacfefdc7c2cf2b : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/619cbb6720d649278f78751cbf8f154e : SUCCESS in 19m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26bb5f57f84a494a9597e2da7fb98924 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fdabee90d1f9487aa9c938f65a5efa80 : SUCCESS in 15m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/304e20ccc9ce425d8996b33be7fc446b : SUCCESS in 17m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a50991f207bc45c796563e5289ac7bae : SUCCESS in 19m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/de55e6e35d944f7f946e1975eb1c4762 : SUCCESS in 21m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9ce553009a6f457fa78f6728625085f4 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1f5ed5852e3941328097ad9348bb21f2 : SUCCESS in 13m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f19ead235ddc418185b7c268c02b1e37 : SUCCESS in 13m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e81b601b814b406ca5976053a5da9f2a : SUCCESS in 6m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ce737a0fa9d348a19f89cdb14f72b55c : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0c5a98d6efcc4c8ca4912af5f65f0f44 : SUCCESS in 8m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/58e2999dd5f146b68ffe7fa0b9f724d3 : SUCCESS in 54m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6f929921151948c49f0fb9bf24b08a9d : SUCCESS in 43m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ae9c7854573473283f1a446c1b59815 : SUCCESS in 52m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed5fc1d2bd7e4bb6966248ac1550be39 : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f0fd841b94f4ca2b657f2f6f1b37802 : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2bd98fda06c64be78a0373d2b98dfe6e : SUCCESS in 18m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59869b451a0844239b8330ee01c33cdc : SUCCESS in 55m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/586374d4af404da298d162fb95dac920 : SUCCESS in 46m 31s","accounts_in_message":[],"_revision_number":3},{"id":"dff0a4b285375e61dbb0cb44a19d65d3d7c8a961","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 15:10:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a8736367f1c3e5f72a629807868ba7fe6a640d70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 16:10:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f11f7cc987443ea87efd6e114fdf34f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dc9ca74bd4d3478d877b43d82dde805d : SUCCESS in 1h 47m 33s","accounts_in_message":[],"_revision_number":3},{"id":"a0b62593fe1c8d624cf7595cf623b1fa49ca6033","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 16:24:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a040877a53e45118691188f2b89db89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32f04ec8e5e4077ac1bb42bf73782e7 : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/18b4e53789b04a40a33252ce0429d32f : SUCCESS in 1h 05m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7dfc7a23629f42189383521f774de7ee : SUCCESS in 59m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ea40f0ed40c74d3590ca9d465aac3cc6 : SUCCESS in 1h 04m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/735fdb6bdcff4fd0b920d86bead22258 : SUCCESS in 1h 03m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/90b3449481964bcfb9ce739f2dbd6455 : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b66e85cce89545108cf90df36d54bcb8 : SUCCESS in 19m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/338c57731c1d4b1d925a306911463d58 : SUCCESS in 22m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1e8c62c046b34150bc7f99da694ddb2d : SUCCESS in 28m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/623185bdb6854423a254c1b5b8266c39 : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/11c0fd7d0b8b4bef9ca63b404ee1eb8a : SUCCESS in 13m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8c8b9f4e5cdc4b1996ba658c691d60dd : SUCCESS in 23m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ba4ba1e72ac942a785982a5cb836f131 : SUCCESS in 8m 47s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/00d3e6a3a71c46c1a6f4ea85efdbe2cd : SUCCESS in 10m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fae9df8811b4fd798f0b5f600c86811 : SUCCESS in 54m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4eb14151214d460cbd16d82b27410482 : SUCCESS in 54m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aaf4614e8ffa409ab3f18de92b754817 : SUCCESS in 53m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd916384a73c431d9b7163c4d433e90c : SUCCESS in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b17002a84c2496897503de6af07e826 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4b5b1f4fe6b247b7b077a9d301243682 : SUCCESS in 23m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e03dda1158f448259a450fb0124002bf : SUCCESS in 1h 01m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bbeb3154b48471eb8abb70a9df79730 : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":3},{"id":"cbf1981383e0e3ba2853466ac0ca455c05140996","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 16:24:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"86cccaae2d370e281a8c72ef5d9c736b1a9af1c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 16:25:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23b4a430c3e34b178d1b97a8a809565c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce73b15255e4b769268f34c9a7f51c4 : SUCCESS in 57s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fa473c75fd014f13bea552666a77e6c4 : SUCCESS in 51s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"351ee7279004e41a8bfa4d0772182ccbb88d41ed","revisions":{"01d74ca6b69be6a82e0fda5b3a0a49bffb9a65b9":{"kind":"REWORK","_number":1,"created":"2025-07-10 16:56:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/954640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/954640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/954640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/954640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/954640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/954640/1"}}},"commit":{"parents":[{"commit":"24b42f4fd2996aa2a3db03f7f342b44a705d9882","subject":"sq? refactor S3Request._validate_checksum_headers return type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24b42f4fd2996aa2a3db03f7f342b44a705d9882"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-10 13:51:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-10 16:56:00.000000000","tz":60},"subject":"sq? add test for sending too much body with checksummed req","message":"sq? add test for sending too much body with checksummed req\n\nChange-Id: I8f198298a817be47223e2f45fbc48a6f393b3bef\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01d74ca6b69be6a82e0fda5b3a0a49bffb9a65b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01d74ca6b69be6a82e0fda5b3a0a49bffb9a65b9"}]},"branch":"refs/heads/master"},"785161fc97de2b46f48bf044ff102f4f9f2de2a1":{"kind":"REWORK","_number":2,"created":"2025-07-11 09:34:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/954640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/954640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/954640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/954640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/954640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/954640/2"}}},"commit":{"parents":[{"commit":"be56c1e25822a23a6f6ad30d190782a395f3149f","subject":"s3api: Validate additional checksums on upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be56c1e25822a23a6f6ad30d190782a395f3149f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-10 13:51:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-11 08:38:07.000000000","tz":60},"subject":"sq? add test for sending too much body with checksummed req","message":"sq? add test for sending too much body with checksummed req\n\nChange-Id: I8f198298a817be47223e2f45fbc48a6f393b3bef\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/785161fc97de2b46f48bf044ff102f4f9f2de2a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/785161fc97de2b46f48bf044ff102f4f9f2de2a1"}]},"branch":"refs/heads/master"},"351ee7279004e41a8bfa4d0772182ccbb88d41ed":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-07-11 09:48:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/954640/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/954640/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/954640/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/954640/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/954640/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/954640/3"}}},"commit":{"parents":[{"commit":"be56c1e25822a23a6f6ad30d190782a395f3149f","subject":"s3api: Validate additional checksums on upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be56c1e25822a23a6f6ad30d190782a395f3149f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-10 13:51:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-11 09:43:48.000000000","tz":60},"subject":"s3api: add compat test sending too much body with checksum","message":"s3api: add compat test sending too much body with checksum\n\nAdds a test that verifies extra body content beyond the content-length\nis ignored provided that the checksum value matches that of the\ncontent-length bytes.\n\nAdd comment to explain why this is the case.\n\nDrive-by: add clarifying comment to unit test.\n\nChange-Id: I8f198298a817be47223e2f45fbc48a6f393b3bef\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/351ee7279004e41a8bfa4d0772182ccbb88d41ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/351ee7279004e41a8bfa4d0772182ccbb88d41ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
