)]}'
{"id":"openstack%2Fswift~954827","triplet_id":"openstack%2Fswift~stable%2F2024.2~Ica2e1cb018106164c6cea0e6e8e9c6bce32d416b","project":"openstack/swift","branch":"stable/2024.2","hashtags":[],"change_id":"Ica2e1cb018106164c6cea0e6e8e9c6bce32d416b","subject":"CHANGELOG for 2.34.1","status":"MERGED","created":"2025-07-11 23:45:19.000000000","updated":"2025-07-12 03:28:23.000000000","submitted":"2025-07-12 03:27:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954827","meta_rev_id":"1f4166a87e83b4f63c2995259e959869b4b22653","_number":954827,"virtual_id_number":954827,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-12 03:27:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-12 03:00:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-12 03:00:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-12 00:11:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec8409e7af357db526a5812f792567e26562a31a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-11 23:45:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71d477a3d9c4044fd82d905516d3dfa8747d7908","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-11 23:49:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cc343b5275a11c316240bcb508843f0bbb67011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 00:11:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90f86e2d1b764794b6e46b861a6101f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/874d601ad7d3480fb663a2bb60ba697c : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbcc4a68f6b046508e34b6ab74c03fca : SUCCESS in 6m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c314f56c4e2f4ecd87126121f086ee13 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/91d7a0fcb8c84adcb20dbe6f74a71122 : SUCCESS in 20m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/67c2c11c12974a07a9732b0a7cb1fbbc : SUCCESS in 7m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7d6eb7e5b7744492b6b58a16dd629888 : SUCCESS in 7m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/74762fdadeca4f9faa82e3ab780c17dc : SUCCESS in 21m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f6b12753e8864957a1a8f4aa36ad4204 : SUCCESS in 9m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c46f67d532aa47fe954413a07e37340a : SUCCESS in 20m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d524f9a27af4ca6ad72efe0ee453608 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"01408aa7aacc9765123a1b7cd8b7f393d68c974f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 00:23:11.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57d7c66f063d4fa19c6008b5bf7a1403\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/78eabdd24add4b4eb8e06d1bf0e07a24 : SUCCESS in 15m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f4aa673bbadc45e38c0e0514173976cd : SUCCESS in 15m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dfe515cd482f48c6ba33ae49d4c298de : SUCCESS in 17m 54s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/e342bf6507344dabaa94f6a9f16799bf : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1aa8e9380a54e539a30293ab33a6043 : SUCCESS in 19m 35s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/91a8365828434cbea7a0e571a19ce063 : SUCCESS in 16m 43s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/0cc3d55d5f5c42169fe0c6fcae505183 : SUCCESS in 16m 02s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/c36a0ea8bb8c4bc1b4ce7d7137205262 : SUCCESS in 16m 27s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/0d24429e54a54cca874504416ce08b85 : SUCCESS in 16m 05s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/c50d2d12148b4cc58eb05d1a1e79d44a : SUCCESS in 17m 43s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/9ad482e54f774f32b67321f7d8814b5f : RETRY_LIMIT in 8m 00s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/7a8d6054a5f14c75b970b8f96bdedbd8 : RETRY_LIMIT in 5m 24s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/cc981390323942788e1ed94b2c21630d : SUCCESS in 18m 58s","accounts_in_message":[],"_revision_number":1},{"id":"490ff76a7329081451d42502e0c860512f098c8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-12 01:00:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"490677d84b214e010e082dce0a0385f36b7b7711","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-12 01:01:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b2f4cf65403d96da85a95e73bad31ce88213a45","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 01:28:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2381d31800ed4346b2406e52ffa970fc\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5aad4798ce8c4e719930569adc754978 : SUCCESS in 14m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0f23e988b5b4433eab3b535ed87b3ac5 : SUCCESS in 20m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ed77015f3c9344f2a24e8a269b2355b6 : SUCCESS in 21m 07s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c812b2154e994a57ade30be0a44a2b20 : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e0f8f51f6684614a1b22f2e5262d5cf : SUCCESS in 19m 21s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/74faa60b5d5a434a96eb856d0d55ef28 : SUCCESS in 19m 04s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/e50f9ec54a6d4c228562bff06e77bebc : SUCCESS in 18m 22s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/18272d7b38004d32857100ba98ffa6bd : SUCCESS in 17m 39s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/9ef19639d3584c9cb7dcc4d310ef19ba : SUCCESS in 18m 04s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/3daa27016f4247078bee869afc63bab7 : SUCCESS in 15m 42s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/f1191fcf725d4507b377a3431087a46f : SUCCESS in 8m 24s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/a4c96238ed874af6b1ae4fff9cfceb6a : SUCCESS in 17m 35s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/3fc9720f336d43a49b116e6786c94ed2 : SUCCESS in 18m 39s","accounts_in_message":[],"_revision_number":2},{"id":"9249427d2e04cc2588a06b79b1d6778f2e85d07f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 01:29:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c369cf9769814e62b8d5fc658f2b7b72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ae9664ab249449ba8a20605460f39cc : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce62506cdb97429bab125f3b2f0d5418 : SUCCESS in 4m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/84eb47c9756747eca2c15e9f6d15d0f7 : SUCCESS in 3m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59a342f3673c482f82a285b55dde265d : SUCCESS in 15m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/44e1fc6dbbf5491b898795d08467d470 : SUCCESS in 15m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ca00787583234686ae9797ca8e9df6a4 : SUCCESS in 16m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/690adc8dd166452db028539bf69bfb75 : SUCCESS in 23m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1f389859af12463f8ce888a8aa1f581b : SUCCESS in 10m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f6a18ad66914d259000764932c73c06 : SUCCESS in 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a2abce539d485aa44aef64054b5cc2 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f80f7cbc6c60ed9d98b08383b88874388cdf43aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 02:28:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b6f1a8e67784e46a4f9ce2d0da4f71e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9850fbb96f4422a8cc03f7f08ee5661 : SUCCESS in 59m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea3da8c9b0dc4376b525be287586b308 : SUCCESS in 1h 02m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3364818f1cb46108f436936e2b19986 : SUCCESS in 50m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b0fe21de39846acb308e6e9c61164ee : SUCCESS in 1h 25m 00s","accounts_in_message":[],"_revision_number":2},{"id":"cb07bd359a7a65ec59a0cfdbefa0522260ca745f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-12 03:00:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6598b10e2a3d9aac3a29992992c714a6d424dea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 03:01:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"78a0bb270686a63b08d99500d5015418dcdccbf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 03:27:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f72cdbab7ee42c89564e78c7ab39e8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df3842e5fe4447569632b888ec03f561 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2884e661c56c498da48022b602bb0c99 : SUCCESS in 6m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8b96fc41caa24a949cf34e06d6d992a4 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06f2cb76f4324732929cd344db982abd : SUCCESS in 17m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6338f540c33d4698b2294bab9e793b69 : SUCCESS in 17m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d538e522b3f7479086542f749fa742de : SUCCESS in 18m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dbf276e664114388ba16a4cd71d0b47c : SUCCESS in 15m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4311ff3c1b254658b8f72c5c8da075f3 : SUCCESS in 23m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/80cc4f8f6c36409eacb127c483a721c7 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/24a0260557184f37a9f9d5fd24f0d478 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3125c100dc50467294eed4196430b9e2 : SUCCESS in 19m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a2868399fa53499fbdf226ad45322029 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/791e07bf2a234c38bdd712204b869c98 : SUCCESS in 14m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fa9112aa085c4486bc02eab06a6ff119 : SUCCESS in 18m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eb6f65f2e2d648178411deeaacb22053 : SUCCESS in 6m 45s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cd62fb9dcd6a4813b124c572ed0c1042 : SUCCESS in 7m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d640406a5b14611a4eedef4d1f3659c : SUCCESS in 18m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/119ff0e3dffd4c33a7086c1fa3ab9bc0 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"426eac6588f1ded532ef0dcc2ef7c8be2cebf824","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 03:27:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1f4166a87e83b4f63c2995259e959869b4b22653","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 03:28:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5304cca43fcb412a95836efda907147a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d402fdea3ff643a9b48ddfd327feaa4f : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14a4fd3cf4d64683aba7dc328806cdbc : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f25285c5b06e4fa4b00abef7a15ac985 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5f776b6ee0235544dc46f0750e0c67f0a448b6a2","revisions":{"758f06160000b7f3fa6f00d2026fc7234b5c2e5a":{"kind":"REWORK","_number":1,"created":"2025-07-11 23:45:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/954827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/954827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/954827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/954827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/954827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/954827/1"}}},"commit":{"parents":[{"commit":"1534385bcc10de4aaa7aa2762bf9e7961bb2ef17","subject":"docs: Call out that xprofile is not intended for production","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1534385bcc10de4aaa7aa2762bf9e7961bb2ef17"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-07-11 23:44:35.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-07-11 23:45:01.000000000","tz":-420},"subject":"CHANGELOG for 2.34.1","message":"CHANGELOG for 2.34.1\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ica2e1cb018106164c6cea0e6e8e9c6bce32d416b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/758f06160000b7f3fa6f00d2026fc7234b5c2e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/758f06160000b7f3fa6f00d2026fc7234b5c2e5a"}]},"branch":"refs/heads/stable/2024.2"},"5f776b6ee0235544dc46f0750e0c67f0a448b6a2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-12 01:00:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/954827/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/954827/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/954827/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/954827/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/954827/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/954827/2"}}},"commit":{"parents":[{"commit":"9cd6757168c81c3717e26f1859630ab3e1c549e0","subject":"[stable-only] Fix rolling-upgrade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9cd6757168c81c3717e26f1859630ab3e1c549e0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-07-11 23:44:35.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-07-12 01:00:32.000000000","tz":-420},"subject":"CHANGELOG for 2.34.1","message":"CHANGELOG for 2.34.1\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ica2e1cb018106164c6cea0e6e8e9c6bce32d416b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f776b6ee0235544dc46f0750e0c67f0a448b6a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f776b6ee0235544dc46f0750e0c67f0a448b6a2"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
