)]}'
{"id":"openstack%2Fswift~954900","triplet_id":"openstack%2Fswift~master~I863a27dca9f030cf7c876131c80fa6e8e101bd36","project":"openstack/swift","branch":"master","topic":"otel_tracing_decorator","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-07-14 08:23:00.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I863a27dca9f030cf7c876131c80fa6e8e101bd36","subject":"Instrument get_hashes with request tracing","status":"NEW","created":"2025-07-14 07:25:46.000000000","updated":"2025-07-14 17:05:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":9,"unresolved_comment_count":8,"has_review_started":true,"meta_rev_id":"da1f9a57d164e85dbc77a9fd6e864d73ee13455c","_number":954900,"virtual_id_number":954900,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-07-14 08:23:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-14 07:44:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-14 08:23:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-14 17:05:42.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"fe33d07d0c69e6474651e3ff3678dac050e083e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-07-14 07:25:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dcdc323d089b6aa5792b232e864877a157d5d7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 07:44:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/436f81eb379f4c39a2303d73083a04ec\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7eb7dddecc6a4ef283fa0a9e22caa5a3 : FAILURE in 9m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f0aee6531f449d00c333004f0e4aab81da456e76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 08:23:00.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b2479df1bfd40569f53bf37aeaa56ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a53fef2f12c84843a1a50d2ba8b0d82b : FAILURE in 5m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/db682a6af1c54ed2b0d1503d58f06541 : FAILURE in 50m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f77d3ea93ff349b19d207c543b2ee6d1 : FAILURE in 50m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a1bac021f4634c7596f396794993436f : FAILURE in 21m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e5676e438aa4246b43da2f0ce6cdf96 : POST_FAILURE in 22m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53e5a6ae455e4d9b81abe2cdb32047f5 : FAILURE in 7m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3216c1310e4641bc90278221dc4cc8dd : FAILURE in 5m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/233a24b77fe840a093718e359c929f16 : FAILURE in 7m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/caea8203fd83456fb220dac006fa5b57 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/732cc94103694ababfe227c6e0abb00f : FAILURE in 5m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/de8710c5259142e49697b8d1de07c281 : FAILURE in 5m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b8d14b59957d4c19b991b1a120ec0704 : FAILURE in 10m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/622d4061b8714d18bbab2d3ed1dd846b : FAILURE in 9m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8544006214ea4d4280e8963fc1f669a9 : FAILURE in 7m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bb94418d17524b279e06a38e43785de9 : FAILURE in 2m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/db830a55f8fa44a8980c8e539af82f33 : FAILURE in 3m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a26bbb6a3c44da1a7912bf7639b468a : FAILURE in 5m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7fb73e9faa2c40ec8a9e75fad7cbdc40 : RETRY_LIMIT in 4m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5017506162ad487e8f3e46f62ee6a058 : RETRY_LIMIT in 2m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/15a67e610501426f91d3268e647a8da8 : FAILURE in 2m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0cf49de360fc4601bbf72bb3d97b2b99 : FAILURE in 18m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/905f40c19c924c8db5d24cd29ae2eb46 : FAILURE in 13m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a374a2fba863479e841e2c8bd23f96f3 : FAILURE in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d69a222bd2404e6c951150b113d01bf6 : FAILURE in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d37f4c15dd394bc8846a0aff93a15f61 : FAILURE in 3m 45s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07b55ddb85ec4b0298cbc77aa1eff533 : FAILURE in 23m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad13323faa2246278094ee36259b28fc : FAILURE in 23m 56s","accounts_in_message":[],"_revision_number":1},{"id":"da1f9a57d164e85dbc77a9fd6e864d73ee13455c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-14 17:05:42.000000000","message":"Patch Set 1:\n\n(9 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"762d59b5864c69b31a87202b0fa9b91a69eef8d8","revisions":{"762d59b5864c69b31a87202b0fa9b91a69eef8d8":{"kind":"REWORK","_number":1,"created":"2025-07-14 07:25:46.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/00/954900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/954900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/954900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/954900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/954900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/954900/1"}}},"commit":{"parents":[{"commit":"5b7daef87bbf30b91814d9819babe438ee09583b","subject":"Tracing: Add overview documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b7daef87bbf30b91814d9819babe438ee09583b"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-14 07:20:38.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-14 07:25:18.000000000","tz":600},"subject":"Instrument get_hashes with request tracing","message":"Instrument get_hashes with request tracing\n\nThis starts to open up new questions like. How should me manage tracing\nof non wsgi apps in swift, like the replicator or reconstructor?\n\nMaybe they just build the tracer and store in a requests env dict as\nwell? which we do because of eventlet, and some non wsgi apps also use\neventlet.\n\nChange-Id: I863a27dca9f030cf7c876131c80fa6e8e101bd36\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/762d59b5864c69b31a87202b0fa9b91a69eef8d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/762d59b5864c69b31a87202b0fa9b91a69eef8d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
