)]}'
{"id":"openstack%2Fswift~955358","triplet_id":"openstack%2Fswift~master~Ibb20a3cae3db4ed55c2f686164bc28bcd7fe7397","project":"openstack/swift","branch":"master","topic":"bug/2117241","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-07-21 00:35:28.000000000","reason":"\u003cGERRIT_ACCOUNT_36393\u003e replied on the change","reason_account":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"}}},"removed_from_attention_set":{"36393":{"account":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"last_update":"2025-07-21 03:50:43.000000000","reason":"\u003cGERRIT_ACCOUNT_36393\u003e replied on the change","reason_account":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"}}},"hashtags":[],"change_id":"Ibb20a3cae3db4ed55c2f686164bc28bcd7fe7397","subject":"Fix logging storage policy index for staticweb requests","status":"NEW","created":"2025-07-18 10:22:02.000000000","updated":"2025-07-21 07:20:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":15,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"145833714164543dd677b70487f11d6054654e89","_number":955358,"virtual_id_number":955358,"owner":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-21 07:18:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-18 11:41:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-20 23:35:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"d1c8a1a38216bd69e7f40f389e301858d9768063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-07-18 10:22:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"465b3946b170d5f0be2c8cc85508aa9c9b734961","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-07-18 10:24:53.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5e4f0481d3882754c1069d865892e23c0bf0da0","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-07-18 10:28:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47f4be34a7fc751dbab66a6bdb53a909f872cff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 11:41:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac0812ee68a647bebab8b5825b3fdd4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4d3e7dbcfb46f9888544a9f94529b5 : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/ebcc40a1aafa4b4f8b9eef3593f5f752 : SUCCESS in 47m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a7e8c15195834ce7bfafee56d025bd50 : SUCCESS in 1h 10m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/506c2fb4babb40b698c7820e1428c600 : SUCCESS in 55m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8a18b0b875a478198c8329a801505a1 : SUCCESS in 52m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3579f0d2e004cac94c6655527b7e655 : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a83fde8a2ff24b0cbd2667211bef10c9 : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8daf775cd84e4d81a63b13bf943527cd : SUCCESS in 24m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b9e52ee6a4e2489eb7da4bd1244f117a : SUCCESS in 4m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ce30b30dc06d458f949832bd14131b50 : SUCCESS in 21m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f465ce98c47c45dfa44f1e897dda4563 : SUCCESS in 18m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d3ef09e49c07426cad8933d6a6dda7b2 : SUCCESS in 27m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f858906c062b465cb84de584a170545d : SUCCESS in 30m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/61b0e50a2b334373aea79223719da309 : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/04b4aae34f394ee79958ca9ff0195375 : SUCCESS in 14m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2788cacd818b493285f163968bca97a8 : SUCCESS in 12m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ed0c59b2edb49318eab348d32f9fa63 : SUCCESS in 8m 08s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cd6363c6336e4a6f82dd56de2ed46f0e : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c4c4238e7a4a401693acc42a97cbe169 : SUCCESS in 7m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/19d9b75b196d4b9abac48f5a04295e8a : SUCCESS in 57m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d7ca1e5ff6e04397b7c570c3886e37b3 : SUCCESS in 54m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3c36302a3e824f8d8f45ab731262e3d7 : SUCCESS in 49m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31c7bd3c295d4b019a3afbb6d7992190 : SUCCESS in 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6306a917ea2f4849a3b07129d66b5fbe : SUCCESS in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf0691c9bdd542559f4942d55884db63 : FAILURE in 25m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6ec74763d20408fa2b8458f04ba861d : SUCCESS in 43m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95b92eddb18f43438158c2ac0c4f0f37 : SUCCESS in 46m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6b12c751f04672c7119c80da78e1ee831da197c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 12:22:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f4edf0a6a574939a9d1a238ec940bdd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e4b3db6620b404c865f2cbc1fbc0b13 : SUCCESS in 1h 50m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ee60bb29bb66c5b914c7812d9ab4564c625c292e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-07-20 22:56:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ab7eb4b8f03994b72b6292d5d5b9424c0be85b3f","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-07-20 23:01:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a88d2f3c6aea822f905eb46b9ac422cc0ed4677b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-20 23:35:03.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e3cef76756ec2ecdf0ff727ecf95e63d28343f8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-20 23:57:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9e2100681954fdc9d4396e552a6add4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c45ea31d29ea444f8f7d69038b6555d4 : SUCCESS in 59m 11s","accounts_in_message":[],"_revision_number":2},{"id":"591d64ca441d6ec0bec67b6c84303c096e31c51c","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-07-21 00:35:28.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d4afca77b264ff624a97c93fbbf844b1631e742a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 02:03:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/465d55eb9e284712ad855e1f72d2d07b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5400f6d02ffe4648957bd7aa57d1905d : SUCCESS in 5m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/a7b6108fed8b47df865c58167e9cea53 : SUCCESS in 1h 00m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c07fe700a9bb417da472d0c7f111bc35 : TIMED_OUT in 3h 02m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cffec854081e4e0492b4c32960f0a905 : SUCCESS in 49m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7133769ea163415f8b5159b467e3d619 : SUCCESS in 46m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff78de39cb444b979f3d15cd8c390603 : SUCCESS in 12m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5bd98d0e5adb4b6690ef451d55572f7e : SUCCESS in 15m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e85790c1aa0a4c3db1d9c6e1166c0ff3 : SUCCESS in 15m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4cacccc3b9414c9a8e28f6d79fc40c8f : SUCCESS in 4m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/67db4ff2c92b4cc3a70afb184e182af3 : SUCCESS in 20m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fa03a916942a4d96a10c8a1ddced4143 : SUCCESS in 7m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b952cbf076674966a9a697e5ee092717 : SUCCESS in 22m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d0dbb37f53474bffa9299db11d5bdd07 : SUCCESS in 27m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1f34b11aa5a04408aa4964c72b21587f : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ecf27f4ea30d42d7810912873292da78 : SUCCESS in 14m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6370197579bb459ea6ad619cc7717258 : SUCCESS in 14m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/43aa4f9b765646d7887d42dc5163be53 : SUCCESS in 8m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3c8670625e074c5d8dac463411d0a2b0 : SUCCESS in 13m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f55b2e14e41544e18d58adc892c6606c : SUCCESS in 10m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a833244562da421cadd9b63cd935b64a : SUCCESS in 1h 10m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4ba54dfb74b4622a1c8cc2f34b7cc8d : SUCCESS in 20m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c079d40a2afd48e0a72022a9acc2010a : SUCCESS in 21m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f032a3307464d0b97cb31dcf25e274f : SUCCESS in 20m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ff8963e00e45f988c9a5cfd9feb5ab : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/922a57cabc164ea9a43c5b89be4d86ac : SUCCESS in 24m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c11ea3280e04e0aa04ef3f2c7b4413d : SUCCESS in 54m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ef730cfc8214d95af569cc66375f800 : SUCCESS in 28m 24s","accounts_in_message":[],"_revision_number":2},{"id":"018d891d8ca04773251dfa502943e778fdc33ea2","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-07-21 03:50:43.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"11312fdb0cd6a2b76a90fffca5e11ebf8b4a7667","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-07-21 05:57:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c4add650e61ef13ee2f57bb83043e240a68b681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 07:18:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e26e6267a10486f800778dfeb8cdf69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3c25eb64c6b41dc889e63f930588311 : SUCCESS in 4m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/4f88fa8e0a4e4283935622e34f8abee2 : SUCCESS in 59m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/48f9fc9f9e1041379a44061af7750891 : SUCCESS in 44m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/545bf89fba33497e9a5030ba2a734bd6 : SUCCESS in 38m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35a50de3326846458a8de179ba5ec4ba : SUCCESS in 51m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27509aba24a64bb487e3e0cd08b0c632 : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/761b71ef5f034775b92cba4a93acafc2 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf1edb3725e54aeeb63f0ccc139bd95c : SUCCESS in 14m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf2d9ede50614ae1b5b801c927877b5b : SUCCESS in 4m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/71bd2e0d89514cf7ac6d54a3df311a7b : SUCCESS in 20m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e3696c75bc4e44a999c313390ab1faa4 : SUCCESS in 18m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/737ef362261d4ce2b1078039292330dc : SUCCESS in 17m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/eb809c903a804aab9e0fa69daf129eb2 : SUCCESS in 22m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7168ce8ee3b04aadacf47ee119b3393b : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cf678eb67ac845909fe6fcb1df5e5ba4 : SUCCESS in 14m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fd641d1325994c49b50159a65fad525d : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7f5fd071a00443708eca380bcbbe876a : SUCCESS in 7m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b42527ac5e64961b5ebbe50d14e4b79 : SUCCESS in 14m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/605136e4732148658b41d8dd5c4e4812 : SUCCESS in 5m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa2ff54fd07643b589c0cb12234f57a1 : SUCCESS in 45m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d71cffe74973444791785768fbb1ef5f : SUCCESS in 49m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a19f14a9352349718774c63c959399f9 : SUCCESS in 49m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31848e33721c4b1c804b62aca12688de : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa1d1a22eadb486ab370d067a866da41 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b2019c1b101e4e448be55fec5b17c573 : SUCCESS in 26m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2214a3df8db4cfa9b57ec205c9d69ae : SUCCESS in 27m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48f114e150d44051bab6cd021bbb2ba9 : SUCCESS in 1h 02m 03s","accounts_in_message":[],"_revision_number":2},{"id":"145833714164543dd677b70487f11d6054654e89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 07:20:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eee6b1389d54e0792fa1eab1804ce0d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/259a53c5ad7a4e57992ccd923056a77f : SUCCESS in 1h 21m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"16b0121c4b3973135a3aa6b4c30faa68f192963a","revisions":{"ca2f2bde0774ed888af27b702fac69f9ec17eb09":{"kind":"REWORK","_number":1,"created":"2025-07-18 10:22:02.000000000","uploader":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"ref":"refs/changes/58/955358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/955358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/955358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/955358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/955358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/955358/1"}}},"commit":{"parents":[{"commit":"8af485775acc9ad97a6cbdca64279aac9f647f8c","subject":"Merge \"s3api: Add support for crc64nvme checksum calculation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8af485775acc9ad97a6cbdca64279aac9f647f8c"}]}],"author":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-07-18 03:48:47.000000000","tz":720},"committer":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-07-18 10:20:43.000000000","tz":720},"subject":"Fix logging storage policy index for staticweb requests","message":"Fix logging storage policy index for staticweb requests\n\nThe storage policy of a container is not currently set for requests\nhandled by the staticweb middleware in Swift Proxy Server access log\nlines.\n\nFix this by adding X-Backend-Storage-Policy-Index to the response\nheaders for the requests generated by the middleware.\nX-Backend-Storage-Policy-Index does not actually get sent out to\nthe client as it is removed by the gatekeeper middleware.\n\nCloses-Bug: #2117241\nSigned-off-by: Callum Dickinson \u003ccallum.dickinson@catalystcloud.nz\u003e\nChange-Id: Ibb20a3cae3db4ed55c2f686164bc28bcd7fe7397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca2f2bde0774ed888af27b702fac69f9ec17eb09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca2f2bde0774ed888af27b702fac69f9ec17eb09"}]},"branch":"refs/heads/master"},"16b0121c4b3973135a3aa6b4c30faa68f192963a":{"kind":"REWORK","_number":2,"created":"2025-07-20 22:56:19.000000000","uploader":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"ref":"refs/changes/58/955358/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/955358/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/955358/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/955358/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/955358/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/955358/2"}}},"commit":{"parents":[{"commit":"8af485775acc9ad97a6cbdca64279aac9f647f8c","subject":"Merge \"s3api: Add support for crc64nvme checksum calculation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8af485775acc9ad97a6cbdca64279aac9f647f8c"}]}],"author":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-07-18 03:48:47.000000000","tz":720},"committer":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-07-20 22:52:47.000000000","tz":720},"subject":"Fix logging storage policy index for staticweb requests","message":"Fix logging storage policy index for staticweb requests\n\nThe storage policy of a container is not currently set for requests\nhandled by the staticweb middleware in Swift Proxy Server access log\nlines.\n\nFix this by adding X-Backend-Storage-Policy-Index as a request header\nwithin the WSGI environment. This allows get_policy_index to find\nand set the policy index in the access log line in a way that doesn\u0027t\nrequire applying X-Backend-Storage-Policy-Index as a response header\nto all possible responses generated by the staticweb middleware.\n\nCloses-Bug: #2117241\nSigned-off-by: Callum Dickinson \u003ccallum.dickinson@catalystcloud.nz\u003e\nChange-Id: Ibb20a3cae3db4ed55c2f686164bc28bcd7fe7397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16b0121c4b3973135a3aa6b4c30faa68f192963a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16b0121c4b3973135a3aa6b4c30faa68f192963a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
