)]}'
{"id":"openstack%2Fswift~955452","triplet_id":"openstack%2Fswift~master~I293bdfe69f10573ef7b3440cd31dd67f22b85f67","project":"openstack/swift","branch":"master","topic":"ringv2_history_chain","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-08-21 08:33:00.000000000","reason":"A robot voted negatively on a label"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-06-01 10:29:47.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I293bdfe69f10573ef7b3440cd31dd67f22b85f67","subject":"ringbuilder cli: add set_history_count command","status":"NEW","created":"2025-07-21 07:11:30.000000000","updated":"2026-06-01 10:29:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0b8b940088126146c9a0b1162d670ec9de13baf7","_number":955452,"virtual_id_number":955452,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-01 10:29:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-21 08:19:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0bfca22234d8af2d6aed1d692482771f9349bcb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-07-21 07:11:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fa9a49a7822a1cc708c2527e0d94acc8d9b7759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 08:19:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/501fbb26e9444ed09dfaac0f7b46d426\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b3683ff778465fa4c1b67545193d8d : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/ae0627be2be04bf1bad6018f7bec23d5 : SUCCESS in 55m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3f5e61f30a1a4aa4b63c6ac62817bf4c : SUCCESS in 1h 02m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ce4953a3381b45de82fb4399e64e1131 : SUCCESS in 53m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2dc5dc941774947af2cdc752b2e6619 : SUCCESS in 57m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/686b81173c6c4b82a3501b3e599c4c01 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88c928f88ce845f5827fa2db149c60b6 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72fb940e074e46b5ba581ee8b0f9cd98 : SUCCESS in 20m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/256999853cb34ebb987252cc88a75bc3 : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3992ea33922a48b2824f8813f8820fa0 : SUCCESS in 19m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/55956de3db2745ea9849f1962e80c094 : SUCCESS in 19m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a53c3c5a132f45429e119ce155ddafba : SUCCESS in 27m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d8fb783bb20d4775a5e82e85848b7b6f : SUCCESS in 28m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5e4cb86dd9f2494c9f23c0a3082703af : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9bff838a288c4f208909ce0c133c2248 : SUCCESS in 14m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ee182bb6fe0444328e53f6c72dca1755 : SUCCESS in 14m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/539df4a2116948298c417b82a23634ba : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd3a40c01e184622bd6f1e978407be32 : SUCCESS in 14m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c9697b797b24e81a86fe75f7019b3c7 : SUCCESS in 7m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/420df6983da34b0b90bb3af1c868cb83 : SUCCESS in 37m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e807044b8db249c984484229e86686b1 : SUCCESS in 43m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a99cf4c8743b4f8d9991cc3ebff874d4 : SUCCESS in 56m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19fc88d43c814715b4efea88421eae1e : SUCCESS in 19m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56441bd93ee5498890c50cdbe7d3ddee : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7679b96549f45059bf64e611309548b : SUCCESS in 16m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1892b3b1572e4ed480b8df2ffafcada9 : SUCCESS in 48m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/729475dbc94a4cecae0a4ea06af1bbb0 : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e772abf4571550581e9dd12367c9c36d09240d8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 08:48:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc8ab4a2d1a04f62a8f3ea9148154c12\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b60e393d63044dab0910bafd83f87a8 : SUCCESS in 1h 34m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4319fa63ac5d007d2422eed3a08698a777dca156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-07-22 07:01:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9de6ed2a4fb1820944ca7604684405019e6d1e39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 08:11:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5222a98415e4335a495895214a2e31f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a50323649b774b68ada510f3732b874e : SUCCESS in 1h 07m 44s","accounts_in_message":[],"_revision_number":2},{"id":"44e98258b0d1c8e898c66e31b14947ea9c2f8750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 08:12:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1fe7dc0054548969bea5cc890a83d07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b048e988b754e31809110bd13ed3ade : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/8c35c850e9e643b98360ec4ae0d79b28 : SUCCESS in 59m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e499342cb33f4add80fb3a48e8cd48ea : SUCCESS in 56m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a5ef24d3e5684e6cb3168406ab941fd8 : SUCCESS in 49m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8df6b82304104ee5a292c419cb08a179 : SUCCESS in 58m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6da93b23d40464681970bc869cf7f5e : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/047a531314cf47d2af75b95990780272 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9e05e3ae850440ab78a8840c6a3408e : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba4a5a6798034d0e84cb5155a87c9c57 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/44566df63536414f9e6df955c34d6d85 : SUCCESS in 20m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/141a3633215747a8aaf5534d395a281a : SUCCESS in 19m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/66ea8c6d4bca409295a782fb781fcb91 : SUCCESS in 17m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3b0a3bd060d343918223eaacf3d90624 : SUCCESS in 27m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c353c63876da480db164db79103d9271 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9816ec540c18401bb1e33c736bde0cd4 : SUCCESS in 12m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c632b8ecfc0842f8937e3ec4099196a8 : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/119e05fca19941a8ab62a12495a7df50 : SUCCESS in 8m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/341efd1b976e4378905dd160dd9344bb : SUCCESS in 13m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/189e31f11bfc4f86a474afa45140bdee : SUCCESS in 7m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5eb41a3903564bb5ba640813f5d19689 : SUCCESS in 53m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c16ee6dd4ae44809b21154d293cc2bd2 : SUCCESS in 47m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0c53c7633dd4449587cbd42f29857171 : SUCCESS in 45m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65189a038ee14378a714d8b44576e2fc : SUCCESS in 20m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/364e68ed4da7499f9bdfe90273f0b6e9 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d00b10fae2446afab0ef7f7c9a89306 : SUCCESS in 17m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6beaa0d2e9e94135ad6024db24a8dd7b : SUCCESS in 52m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/051585483399487a99ac61b88b38752a : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ec920c141ff7987883d351bf45fe7f6cb96d7b0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-08-13 22:59:15.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d15f7158bd53f0f0e195eb4da0fc03cbf09fdfe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-08-13 23:02:20.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8622dc2fb6cdc6332b3a729517c71263309292bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 00:40:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2544713ee8194f1cb65f50058ef02d88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e66bdb35eff42448fc6d7812cf03f82 : SUCCESS in 4m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/4c48a82cde594a258a7e201a96c9afd8 : SUCCESS in 28m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d0d64a21104c4b768f5dd7c82787151f : SUCCESS in 1h 02m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fdfaf72a2bc94bafa6f06e0c2f3396b6 : SUCCESS in 56m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30b8c8ee05ea440980cc6ee54d5d74e0 : SUCCESS in 46m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56c8004cadca45fd8620b2b4fd388e76 : SUCCESS in 15m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de7b473edb8d43209824757ad71e6142 : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc6e8f0012ac449ba925961d539e4f11 : SUCCESS in 16m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/62b33bb2aea141c6a0ad9aa3ee8a19ea : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6f5b740925d64f5084b7e102085cfb1c : SUCCESS in 14m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e93d5b38f3d492ba77f0359e594f91e : SUCCESS in 14m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/af959d07614940589040db5b0a1519ee : SUCCESS in 28m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1ca04fbdaf81402c9f4c1c441381c503 : SUCCESS in 23m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c1b9d61e83904dce82b705ea6eb4c46e : SUCCESS in 12m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/648b63ae83c64aaf94a3f30e81c053e8 : SUCCESS in 14m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9f2a2a9710be481884fae658c4e29e07 : SUCCESS in 14m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bcd998736b9944979f15f9291ad245e5 : SUCCESS in 8m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/179f3a4f0c4f4f61af1645d81d7bc1de : SUCCESS in 12m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9903d27089cd48ea8d8fc7da11cdfef9 : SUCCESS in 4m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa3e2f13116641b59536346f8e781579 : SUCCESS in 43m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e208568617bf4e87b07738764274b455 : SUCCESS in 35m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95e845f238ee45c39ffbc56d724d260c : SUCCESS in 50m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddb521665d1b453da2d4dbdcdc7d2f72 : SUCCESS in 19m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feaafc8cf5674760abc24abb63c71826 : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4414170a69bd418b99b07f2adbfe3fae : SUCCESS in 24m 39s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68f8895e341b42e58a4018ed9223e2df : SUCCESS in 49m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95406fa5243f4cc3b806d0d17cb8c080 : SUCCESS in 1h 11m 22s","accounts_in_message":[],"_revision_number":4},{"id":"862a34232c9d36a05cdaa61dacfffcab7920bf5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 00:48:55.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8795d60d4db2490aa11011e6cefa1da3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/97c89ac67ed94d0ea149a2109b22216f : FAILURE in 1h 40m 03s","accounts_in_message":[],"_revision_number":4},{"id":"f74e578e7b7f4f2571ebe9becbb55bd277f0fd22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-08-21 07:23:43.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"b2a608fb108bee44ae935695c70645fa46d0bb3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 08:33:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41bf03050e5c45f590ebae7bb51640e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b46f63dc8db9409ab456883159cb282f : SUCCESS in 11m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/08a8c4da67354131a2a458ef121dbab4 : SUCCESS in 42m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d95127ae4314271931097fa66d2eac9 : SUCCESS in 1h 03m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/292a8c08dd9b4de69f2119aa0161f46c : SUCCESS in 42m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7158a5276d748f19a0e75895fdc9030 : SUCCESS in 36m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f651a59237ab4183bbab4086b39472e2 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af974bb435804dada188098adb199fe0 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/517aa210862245f28f1398c830a2d080 : SUCCESS in 20m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bd725ed85619417489b3c06a2d94abeb : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d8ee151f251d4b63acffa29d1105e4ea : FAILURE in 24m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b2da210d52f94b4a87a491560a9074d9 : FAILURE in 8m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b1c7a4a66abd4306ab35ea5fea96ab1b : FAILURE in 28m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/28f9859b79524a55bc7460ab1c0ca1f3 : FAILURE in 28m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1d5cfd315a7c440baf325b63374c0d10 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ae1c548be1b54542bfd3c4eff0cc6c23 : SUCCESS in 14m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4bb6e8ebb4e24109ac7be0cc10769891 : SUCCESS in 14m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4bcdeaebc39442da81264aa176f55390 : SUCCESS in 4m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76628a173e654ae8914126f68f075cb7 : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a98c39ab23764bf4805a3e2eab216255 : SUCCESS in 5m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1501523accd8453bbcaa1ac515ff7d5f : SUCCESS in 47m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b017f56824ee4f4e8333443140d40634 : SUCCESS in 50m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b15b3551fda54515a00769231e8518e0 : SUCCESS in 52m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20182db25ca145daa9424a2bfbd1ac14 : FAILURE in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9479a1819251484da572e4af3a01b3ea : SUCCESS in 2m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd50277b215e466dbef1229fcc9385da : SUCCESS in 16m 45s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bc83cb966dd4790aa1056474d6769cc : SUCCESS in 50m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32d820da1ad54cc594d7d1baa57ee847 : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":5},{"id":"d70bfd8de52f1918d636f46f42c501bd5f616801","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 09:36:51.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56e1e38f4ed940648d46d130d1222d3d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/310b845919fa4686a88686e78f5adcfc : FAILURE in 1h 59m 09s","accounts_in_message":[],"_revision_number":5},{"id":"6dd50b0a624c940c21ddf1b6726eb49be9f59a2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-08-27 07:54:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"27bf4e14316703d3769bd68698b97681d44d653c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 09:13:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c92c314921264c90864b919267a410a8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c5618754cb640debc8233ea14b030ea : SUCCESS in 1h 07m 01s","accounts_in_message":[],"_revision_number":6},{"id":"90d540bf4c175f7a82f121eb7559e22610841a00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 10:00:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89c483fa41ea4ebf91de3e5affa2d697\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d52529dd61d48a4bdd11e45da0da6e2 : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/efa3dfcf0f944298af964ba060c5cd10 : SUCCESS in 59m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e1d0286518fa4f03b854d53e741215a0 : SUCCESS in 1h 02m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a4bf70516e5d4ca197904bd0a19cceff : SUCCESS in 52m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdf09a01611343c8bb83420690b78293 : SUCCESS in 1h 03m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc0b76ac92c4472597887b8920095ca8 : SUCCESS in 16m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab81308651e84208a9f0fdf1e8430f16 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9eb667d65e694dbaa73cc6986943326f : SUCCESS in 20m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fe720c9b74fe4acdbed9dd7c5e57e991 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dcb5d2adff244fce8c369e8e71d35729 : SUCCESS in 28m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c583e6e1257443cebfb1f1c7d2988103 : SUCCESS in 8m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/502e45ae99664e0a92517776af5caed4 : SUCCESS in 10m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/73af5151bfde4d88861db53e55829457 : SUCCESS in 24m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fa6acb31b8b6405b84b1ad33d87136df : SUCCESS in 9m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a24d4278c28044d296c7dbfad72d25e3 : SUCCESS in 14m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6be27105b5f14c1ea13d83d654ed89df : SUCCESS in 15m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9916bb2f7a3f4c48b0f9a88cc6d413b6 : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4b3f55c4fa9f4b249c6adf7f6526d9ac : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2429ff6cc3074f4794a5804ea1927de7 : SUCCESS in 6m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cc4c76603aa4019a20e6c7e7b54ea8e : TIMED_OUT in 2h 01m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e76f6d62950d4e71bbf14d95a246f720 : SUCCESS in 22m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c6f0f29f1134074be1c44de365e020b : SUCCESS in 48m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90a7546a13a246458121daad9928ca1d : SUCCESS in 28m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/256f9074b6fb4466b8ca634552969a90 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/54c12cbce1ea44808733978959fbb0e4 : SUCCESS in 18m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/157691f0e376484ba4938d79cb902287 : SUCCESS in 49m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/078141c461444ee9bfa7114095f15c64 : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":6},{"id":"a7b4172d62b89a7461707e135ab3df228aedc600","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-18 01:45:36.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"191914c7e3de8c5222227a84197e79a2fc3daf7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 03:48:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f49388b90f947c6b2d76d211cd502be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/736f7253f005417490bdb359ee73cf93 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/77d8b587366d4f5c8f6e84ea5297fe5e : SUCCESS in 57m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5da8379521fc45d6b8876e11503216bd : SUCCESS in 1h 08m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1222db400fba4646b14dca2e7da1e858 : SUCCESS in 1h 02m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0f2dc0f109245ba9343c6f7f1ed15bd : SUCCESS in 1h 04m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75113c03cf0244e6b3e3f1007b09e14f : SUCCESS in 16m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99392caca7ab446cac811fbb62195f99 : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a04ad5e44634abeba0258bc61ea7e1c : SUCCESS in 22m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/07e2ca35b746411b95806383852f8499 : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e6cfee937e734b239b486c304974471e : SUCCESS in 11m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/56d194c1d2a54678b930cbab6e5d026f : SUCCESS in 19m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9178d02b72074e8581ec802119e64d24 : SUCCESS in 29m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d0388513f8604b43a6b6fc1831898a74 : SUCCESS in 14m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7e148a94b0b84a9aaa760b87841e91b7 : SUCCESS in 8m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6c0cb4d38a7f4edc8fdee0ccb3514b65 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/04a3b64a92d14c59ba3b95f9ffd76843 : SUCCESS in 14m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6a755a6d441843ac99030dc4f76adc64 : SUCCESS in 4m 17s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab395ccfb4164fb4bfcbce478803bbe2 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5eed3d7b8f6d4c4996cb6b6eab102ccd : SUCCESS in 8m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a79d5c7339d481298805562cbbc0673 : TIMED_OUT in 2h 00m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9969798e96243f08249c10769ca00f6 : SUCCESS in 51m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/212a500da9354d7fb26eeeebee47f02a : SUCCESS in 27m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a799caa10b2f48f3a958e0d14585349f : SUCCESS in 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac2bb783d22411f9f43a506bba693c6 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c9f88589e4546389db168adf7cd3afc : SUCCESS in 15m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc33015812654894b46f6a854bd99dfc : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":7},{"id":"f86277feeb94034e911a7406ff1b0944411c562d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 04:25:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/449067aa28f94fe8a7577190c44aa2dd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/18cd40ce19b6409dad728856be75f156 : SUCCESS in 2h 37m 26s","accounts_in_message":[],"_revision_number":7},{"id":"25fcc8e3eb151e2e8791712e5d3c1bbe2831ad64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-04-07 05:01:51.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"34e7c252665a2c9cfbfac6032b5083d41229a17e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 05:31:02.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07385000318b4afd8b37f06f7a588291\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3e7f21e865445afb840a1cd3910b53e : SUCCESS in 27m 05s","accounts_in_message":[],"_revision_number":8},{"id":"228d6b8ab9a6a2c2f8f02ec89a926fe38d89416b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 07:11:12.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74452bce06c543a3be0d0a10c5c8a4bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e881d3e841334f6f9fa8f75b6fdcefc5 : FAILURE in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/e26387ef88b04aa0a42972fd1fc922ab : SUCCESS in 1h 03m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1944fbc7f2944a668933c7ba69ce941b : SUCCESS in 59m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d3546a030f64e0aabe01f4c3ca5a103 : SUCCESS in 59m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a0743423aef470ea803ffbfa73b1d69 : SUCCESS in 59m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c3ac8694aa34b4989b1e4aef42e29a0 : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/686473def6a84e44969b3df4cb92d959 : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ca713f24cd3436491c6cbe86e878574 : SUCCESS in 20m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/99187d86978943a4808e6069cc035782 : FAILURE in 4m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/451376aa5e0c47cab636acea14657c6b : SUCCESS in 25m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bfaaf84979c44d6781c5c01cfcb2e045 : SUCCESS in 19m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5e68138028084b7ea77abed16689afe4 : SUCCESS in 28m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f770465ab50142f0ac60ca4c6ab4e65b : SUCCESS in 20m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6e0bd00db1874ff893c0e31c64c5eda1 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/49351f73737742d094f57d1880a190a4 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4912f5691e814d5bac4723295eef01f3 : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bf57ca7a86754b46acc6c3dd45924352 : SUCCESS in 6m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/770e722b7bc849fbb86a5e3686a43811 : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a1521cc5cbf44fdb00e65a8a8e00354 : SUCCESS in 8m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/58907507e9154d92a7cede77dd99f923 : TIMED_OUT in 2h 00m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/044ee15fcea94d4886d9459b868baf55 : SUCCESS in 51m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8851e84cc326458983c5fac9ac0d17e5 : SUCCESS in 52m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0900a53bf604e42945c5ee270aee86c : SUCCESS in 27m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17b700f0a5eb43108f669ce8a1fbda9e : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6687b202ae434097836f44b36279d4fa : SUCCESS in 19m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a995817ba7ba44c7af4c8242f2fa427f : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":8},{"id":"6bb29771a532f997954ade09344e669da6bfb7be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-01 08:21:52.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"b31cc1f88f0a90b87b080bb5aa8cce8c586fab8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 09:45:48.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd488e538f5b42ebafa02516e733b654\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/06e946b9e62245f58cb95dccceaea1f5 : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":9},{"id":"0b8b940088126146c9a0b1162d670ec9de13baf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 10:29:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7cd7f739a72439ab3854d47018ce7c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c1307cebe94a468afd4f68c943d385 : FAILURE in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d43630bed64467d8bed357aa61c5234 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/1db9fc158e8a4717bac965f9afdf738a : SUCCESS in 50m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c55043ea915d48aabb8a64bc92920a24 : SUCCESS in 1h 02m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d560e169c314433a4ac3414b09822df : SUCCESS in 43m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e66a52571f0f4c49aab2a096ba82808f : SUCCESS in 1h 08m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4a53801c43443c1a25c825cdfe15789 : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7e06cc4a069409db4000b1cb91e0b1b : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a205feabe2ff41ffbf19074e235a4e09 : SUCCESS in 11m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e4131189748647a68d0ce8c1acae0d18 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c2593045a5c341e2ade0e728571ba014 : SUCCESS in 24m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b1bd7960351e497daa230afcfff1f308 : SUCCESS in 23m 27s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c64a60da1a3346ef9616675264afda96 : SUCCESS in 22m 55s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8d7132ec356d45368f55b762c2e74a63 : SUCCESS in 22m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b7572c9c29db48e1a21939ce779850a5 : SUCCESS in 8m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ce0f9e616a134ccd9a7588c1e44c6c6e : SUCCESS in 16m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a22e67e337354484bb3b6abb8456f0f1 : SUCCESS in 8m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fb9cf6c4287a49a9ab5a022ecc908ed7 : SUCCESS in 6m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/905502b99461424ba766d77187468ce5 : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/07ef2e408d6647daa55ccb8a9064617b : SUCCESS in 9m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbda8151395b44c697f85d80df98630a : TIMED_OUT in 2h 01m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3cd15b577312463dbfc648f6f6a03506 : SUCCESS in 21m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6510b7b827ac48198c32eb3eb6b6ecba : SUCCESS in 53m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b1f52e31c794650a6c61052565ae6d5 : SUCCESS in 27m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f1688f01f664b4e88bad0223dec8b72 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b8d6ef72be9c4bc3bfabb79b5a90d14c : SUCCESS in 18m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28f84b550460415db2792cd3315a8de7 : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"fa5bd9211b4d5dd02da97a19df2318b08e0149dd","revisions":{"1e73f74db8d7d24ef113f069481aeb108b348e21":{"kind":"REWORK","_number":1,"created":"2025-07-21 07:11:30.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/52/955452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/955452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/955452/1"}}},"commit":{"parents":[{"commit":"fc6500bcf06865a8db24e1e96f2faf7beccccfeb","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc6500bcf06865a8db24e1e96f2faf7beccccfeb"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-21 07:09:34.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-21 07:09:34.000000000","tz":600},"subject":"ringbuilder cli: add set_history_count command","message":"ringbuilder cli: add set_history_count command\n\nThis patch adds a `set_history_count` command to the ringbuilder cli so\nthe default value of 1 can be changed.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I293bdfe69f10573ef7b3440cd31dd67f22b85f67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e73f74db8d7d24ef113f069481aeb108b348e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e73f74db8d7d24ef113f069481aeb108b348e21"}]},"branch":"refs/heads/master"},"65c5bd716816cd4f69b19df7d4b86123b2925f20":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-22 07:01:07.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/52/955452/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/955452/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/955452/2"}}},"commit":{"parents":[{"commit":"2ce49da558095a5839bd94e2b00ce651e300050b","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ce49da558095a5839bd94e2b00ce651e300050b"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-21 07:09:34.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-22 07:00:48.000000000","tz":600},"subject":"ringbuilder cli: add set_history_count command","message":"ringbuilder cli: add set_history_count command\n\nThis patch adds a `set_history_count` command to the ringbuilder cli so\nthe default value of 1 can be changed.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I293bdfe69f10573ef7b3440cd31dd67f22b85f67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65c5bd716816cd4f69b19df7d4b86123b2925f20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65c5bd716816cd4f69b19df7d4b86123b2925f20"}]},"branch":"refs/heads/master"},"926cda0d689bf9894b7e932ba5e07dc391309429":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-08-13 22:59:15.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/52/955452/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/955452/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/955452/3"}}},"commit":{"parents":[{"commit":"edc883c9b2fd5fe90cb841a68263a5a241ff45d3","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/edc883c9b2fd5fe90cb841a68263a5a241ff45d3"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-21 07:09:34.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-08-13 07:49:26.000000000","tz":600},"subject":"ringbuilder cli: add set_history_count command","message":"ringbuilder cli: add set_history_count command\n\nThis patch adds a `set_history_count` command to the ringbuilder cli so\nthe default value of 1 can be changed.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I293bdfe69f10573ef7b3440cd31dd67f22b85f67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/926cda0d689bf9894b7e932ba5e07dc391309429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/926cda0d689bf9894b7e932ba5e07dc391309429"}]},"branch":"refs/heads/master"},"55828522e8b3825934178596ea3218beae8b8130":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-08-13 23:02:20.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/52/955452/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/955452/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/955452/4"}}},"commit":{"parents":[{"commit":"37f6b36410bcaf6bdbb5bb2a7764f5638430d670","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37f6b36410bcaf6bdbb5bb2a7764f5638430d670"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-21 07:09:34.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-08-13 23:02:04.000000000","tz":600},"subject":"ringbuilder cli: add set_history_count command","message":"ringbuilder cli: add set_history_count command\n\nThis patch adds a `set_history_count` command to the ringbuilder cli so\nthe default value of 1 can be changed.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I293bdfe69f10573ef7b3440cd31dd67f22b85f67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55828522e8b3825934178596ea3218beae8b8130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55828522e8b3825934178596ea3218beae8b8130"}]},"branch":"refs/heads/master"},"abee9663fe0aef0c5ef4ef98c9c817b4fb680d99":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-08-21 07:23:43.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/52/955452/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/955452/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/955452/5"}}},"commit":{"parents":[{"commit":"2de55911a2e2449eb4364f257b50bc02521a5aad","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2de55911a2e2449eb4364f257b50bc02521a5aad"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-21 07:09:34.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-08-21 07:19:16.000000000","tz":600},"subject":"ringbuilder cli: add set_history_count command","message":"ringbuilder cli: add set_history_count command\n\nThis patch adds a `set_history_count` command to the ringbuilder cli so\nthe default value of 1 can be changed.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I293bdfe69f10573ef7b3440cd31dd67f22b85f67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abee9663fe0aef0c5ef4ef98c9c817b4fb680d99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abee9663fe0aef0c5ef4ef98c9c817b4fb680d99"}]},"branch":"refs/heads/master"},"14d4853bf450e08996ffb78f58602d704b6df22d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-08-27 07:54:45.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/52/955452/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/955452/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/955452/6"}}},"commit":{"parents":[{"commit":"801b6e86e465bfcc61e7e3447e53dda78964b850","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/801b6e86e465bfcc61e7e3447e53dda78964b850"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-21 07:09:34.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-08-27 06:44:45.000000000","tz":600},"subject":"ringbuilder cli: add set_history_count command","message":"ringbuilder cli: add set_history_count command\n\nThis patch adds a `set_history_count` command to the ringbuilder cli so\nthe default value of 1 can be changed.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I293bdfe69f10573ef7b3440cd31dd67f22b85f67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14d4853bf450e08996ffb78f58602d704b6df22d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14d4853bf450e08996ffb78f58602d704b6df22d"}]},"branch":"refs/heads/master"},"adf6da530f47744805cc0517be1dd459c5c2a440":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-02-18 01:45:36.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/52/955452/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/955452/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/955452/7"}}},"commit":{"parents":[{"commit":"2b7598f4049c17827cdcb5840b69dd30a2c682be","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b7598f4049c17827cdcb5840b69dd30a2c682be"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-21 07:09:34.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-18 01:37:58.000000000","tz":660},"subject":"ringbuilder cli: add set_history_count command","message":"ringbuilder cli: add set_history_count command\n\nThis patch adds a `set_history_count` command to the ringbuilder cli so\nthe default value of 1 can be changed.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I293bdfe69f10573ef7b3440cd31dd67f22b85f67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adf6da530f47744805cc0517be1dd459c5c2a440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adf6da530f47744805cc0517be1dd459c5c2a440"}]},"branch":"refs/heads/master"},"f8d42caafbedb7ff9abcc1e4481eb86151d7465a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-04-07 05:01:51.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/52/955452/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/955452/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/955452/8"}}},"commit":{"parents":[{"commit":"7a731ea72f362e0cf095f6f87c35b933565e2002","subject":"Add PartHistory class that is a sparse/dense array hybrid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a731ea72f362e0cf095f6f87c35b933565e2002"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-21 07:09:34.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-04-07 04:58:59.000000000","tz":600},"subject":"ringbuilder cli: add set_history_count command","message":"ringbuilder cli: add set_history_count command\n\nThis patch adds a `set_history_count` command to the ringbuilder cli so\nthe default value of 1 can be changed.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I293bdfe69f10573ef7b3440cd31dd67f22b85f67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8d42caafbedb7ff9abcc1e4481eb86151d7465a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8d42caafbedb7ff9abcc1e4481eb86151d7465a"}]},"branch":"refs/heads/master"},"fa5bd9211b4d5dd02da97a19df2318b08e0149dd":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-06-01 08:21:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/52/955452/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/955452/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/955452/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/955452/9"}}},"commit":{"parents":[{"commit":"387de8e7865909259113b7b50856b9b6b22d6be7","subject":"Add PartHistory class that is a sparse/dense array hybrid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/387de8e7865909259113b7b50856b9b6b22d6be7"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-07-21 07:09:34.000000000","tz":600},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-31 14:23:39.000000000","tz":120},"subject":"ringbuilder cli: add set_history_count command","message":"ringbuilder cli: add set_history_count command\n\nThis patch adds a `set_history_count` command to the ringbuilder cli so\nthe default value of 1 can be changed.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I293bdfe69f10573ef7b3440cd31dd67f22b85f67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa5bd9211b4d5dd02da97a19df2318b08e0149dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa5bd9211b4d5dd02da97a19df2318b08e0149dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
