)]}'
{"id":"openstack%2Fswift~956200","triplet_id":"openstack%2Fswift~stable%2F2025.1~Ib0330e863ad15e362dfa587517a26109c74eefec","project":"openstack/swift","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-08-05 19:40:10.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ib0330e863ad15e362dfa587517a26109c74eefec","subject":"s3request: refactor error handling while reading input","status":"MERGED","created":"2025-07-30 20:51:05.000000000","updated":"2025-08-06 23:12:48.000000000","submitted":"2025-08-06 23:12:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956200","meta_rev_id":"521cf0d8f93edfb9160bf40f85b47b0106773942","_number":956200,"virtual_id_number":956200,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 23:12:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-06 21:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-06 21:53:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-30 20:51:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-07-30 22:13:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"402cd77b5507de56cd85ed6a0e9319a51edfa8f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-30 20:51:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4edf508eda220efad61a29cfed01d348b77cc876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 22:13:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a22703de36aa42abb9eb34c11578fccc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/580d6df9415449518d9d1e79a49f426e : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/f9b6e78bfb9c4af8828416a5250aa7b1 : SUCCESS in 57m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/744f5822a7e54edba66b4fd446bf907b : SUCCESS in 57m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d252a2f63e54e08860436ea160b990c : SUCCESS in 49m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efe0a65878bb497e827ee81f2d849177 : SUCCESS in 1h 00m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f1a45a16f904cfd981172c465c7b14b : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b11bf1cccd2448dadb079f4405317a0 : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d91f26538c1745f294f1a1a6f0585c7a : SUCCESS in 23m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70ddc82b12a6419eb2391acd5aac9fc3 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9efe4c44fee4468382878dd12c33fb9f : SUCCESS in 19m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/52d52e70b1cb45b3a9fed43bf9eca00d : SUCCESS in 15m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5ea5e2ff4d984d28b81eb90a6d1b9ccc : SUCCESS in 27m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a3aac1124075462b91475fa375f6a479 : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d2dc72ca964f43a08acefac28e3aaa4a : SUCCESS in 20m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9d034967828e4a098fdaced978accd40 : SUCCESS in 17m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f4d434aba44f48e292d23de9e787dba3 : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0c22850856164f479404e7d5f1866ff9 : SUCCESS in 16m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/20d898d2ee5f480ea35a9ac185593517 : SUCCESS in 7m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f43613313f94cf1a0b931be2ee57de2 : SUCCESS in 59m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bcd58acd79f54a47971416dd26104db2 : SUCCESS in 38m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/196563edf7c047a8853bdaaabfabb673 : SUCCESS in 18m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cfc50ccd77d43ad9253f2e956374c11 : SUCCESS in 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c879a7bd8b402c91f6f32582eb5d0b : SUCCESS in 2m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/599aa78d0ba24699b22abfcd5b09ef70 : FAILURE in 8m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05ee2a9bf9474dd593925b92733b4afd : SUCCESS in 52m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3353998f0da24ebaadb11fb34df85773 : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ef744ec9029606d067953c3b834ccf5b5686f865","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 22:17:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd4ba789f04a4ee4b112934c6389bdb3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/89deb70738324014a8d1884dfdaf6e28 : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"576fcae01dd41ac93476b1c33b9d205c7713ca61","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-05 16:14:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"766ac4b581b9ae120ad0cf7250b2a24fd6fe5454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 17:23:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a70d63d151684512a0d73c381f78ba04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6c154aca3f24d83b0fdb81ac9b686c2 : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/8607d13c84d2474d9db7296819615103 : SUCCESS in 52m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/58cd463452c449059607e54095bb4184 : SUCCESS in 51m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63c34ccefd7a41899032b40d44bf352f : SUCCESS in 50m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff1d7ba4bb684eb18b110ed0c54d2ba0 : SUCCESS in 51m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75df980220574b8abda777cb578ca08f : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/122e5d3950a64b8cb7af5b8803544083 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3cb8a4831d1049518a730888b701dbea : SUCCESS in 18m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/21cd060fc56446b09c91399e4598ea66 : FAILURE in 3m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2416179ec5cf44869d1254e46c68ebc9 : SUCCESS in 14m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cdc51e61295d4874811af00dd033cb75 : SUCCESS in 19m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c59f993d6a3a45cd8043ee16af642ab3 : SUCCESS in 19m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4cd635816aa043c291033351d6b0d2d3 : SUCCESS in 7m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/da1f4bc65fc4401a9347ba13cecc029d : SUCCESS in 15m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8902d617e81e418391d70510ccb60ae3 : SUCCESS in 14m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c535464c91349fe8b95df0619abc3c5 : SUCCESS in 8m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d23e11c7a2a94599bd6b743439660d18 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/26d690cec205457583176ac9785efc09 : SUCCESS in 8m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d68e0b30b104009afa4beda40fd8939 : FAILURE in 58m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/400e7327540349129be84f8d92dcc7f2 : SUCCESS in 46m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff56b94243fb49d7b860a71c95e3df85 : SUCCESS in 46m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2be2001d35ba4e21ab52623e6df0418b : SUCCESS in 19m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee442584c564dccae129462d724f31b : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b7ace091b090421b94904915f19ab2c6 : SUCCESS in 25m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f7e93392b9f40fb834b4b098f8b4e81 : SUCCESS in 27m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5646f46795dc4bd6b7fa05886d80dc8e : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b822445c015194436b8d609170bbc09483980e40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 17:47:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21f2df6f479845179eb4601e32648ac1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7288d8841ae542c59360ff96989a6bd7 : SUCCESS in 1h 30m 31s","accounts_in_message":[],"_revision_number":1},{"id":"dd90314e88bb4b31212782f6180819747187c5fb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-05 19:40:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"419a140be9c1682ead28dc0a1c67484e163f6e16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 20:43:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c2585a2fb5645578b6b7fb3a86b859f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a3127c5909848ec9f76abdef25f5d91 : SUCCESS in 1h 01m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3818c215b3471edfe76b9ae14d2d13547ea819a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 20:46:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9030597090f4d89ae2fb28442c8b3fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47dcbc36b70b42e992e039f1a5cff427 : SUCCESS in 10m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/7d7a617b75a8476dadeedb53b160a320 : SUCCESS in 26m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c98d86b03ab5440389855b4f47f9f9d9 : SUCCESS in 55m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b90e013467b947828bc05a2502cc9a19 : SUCCESS in 51m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d17583856099423f96e514daa3298192 : SUCCESS in 1h 01m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df335ba0d7f343f59e9d65900167f964 : SUCCESS in 9m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ebe979e5b344ec8805ebfd53a2df59a : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f1a119646654b9282311b7b9bd973fc : SUCCESS in 21m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/779e25d9cfdb4cad8a2412b3432703d8 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cbf18112b2b0409092050cd4a8b7b1c7 : SUCCESS in 18m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6f0cc99703d147eba14ce75c800ac3e1 : SUCCESS in 8m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ff69fc7f0a1b4663950e45cf71d1e54d : SUCCESS in 21m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/330bd6b29e5a45eb8a03c38068ec9b5a : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/073db33c42234201a099326b368bb2fb : SUCCESS in 14m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2c4070eab1aa4cedb4a547c582dc2e71 : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5fc7ec5eefbc49c5b37d45c1dce852f7 : SUCCESS in 5m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c925fb70fe0496eace1992a8b31a422 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/696265c4e92a446db0fd6103676a1b28 : SUCCESS in 6m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7fd59aebbd640858417dc4b96d710c5 : SUCCESS in 53m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/036191206a964d21ae1b87a0609fb48b : SUCCESS in 41m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4cf148085b0e46bca3aa531846d1f7f3 : SUCCESS in 40m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/061f60f3f93a4d1b9dd0d659005372e9 : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eefa724fcc994724be5ea5b49fe1b271 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c8a94b4e3f22459e97995fa8b5013450 : SUCCESS in 17m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65088171226849348b6e3112da254ad2 : SUCCESS in 48m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/202be5a5fba846e796d82a972b8252db : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0cf7f668922c2eb7f1a037eb6d3e65805c4d8f7b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-06 21:53:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d18a84474d531c0e84d8057e03031c39ce68a42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 21:53:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13166abbb2308f614d04e947e38051ccf4e8a23b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 23:12:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19c44cf9950b4b8fa95dd83f01b2f6a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e2b1fd5d4da456fac291fbca5fbe5b8 : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/f7d5aaf815854f4aafb0597eae62b493 : SUCCESS in 57m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7705d9e9a6d743ddb7311bba1ec7e4f7 : SUCCESS in 1h 00m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3abee10446464cb5b7bafbd287235d7f : SUCCESS in 57m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2d83015355b4b28b160878e0c7e6a2f : SUCCESS in 51m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0afede100a674ad592293749bc4be9fe : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8393ea8ccc3942448cc744d64a3caf90 : SUCCESS in 19m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/479ea257493b48d183df8a6030ea9232 : SUCCESS in 19m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9247444c72254b4fbcf4a314fee1b7e7 : SUCCESS in 27m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6d0df6a5505b4a579e18ddd55b93a696 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bf131a05ec5b4c7b8ff1dd4bc4cdaa29 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f091ff64bc424c67aa1135f810308e29 : SUCCESS in 13m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b2d1d3aa69bc4923975ce13f571a62cd : SUCCESS in 6m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9cc57af420d7448a8ea1a1ee73cc2272 : SUCCESS in 9m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/53a7468750cf41b39d7f2367276869c8 : SUCCESS in 54m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3225bb1f37f24325843ed411becf8010 : SUCCESS in 49m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/376b33168af64f339a02b73816d060e3 : SUCCESS in 44m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9448e24c31a4028ac432a637b846ee5 : SUCCESS in 15m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/832ec3b2ea7f42e1a372527992706b02 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/32b3c172c7e14113bc1d3fef10b9743a : SUCCESS in 19m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c39aac0541e4794b1e9067388bdb8e5 : SUCCESS in 48m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e3539bb714e4fed85e6f2ad3145cdde : SUCCESS in 47m 59s","accounts_in_message":[],"_revision_number":1},{"id":"521cf0d8f93edfb9160bf40f85b47b0106773942","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 23:12:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0d17af87f78937e69bd919005951137baae3e89","revisions":{"f0d17af87f78937e69bd919005951137baae3e89":{"kind":"REWORK","_number":1,"created":"2025-07-30 20:51:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/956200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/956200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/956200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/956200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/956200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/956200/1"}}},"commit":{"parents":[{"commit":"3fafacbba15d1be01b2662767fc30edf0262288c","subject":"s3request: refactor to introduce SigChecker classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fafacbba15d1be01b2662767fc30edf0262288c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-15 15:28:20.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-07-30 20:50:51.000000000","tz":-420},"subject":"s3request: refactor error handling while reading input","message":"s3request: refactor error handling while reading input\n\nIntroduce a single error handler to catch errors while reading the\ninput.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib0330e863ad15e362dfa587517a26109c74eefec\n(cherry picked from commit addcc1e32b42da93978171f4f56f237f930945fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0d17af87f78937e69bd919005951137baae3e89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0d17af87f78937e69bd919005951137baae3e89"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
