)]}'
{"id":"openstack%2Fswift~956201","triplet_id":"openstack%2Fswift~master~I74b866dda986e077d079e0cb474c614acdb3a085","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I74b866dda986e077d079e0cb474c614acdb3a085","subject":"Forward-port stable-release CHANGELOG entries","status":"ABANDONED","created":"2025-07-30 21:00:44.000000000","updated":"2026-07-02 20:05:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bfe5f3c7ab71e33e725ed1705e1caf24ba855765","_number":956201,"virtual_id_number":956201,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-30 21:44:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-30 21:44:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a61c5d862d286e2a408d84b602527b8c41efd22a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-30 21:00:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5963a69b55a06c7a59c8e247978fcc0e3804ef87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 21:44:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/749553f9674948fab6668f02a00f0b64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9693239aecf542c3aa3340cbeb8986cf : SUCCESS in 9m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a627211b47e74f018edf651e5246a32f : FAILURE in 3m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8e8a80a404314738839062861f56034d : SUCCESS in 8m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a8cab84471a148a9a063c36845380c26 : SUCCESS in 21m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6aeec9dae61443bf8b0922d87794169e : SUCCESS in 32m 30s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/44519346bb8e4553a3e56263bc1a944c : SUCCESS in 22m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f89337a4a4b42e88cecd1ec29074a1c : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5a988ff5e64e6eb525061ca002a425 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"fa94e8951c2cc8fa7c78282c053074eecb2aa4f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 23:40:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ef017577d2c49cbb4362f90a37f271e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23f239e4238e4810995e7df26761e919 : SUCCESS in 1h 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"bfe5f3c7ab71e33e725ed1705e1caf24ba855765","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-02 20:05:12.000000000","message":"Abandoned\n\nDone (with others) in https://review.opendev.org/c/openstack/swift/+/991522","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97e961cd92b7e346201f27c4bbc907edcc832e9b","revisions":{"97e961cd92b7e346201f27c4bbc907edcc832e9b":{"kind":"REWORK","_number":1,"created":"2025-07-30 21:00:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/956201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/956201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/956201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/956201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/956201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/956201/1"}}},"commit":{"parents":[{"commit":"f39133055fff953f69f6b7ea62c47127d5b5c9a5","subject":"Merge \"sharder: make gap and overlap warning logs shorter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f39133055fff953f69f6b7ea62c47127d5b5c9a5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-07-30 21:00:25.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-07-30 21:00:25.000000000","tz":-420},"subject":"Forward-port stable-release CHANGELOG entries","message":"Forward-port stable-release CHANGELOG entries\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I74b866dda986e077d079e0cb474c614acdb3a085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97e961cd92b7e346201f27c4bbc907edcc832e9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97e961cd92b7e346201f27c4bbc907edcc832e9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
