)]}'
{"id":"openstack%2Fswift~956359","triplet_id":"openstack%2Fswift~feature%2Fmpu~I7bf3c6de245dbd2f5dd0e79a41b86d2c7c524132","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-custom-async-updates","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-09-02 10:54:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7bf3c6de245dbd2f5dd0e79a41b86d2c7c524132","subject":"mpu: make custom containers updates sync","status":"ABANDONED","created":"2025-08-01 12:01:59.000000000","updated":"2025-09-02 10:54:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dcd28eb6331d0152651b0b64a8122a91f8ff4395","_number":956359,"virtual_id_number":956359,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-15 19:17:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-01 13:16:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98c32ddd6d25769d101001eabb790ac3cb93c55a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-01 12:01:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f4130e06100848033bf808a87a44b25841f0ded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 13:16:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e3f3bae3b5f4b3287e3cf46aae5b37e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a14c8a5c4b64751a9d1e109897f6bda : SUCCESS in 5m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3b5226eebd44403ad439a847ed0faf2 : SUCCESS in 11m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8081f20e8084e999610cf517c778587 : SUCCESS in 11m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7127854a8b74b108aef25667956c04e : SUCCESS in 11m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3870410b4dc94e8f8fd9e58f5e2723cc : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0ff8efa39c5e41fa87ecac0fb3b6c0df : SUCCESS in 17m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a1955d0968554ab099c0a9cff1419532 : SUCCESS in 15m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/25a51d2e4e00419dadd8cd91b1e5be64 : SUCCESS in 17m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ec970fcb662a45adb8eb090475119611 : SUCCESS in 22m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cd04aa2a88c94844b9c4105d072e7594 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/889e19ee89ab43e492992d7f9b8dd94f : SUCCESS in 18m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d7ef296f519d4f38af397815d6c7543b : SUCCESS in 17m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/487b7f9915b640fca51da00a30b6ef0c : SUCCESS in 6m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd231127fcb847358af2957f33df1c41 : SUCCESS in 8m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/634655899fd345d9ad41e4e2dfce01f8 : SUCCESS in 4m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a78a1b5d5cb4edc8e1dc95a7bc0b2f8 : FAILURE in 1h 03m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3139db1149340e4a6d451a8e151ef35 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ba6d58f65b746f8a33b781f1c995224 : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7de10ccdd28e48d2a9c9de6db3b17ac6 : SUCCESS in 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b26f72a306a407838dca06ee14a04d6f4026703c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 18:13:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a5cae4d543748af9b549aff286a372c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a53adccf5fd3497bb0afb2cc00068920 : SUCCESS in 1h 24m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a1686e0796e011689d444777b8f528eb8bb97fcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-04 09:19:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f4c9393ff3a1f7597d14cba1288abb4bbdcd8e85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 10:02:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ecdbfbcfc3a4df08af6877bf154f540\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198f83f9abbe43a3be66cbdd0b8b44a4 : SUCCESS in 10m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25487d623bab4127a57602abb3c4cfc1 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3f5f554790b44578fd6895cf6d36b33 : SUCCESS in 13m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cc4f052c8bf4df792897766a1952f9b : SUCCESS in 13m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0a4d41b71a0348d9b4510927a3531cdf : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b460e7631ada41fbbc91f863897f9615 : SUCCESS in 19m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/52f6150dae08474da11a13a79972a546 : SUCCESS in 19m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/13b999aeb36f494fa3379b97269ec4d1 : SUCCESS in 21m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/def92a678b6f429db7686691ab96c1f1 : SUCCESS in 29m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c52931d2031641bbb7d9e217da06c88e : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/05b938bae0c64f3aa229b45fea97a8ee : SUCCESS in 8m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/04a3c6205ee5420ebf87e814dd198448 : SUCCESS in 16m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0dcfca612f9749bf963a965553f8ee8f : SUCCESS in 7m 58s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c2303ea13eb24de58067b391420e4375 : SUCCESS in 9m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/894af4867a4344c384745446208817ad : SUCCESS in 9m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/007c61b4d75f4911a36c9bc3d470a35d : FAILURE in 41m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bb6fa4bcffb4eefadab55e314015fbe : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6965fd07e224ed6807477a8fb9c07ef : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e597f5caa780408fb4829aa541a93b5b : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b4f2310fd9b12d662aaee9b84d6337d6ca9b1005","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 10:54:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/653b6f3709cc4032b6e61758558ac4b5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5ef422ede71343448784e55ca66959d3 : SUCCESS in 1h 33m 27s","accounts_in_message":[],"_revision_number":2},{"id":"36d50d578dc07ca593d87b3f83b0a46425d88abd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-04 13:17:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e17a5910f816629fe86733333b19a9702f6cde1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 14:25:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/564a22f293fd4fa4a065c39b44d8e1ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1617a96bca584a98b3ebb182c85932a7 : SUCCESS in 9m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc5697fca3424e948ec7b1fcaa427f3c : SUCCESS in 25m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c19daa51e8d4e75b5e401c34f09ea68 : SUCCESS in 21m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b90c5054f4d44f6bcae7983cdfbdd20 : SUCCESS in 21m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fd9c4949299f4f35a8f82668057688a6 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a9e22c9a691a43c1baa04f5c5b47aa18 : SUCCESS in 21m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f48ef68c4cb34a4980727ec60fbf36e3 : SUCCESS in 21m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7b8f897319784587a443bc7d3cf6c1c0 : SUCCESS in 29m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cecee128d4d54d4789a5e706f54d2dbf : SUCCESS in 28m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e64201e017764196810cb1451baec485 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/403092dfadf848be8bd05db7fb2390a1 : SUCCESS in 14m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c695748fc59b4e9aade5f8858926ea5d : SUCCESS in 20m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fd391947e95e4932be40af7c9c96d2ae : SUCCESS in 7m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b533e9b543d64325b69d23e22d180aa0 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d0847e56eaec43f4b99b07a6ba592772 : SUCCESS in 12m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6269e67671b54cbfba4e37a439df7371 : SUCCESS in 1h 01m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a352617d60704b6ab6f3d6f97ca9e791 : SUCCESS in 15m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7129190d7aa49759e9a8842d065aa0a : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/093aedd55578425c8233eca3e0bc53d8 : SUCCESS in 18m 18s","accounts_in_message":[],"_revision_number":3},{"id":"fc83be72bd45c2796a3c50df5e248188514d51ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 14:35:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aa1864fddd84daf9ed14117cacc2c02\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fca1a1f0a7c04da6a7daa238e3a0d2e1 : SUCCESS in 1h 16m 35s","accounts_in_message":[],"_revision_number":3},{"id":"5c295758d100b56bc362c0c65b4d2f04b3504049","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-08 14:58:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"25c567b05c4d8a2abdc2b7f0d6dfbc097f56e99c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 16:02:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4163df231a484c0280d8f01d71a18d2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b454077798814da5893f166741ccafa1 : SUCCESS in 10m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09c063e1c9534848965d950f54d81782 : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c37f4856bfdb4d6fa20f3000ee4f0230 : SUCCESS in 21m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/664ee8dd3b174fae97007a8f5bf534a6 : SUCCESS in 33m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a34ce2c94365476486c96eec8921c851 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ebcc53f018da497f8dd8cb4dda9fe113 : SUCCESS in 16m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/80d0dd8495664c3ea42d8bf88cb7bdf2 : SUCCESS in 15m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5c7e984c748c4087975f91fd144d8536 : SUCCESS in 10m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2ccbca1c014d48e398d0ff03d87b5ece : SUCCESS in 26m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cb8f9a3e5c2f44c18fa10374e061d17c : SUCCESS in 20m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/171e4396f0b1482b98756277d7fbc115 : SUCCESS in 18m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/719add42ae5b4394ac58e464afbdcc8f : SUCCESS in 17m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/672154945a0c41a6a6fc38e502c1c15e : SUCCESS in 8m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c16d5da125c14509abdb957378558d2d : SUCCESS in 14m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/65ab34c9dc3746e7b49f328f6fc314b0 : SUCCESS in 8m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7aa0d5e8eae4c3bb1587d8ebe376067 : SUCCESS in 56m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad672122d17342a8a2072fa9c4294876 : SUCCESS in 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22e5ba06335541c28c6ddd637a03b8bd : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a879cf84fdb3408cb1f7ce48708ed661 : SUCCESS in 16m 06s","accounts_in_message":[],"_revision_number":4},{"id":"758951981fc43d7505229486606454e4fe34ffca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 17:46:04.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0163f6789c44a1a8ae67f08bf73ac3a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a443fcf6bbe4f3393c8b2f1ee7fe22a : FAILURE in 2h 39m 56s","accounts_in_message":[],"_revision_number":4},{"id":"4e0fe9ccccefd48ae6a461ec7d26b209c1af412f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-15 18:05:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3aa3010613814807f0b8daae188cad95520d8cdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 19:17:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a7cd75073e6438ba2aaf2e37fa58e94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97e6412e7cb241dfbf90adf5bed19ca3 : SUCCESS in 6m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea28601ad2854ed1ab325aed5e4bde52 : FAILURE in 18m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32876e41e81f4eada4f95c53e75e250c : FAILURE in 22m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bbd782485a30415da2f05d5817f09d31 : FAILURE in 10m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/73a6366ad302479db06a20eb0b4261d1 : SUCCESS in 4m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/854ea54f345a4c189a07e054047d4ee4 : SUCCESS in 19m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/62fd5b00253746c4a81044cb21e7c58a : SUCCESS in 15m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/57841d6ebd3e4834b36aa6b8558a3482 : SUCCESS in 10m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b3c14f20f8194cfdbc7c4ca1ff03de7d : SUCCESS in 13m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/40a49495af9d4134b1a0b42b89c33b3f : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f8880105df894a088e1a7761383c0b20 : SUCCESS in 8m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/76ef10f1cb75463e83c7a79b8bc6eb97 : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ad7aabaeef9c4b3d8c29aa23e1420587 : SUCCESS in 9m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d43f904fd71436cba054e500cb8759e : FAILURE in 8m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fb903c939ccc4e959de7cb58f1b7e5d3 : FAILURE in 4m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6438783ddf640e2a7d0bff78bf3428e : SUCCESS in 1h 01m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d080881e39374e70a09d8b1870e66c0f : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b261beb42ab460480c4ee099d585ecd : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cc0fad7f4dc422084a136f1dec8e50b : SUCCESS in 12m 00s","accounts_in_message":[],"_revision_number":5},{"id":"3ba96b1c33a20a445a6bf344b7ae9bcebe7fd382","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 20:10:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3993448a7e784426b17ccd938782621b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0dee26fb3e0f492391e3b5f2761483b8 : SUCCESS in 1h 22m 42s","accounts_in_message":[],"_revision_number":5},{"id":"dcd28eb6331d0152651b0b64a8122a91f8ff4395","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-02 10:54:09.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/950689","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3651a0ed895fa582fdffdcbcce00c6d28e6fd4b0","revisions":{"23c12faa2c568bd4b8837e85bdde54651f54ad95":{"kind":"REWORK","_number":1,"created":"2025-08-01 12:01:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/956359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/956359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/956359/1"}}},"commit":{"parents":[{"commit":"95f2785fddc36236466902c22ccc88f9fa90c0d3","subject":"WIP: use version history to detect orphans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95f2785fddc36236466902c22ccc88f9fa90c0d3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-31 17:52:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-01 12:00:39.000000000","tz":60},"subject":"mpu: make custom containers updates sync","message":"mpu: make custom containers updates sync\n\nHave the proxy add container host/device/partition info to custom\ncontainer updates so that the object server can send the updates in\nsync with requests.\n\nChange-Id: I7bf3c6de245dbd2f5dd0e79a41b86d2c7c524132\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23c12faa2c568bd4b8837e85bdde54651f54ad95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23c12faa2c568bd4b8837e85bdde54651f54ad95"}]},"branch":"refs/heads/feature/mpu"},"d6eb88988fb80e6580771211faa0fc4dd5484fa7":{"kind":"REWORK","_number":2,"created":"2025-08-04 09:19:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/956359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/956359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/956359/2"}}},"commit":{"parents":[{"commit":"95f2785fddc36236466902c22ccc88f9fa90c0d3","subject":"WIP: use version history to detect orphans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95f2785fddc36236466902c22ccc88f9fa90c0d3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-31 17:52:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-04 09:18:35.000000000","tz":60},"subject":"mpu: make custom containers updates sync","message":"mpu: make custom containers updates sync\n\nHave the proxy add container host/device/partition info to custom\ncontainer updates so that the object server can send the updates in\nsync with requests.\n\nChange-Id: I7bf3c6de245dbd2f5dd0e79a41b86d2c7c524132\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6eb88988fb80e6580771211faa0fc4dd5484fa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6eb88988fb80e6580771211faa0fc4dd5484fa7"}]},"branch":"refs/heads/feature/mpu"},"1a01a464b93e2b8d8d8bb394ec564e928eddafcb":{"kind":"REWORK","_number":3,"created":"2025-08-04 13:17:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/956359/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/956359/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/956359/3"}}},"commit":{"parents":[{"commit":"95f2785fddc36236466902c22ccc88f9fa90c0d3","subject":"WIP: use version history to detect orphans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95f2785fddc36236466902c22ccc88f9fa90c0d3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-31 17:52:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-04 13:17:34.000000000","tz":60},"subject":"mpu: make custom containers updates sync","message":"mpu: make custom containers updates sync\n\nHave the proxy add container host/device/partition info to custom\ncontainer updates so that the object server can send the updates in\nsync with requests.\n\nChange-Id: I7bf3c6de245dbd2f5dd0e79a41b86d2c7c524132\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a01a464b93e2b8d8d8bb394ec564e928eddafcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a01a464b93e2b8d8d8bb394ec564e928eddafcb"}]},"branch":"refs/heads/feature/mpu"},"63d5528e1e7e1d06611f58722f1a4b9b29837647":{"kind":"REWORK","_number":4,"created":"2025-08-08 14:58:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/956359/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/956359/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/956359/4"}}},"commit":{"parents":[{"commit":"007688df49afd02b344dd2dfe6c641663ef20de8","subject":"WIP: use version history to detect orphans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/007688df49afd02b344dd2dfe6c641663ef20de8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-31 17:52:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-06 15:28:20.000000000","tz":60},"subject":"mpu: make custom containers updates sync","message":"mpu: make custom containers updates sync\n\nHave the proxy add container host/device/partition info to custom\ncontainer updates so that the object server can send the updates in\nsync with requests.\n\nChange-Id: I7bf3c6de245dbd2f5dd0e79a41b86d2c7c524132\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63d5528e1e7e1d06611f58722f1a4b9b29837647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63d5528e1e7e1d06611f58722f1a4b9b29837647"}]},"branch":"refs/heads/feature/mpu"},"3651a0ed895fa582fdffdcbcce00c6d28e6fd4b0":{"kind":"REWORK","_number":5,"created":"2025-08-15 18:05:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/956359/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/956359/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/956359/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/956359/5"}}},"commit":{"parents":[{"commit":"823b18b4c17b58d6544884b4bc996c37f2e1e708","subject":"WIP: use version history to detect orphans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/823b18b4c17b58d6544884b4bc996c37f2e1e708"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-31 17:52:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-15 14:06:29.000000000","tz":60},"subject":"mpu: make custom containers updates sync","message":"mpu: make custom containers updates sync\n\nHave the proxy add container host/device/partition info to custom\ncontainer updates so that the object server can send the updates in\nsync with requests.\n\nChange-Id: I7bf3c6de245dbd2f5dd0e79a41b86d2c7c524132\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3651a0ed895fa582fdffdcbcce00c6d28e6fd4b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3651a0ed895fa582fdffdcbcce00c6d28e6fd4b0"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
