)]}'
{"id":"openstack%2Fswift~956505","triplet_id":"openstack%2Fswift~stable%2F2025.1~Ibbb8c305a8b8bac7ca1a8853208cd031957adac1","project":"openstack/swift","branch":"stable/2025.1","hashtags":[],"change_id":"Ibbb8c305a8b8bac7ca1a8853208cd031957adac1","subject":"CI: run func tests during rolling upgrade job","status":"MERGED","created":"2025-08-04 17:43:19.000000000","updated":"2025-08-04 20:20:55.000000000","submitted":"2025-08-04 20:18:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956505","meta_rev_id":"fa2b9cb63c18392ee3a6b310263305d81ff68a41","_number":956505,"virtual_id_number":956505,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-04 20:18:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-04 19:13:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-04 19:13:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-04 18:55:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3fa29c430ef9b0c66149d193b599b5fda0efd171","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-04 17:43:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dcdd2572e1bd07a14a6b5dae9a4bfab527150c6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-04 17:43:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2449fbce1ea6227a65209ae7329b77ccdd551069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 18:55:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f508b69127d4b7196fc3abafe1a6536\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d88186ea5f84d1e9b538e3917ef37a8 : SUCCESS in 11m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/8210d7d6fa6c4a64a9538c13238262d8 : SUCCESS in 58m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cb2de60239ab44679cddbf2275596329 : SUCCESS in 55m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae65ee7422ee489999198ce72b55c000 : SUCCESS in 52m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e48798ad50444d6c8f80bbef4db2ba2f : SUCCESS in 1h 07m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e255f3496084f9784759f89931d08a7 : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/709bda0701ea428b94f05efdd2e3114a : SUCCESS in 21m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d02f048b48e644b9a6c31385c4e427fc : SUCCESS in 20m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/446dfac4ee524296a6010a1d3707886c : FAILURE in 3m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef93a3a79cfa4e4f9fbf937bd97bcc5d : SUCCESS in 19m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7d562173c15d4aad970942b0412f6455 : SUCCESS in 7m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b59aa36526304ad1bf0fbb5e6769f10a : SUCCESS in 14m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cb666abda3e84a60aec6d82e019630b9 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/807d68fcdca941df9c0f699630b526f6 : SUCCESS in 15m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4869ac46c4074b7da85591c2f4e84677 : SUCCESS in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b665d0a921b4d69b68798c4e0ad1ed7 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1bca2092e50e4c69a1eec880e84dc2a6 : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e50ad4c035ca446096f89b218cd0e9a8 : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c69b48de936942da936c7f59760ba638 : SUCCESS in 58m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/28637e5fbd1e418b8fb3fd78e526b71e : SUCCESS in 17m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea5c8cfb236f49b18376c2c28c5f1111 : SUCCESS in 48m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0905affbef2042959edf09eb089436e8 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcff04087b9e4f2a8c23aad2dcc622dd : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5fd1d03246b4c9a865bc43e5cf91334 : SUCCESS in 24m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/204eb7a4bdbe4ef18c4ef20d9602d785 : SUCCESS in 55m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23c2d999eaf34137a873c25d591300fb : SUCCESS in 28m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c2ff78588e8e8d64f203b341aa2d486bd0e3fdf8","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 19:00:50.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7817ecb7a5f24c99971415c0701e9236\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a754466c77a149e08006da7a7bc5fbfd : SUCCESS in 1h 05m 34s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/75f5be7032664a5488b5ac5a015ecccd : FAILURE in 1h 15m 14s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/6d26bf3c66c74d4f8a4898f665ef9222 : SUCCESS in 51m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ff56201c05644700958604dd8fdf7ff3 : SUCCESS in 21m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f6cd102f940e49dcadf13d271b99d2f0 : SUCCESS in 23m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b31bc5e626864b5c8b4859e9323db963 : SUCCESS in 21m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/edb7ee8ce9b54208b913394a664f44ad : SUCCESS in 20m 50s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8914e52c6ed4f3fa0e90b87e277d2b2 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/32c22887d6dc4cbca3c46434f9cf9f60 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a74ed24a2eb546a3a4ca7b35f6f3f770 : SUCCESS in 18m 31s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/9fbc20e086e14fb6ba37b8b6932e5b0d : SUCCESS in 18m 12s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/52f6a420b3ee49ceaf66bd51de054898 : SUCCESS in 17m 03s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/725fc8c39807483b81fec9a694a6fb15 : SUCCESS in 18m 20s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/1abc0fd1c86a4b999d50e1b5c81a9c91 : SUCCESS in 17m 50s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/4debb2da134c40e58628926539a0cebb : SUCCESS in 21m 32s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/db17a9b0747c4617a7538f2a3e2d14ac : SUCCESS in 20m 26s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/b5e5d4dafd3c4b9696a4297dd4075308 : SUCCESS in 19m 39s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/2097f6179675497897def34290a9d632 : SUCCESS in 17m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fc1b84148bae50d0c0a106319c47544218e30172","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 19:11:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de5bdfb8ef244aa69569757e10812344\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/169c1d01f08247dca6e7558a60a3ffe6 : SUCCESS in 1h 26m 15s","accounts_in_message":[],"_revision_number":1},{"id":"cd80150259b8f6925b8f3b053541f9b054eca460","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-04 19:13:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"412531e6a9c8e87a63e034c6c2c398a0ee25879d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 19:13:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c7a3067d33b0f058dbcb1a0ff7605c611850b66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 20:18:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc02adbe85054ee5b082c07b4e0c416a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7cc34126ec4a39a3f5efc4a2d4b283 : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/b9ac8afa3a1f488b88605ed46fbdabc3 : SUCCESS in 58m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ff074b6181743ebbe5c4fed277c810f : SUCCESS in 27m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/508777d1956d486c9d5fa049a5acb373 : SUCCESS in 53m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9cc4e35322914f5d8906aa3944f6a694 : SUCCESS in 1h 00m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/204e5ff620fc440ead19c3293a82e726 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fc654c77d165406cb6c6ae0f0b9929eb : SUCCESS in 17m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f74a2a02143544a2a3c3240d6498045a : SUCCESS in 21m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1fc7d52c5eec449f99f4b3ba90a28b6a : SUCCESS in 17m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d9845747b9d54f608deaced2c574c447 : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d7d216b3938f423a86ccf6ce811bac57 : SUCCESS in 14m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a1a5b49c778d4f328c3bec08e4b243b8 : SUCCESS in 11m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/048a56620ce846ad97a5f9531e0e97b5 : SUCCESS in 7m 53s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4f93c5d5284b4d6fb4a365f9a493a980 : SUCCESS in 6m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3c4071d37f842edae19cc7037d86ac3 : SUCCESS in 32m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1f55b2a519554bb7a8d75a78ca19baca : SUCCESS in 18m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/080438a583db40c49ff164c283be883c : SUCCESS in 18m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c0f7f32fc9246e7abe21de188e2d4c3 : SUCCESS in 17m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51645f8dad3e44a79bd60de706ca45ea : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/91969f79cf3f48e98bd9a5881180165d : SUCCESS in 26m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fd6182e73f54c5ea1aea081c1487d11 : SUCCESS in 48m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/411fb0461a714e359557bc67d21b085a : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0da4ffbf25be4fd034c629980301d5e6b8d59449","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 20:18:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fa2b9cb63c18392ee3a6b310263305d81ff68a41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 20:20:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c931e0b0dbb74affb3638f8e09ed52cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d5f86bb534423d9e610e393117a9a9 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/37dcff1d8b5f4e64ab20187bcebe8601 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b9ccdc0388bcd65c67f99926ec630b4faf252d5","revisions":{"5b9ccdc0388bcd65c67f99926ec630b4faf252d5":{"kind":"REWORK","_number":1,"created":"2025-08-04 17:43:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/956505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/956505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/956505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/956505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/956505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/956505/1"}}},"commit":{"parents":[{"commit":"095779bf41ffb090747aee5c8670328a8cdc08c2","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/095779bf41ffb090747aee5c8670328a8cdc08c2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-14 17:02:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-04 17:43:05.000000000","tz":-420},"subject":"CI: run func tests during rolling upgrade job","message":"CI: run func tests during rolling upgrade job\n\nWe recently removed the \"func-py3\" tox env in favor of just \"func\" -- as a\nresult, this went back to running unit tests!\n\nKeep \"func-py3\" for older releases where that tox env *is* still defined,\nthough, as they won\u0027t have py27 around to run \"func\".\n\nRelated-Change: I53153c4fde043e964e1daa7bbf2089e0471dede2\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ibbb8c305a8b8bac7ca1a8853208cd031957adac1\n(cherry picked from commit 33f409ceba9a22f8067ea5b10317f7b9a8d3f705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b9ccdc0388bcd65c67f99926ec630b4faf252d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b9ccdc0388bcd65c67f99926ec630b4faf252d5"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
