)]}'
{"id":"openstack%2Fswift~956730","triplet_id":"openstack%2Fswift~master~I9df32c5aae4e681d488419f36982ffc36589d50a","project":"openstack/swift","branch":"master","topic":"traceback-in-prod","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-08-07 00:06:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9df32c5aae4e681d488419f36982ffc36589d50a","subject":"Fix another way mw may encouter invalid swift paths","status":"MERGED","created":"2025-08-06 19:54:42.000000000","updated":"2025-08-07 01:25:27.000000000","submitted":"2025-08-07 00:06:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956730-traceback-in-prod","meta_rev_id":"bcaa815041883dbb7e763b52f679259d7f3cacc8","_number":956730,"virtual_id_number":956730,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-07 00:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-06 22:52:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-06 22:52:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 21:04:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-06 22:52:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"d3ea2a1aaa00797419c94502dbe636c40296cdf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-06 19:54:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65b15b3018f13532204e2ca6568f218dcf381d37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 21:04:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/337c1d85236c453fa71fc47ac009df86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cace82abb78b41aa95200d302eb33910 : SUCCESS in 11m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/9f26e32f543b49d4905e356571937077 : SUCCESS in 57m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1c86f19d28ad47baa575ad60bdbb7bc1 : SUCCESS in 58m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/df436d1a495d4b95bd19843cc751d851 : SUCCESS in 58m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9372bf9fc60a4474bc862ed293211a3b : SUCCESS in 59m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b986cc50a9a45039c73abaaef0fa80d : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f3dec79a7cf44e7b0dd7400bd2a8d94 : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b7d7e0bc57542398066e807b383ce34 : SUCCESS in 19m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/46228edf18014406a0b30af83987c7cd : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3f1a8bf66f174e8eba43468c0c03c4a0 : SUCCESS in 9m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a268c09d3e5f4dfa8d664aef7a0cd3d1 : SUCCESS in 20m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/462468353e8e4f9c92bdd032654bc692 : SUCCESS in 28m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8d6101be9bd04be6a9d958a2f7611956 : SUCCESS in 22m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/625756737ebe4eee9a7353185b9d77da : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cfd5d4fe65f245d4af5a7d1c6b8834d3 : SUCCESS in 14m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d41f8a22777742949f87c4234f800dc6 : SUCCESS in 15m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/75a3e28745124a18845fa3dd23eb3a5b : SUCCESS in 6m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/60bc3d7574aa4bf090e7c2b91df0d991 : SUCCESS in 14m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/50590a748ca94604abf00eedf5c01f61 : SUCCESS in 9m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bc5677eb30447e69e726a241e99a173 : SUCCESS in 55m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e17d48638aa64cedb23a9778c04dbf64 : SUCCESS in 34m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a805a9785ac64dac9f67dd343f4f647a : SUCCESS in 53m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2d0edddc8c14bcb94c0f626d858077a : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ec810aa940346978fb93ca4d414ce68 : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d155f9a768354b9493462bd8be166437 : SUCCESS in 17m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/882c27d6bff3409ba12c0c331218da30 : SUCCESS in 59m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2d0660185904e90a3caae0fdc33f086 : SUCCESS in 28m 10s","accounts_in_message":[],"_revision_number":1},{"id":"af90deed4fb558bf90274ce634630cf9d75736da","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-06 22:52:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee89e2f99763baecfc1e42c46876418506abf02a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 22:52:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ec385641b52e609816dd2f381ba485b81253561","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 00:06:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d642a83c60d4d1ca546a34e6f55913d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4e300a7ac3647c79b13dd4c7f517fd6 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/1803fc7831a449998fe4676a489c0038 : SUCCESS in 57m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/403e0bea003a446e8cf65c86e4279ac0 : SUCCESS in 1h 08m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2ed537b24cbd4b99984922b029cf4a49 : SUCCESS in 29m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f198d21194e4400ba66d2681ed8a8c8 : SUCCESS in 1h 02m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8e3340d96ce141c98ff225b857dd2e71 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9370764d922c4a64adf3961ae328ec5d : SUCCESS in 13m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f881972a49f445b3a1f0ed031ac33109 : SUCCESS in 8m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/48638143b15b4a06bfc30d80e9199cdb : SUCCESS in 22m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0321f70fa94f4c458b7ed3ac3a5a129c : SUCCESS in 11m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7a59d6bc9b394003bdf0aa516b7491d6 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/00b422fe7e914eba9be178384be88ba2 : SUCCESS in 14m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05c0a5fbfb5f47169793d695412deac4 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/65191ca365164bf6b759628b58f1f3c7 : SUCCESS in 7m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9363573cd118407ca5997cec70578efa : SUCCESS in 1h 02m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aaf29441d12f49dd949512311d4434d7 : SUCCESS in 52m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b3623e4581924cbfa9c1de37c58dece9 : SUCCESS in 49m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9d7b675e34d4e1fa99aa23637d477d9 : SUCCESS in 20m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/865a3973570447da94fc8768b2fbdfdf : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ea2b99f6bdd482ab1bae179a10f5fa3 : SUCCESS in 8m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ac132795c394f8fae08746b61377d5e : SUCCESS in 49m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ff803377aa9436bbb023fa1ac986f5e : SUCCESS in 51m 34s","accounts_in_message":[],"_revision_number":1},{"id":"737922eb92ecfe6733e26f5966a8bfbebf9832ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 00:06:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"092d87a633ce2e7479e9627f6ae949b1c46763c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 00:07:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59edbb9adbd841d98e238921f2ed4db4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b72f23b5a648d1a27c8449e8ce86fa : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b1cf49ce4f204756b41c7b237efd387e : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcaa815041883dbb7e763b52f679259d7f3cacc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 01:25:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9899f1d9d3c4dafa80c3246bcb1e0c8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/696880973fcf46578b92c8b7668459ee : SUCCESS in 1h 59m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a770881b59cbf53ab47a3e95616ccded113d86e3","revisions":{"a770881b59cbf53ab47a3e95616ccded113d86e3":{"kind":"REWORK","_number":1,"created":"2025-08-06 19:54:42.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/30/956730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/956730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/956730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/956730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/956730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/956730/1"}}},"commit":{"parents":[{"commit":"86a1acc9e3f50729f3d95e8979e9ed133e1c24f3","subject":"crypto: Fix traceback on non-utf8, non-swift paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86a1acc9e3f50729f3d95e8979e9ed133e1c24f3"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-08-06 19:52:31.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-08-06 19:53:43.000000000","tz":-300},"subject":"Fix another way mw may encouter invalid swift paths","message":"Fix another way mw may encouter invalid swift paths\n\nRelated-Change-Id: I8c342c4751ba3ca682efd152e90e396e9f8eb851\nChange-Id: I9df32c5aae4e681d488419f36982ffc36589d50a\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a770881b59cbf53ab47a3e95616ccded113d86e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a770881b59cbf53ab47a3e95616ccded113d86e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
