)]}'
{"id":"openstack%2Fswift~957412","triplet_id":"openstack%2Fswift~master~I64a85eb739fb72efe41f1ee829e463167246b793","project":"openstack/swift","branch":"master","topic":"p-checksum-pyeclib-bug","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-08-21 00:36:02.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I64a85eb739fb72efe41f1ee829e463167246b793","subject":"checksum.py: fail gracefully if pyeclib is broken","status":"MERGED","created":"2025-08-14 10:31:30.000000000","updated":"2025-08-21 00:36:02.000000000","submitted":"2025-08-15 14:44:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"957412-p-checksum-pyeclib-bug","meta_rev_id":"ef7059640f6441ab4c9e32193fec1a35876b263f","_number":957412,"virtual_id_number":957412,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-15 14:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2025-08-14 15:54:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2025-08-14 15:54:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 12:45:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-14 15:54:09.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-08-15 04:26:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-21 00:36:02.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"7ee9b641a74ac2726a08322ab79d8948329a41c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-14 10:31:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1061382a31f9381ad84ad7ff76d88fb894c8cbdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 12:45:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99a44ac83ab54b579cec78be5bd6572f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3fcf3ef6bb09429ea4bceaecb03871eb : FAILURE in 1h 32m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ec84e9d339c7b534e8fe2c4469531241fa4f312b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-14 15:54:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"994c02537d8fe2682e8765f62363b581714ff78d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 04:26:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35ae4fda200f4c00a7eb81cb6e5b2dcb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d6cdec035064901bc79259d8a442c96 : SUCCESS in 10m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/ecf89f8e909744819df6f12978638ccc : SUCCESS in 1h 02m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d1cf17c96c09417081f4879aa32a0640 : SUCCESS in 57m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/56aa5e5f9f9349e78e7bb2783f5f92a2 : SUCCESS in 51m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b982f0456fb442599f955e234274837a : SUCCESS in 1h 01m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b69b9c8b0d594d998390f08cfa769367 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff1c64ad0bf54639b73dd8ce92536411 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86e9b678157740c58b122f95bbb9ed91 : SUCCESS in 10m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/46035a9818994d01bf019e2bb675de16 : SUCCESS in 3m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d67fb36d5ff243d5a2d543c1e6222417 : SUCCESS in 14m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/12a0d2514a5c4a1aa421dc7e6bb71332 : FAILURE in 24m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e393de2d0b15451aab5c002b3639fd6a : SUCCESS in 23m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/db217bbe3f8b43c099c7262178bf3cc3 : SUCCESS in 11m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/50ce7f0add274ff6aac551d9567f8218 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/01bef9173f5e42b48d3d149af9d290bc : SUCCESS in 8m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/13dad71df4d142d0ba870e373ae0f239 : SUCCESS in 14m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b5e171a8d94a408ca90a97f9ed830338 : SUCCESS in 4m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6fac5a9b63b44f468a97076ee8eb44f1 : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dfa399d3fb8d428d82c912aaea283e98 : SUCCESS in 6m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d9ca61a8e654369a5b08683891e46a2 : SUCCESS in 37m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4781c1ea2ce46f48b843f64f2c12c64 : SUCCESS in 22m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0415900976774d839bdca131996aca9e : SUCCESS in 25m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae49485609d847e5a92b5a7daed6bb0e : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/101755ea548c4d69a89491424049221a : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e9fbde6c4ff44fe83be668ffdca24ce : SUCCESS in 20m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b580ba179b564986b9897db005db1740 : SUCCESS in 31m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29fe3c486ce44e20b935ba7c00d18509 : SUCCESS in 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"485d296a0a4472b4ad22ee090ca9f71cc0544ac0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-15 12:26:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d1d6b7eeb3f9e8b157768043cf489edb02ccc6b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 13:29:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/855f26747f9b419b85f9f2243f345df6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97586af0cacb4d70a99fb95fa7b57d17 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/7b22a04407764a9c8cf9217e962dd29d : SUCCESS in 34m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2143de75f1264b51b995cc1225f184b0 : SUCCESS in 33m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4989bec831ec4720bb4fda9b54cb1cd0 : SUCCESS in 55m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92ca6aa7e447499b8c95bdfc4cb69650 : SUCCESS in 49m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf8ee71ec37f47759e21256fc7605e56 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86ebb1def0fe42f5ba6ac392e4dd104e : SUCCESS in 10m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf760f8641c84f209b55253e3f9aab59 : SUCCESS in 16m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce82880f6b6c4c20a913d6b970bddaea : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7d002d98caa84075b7162849d8058b07 : SUCCESS in 8m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a366d40479fa40148d79cc88239fb231 : SUCCESS in 9m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f64f5c5f4b3c48578478de953411d45d : SUCCESS in 17m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9572dfdc175c4cc48c4f8eb9fc540ec6 : SUCCESS in 10m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/beff83209f5d4b98bc18c9cc09dbd54f : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/690ba19665694f959196bdccb36425f6 : SUCCESS in 13m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4935fa760308448e80d2a6f146861525 : SUCCESS in 7m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/447b031d435a45ee8433199da36fc862 : SUCCESS in 6m 17s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b462127347be43c38a57abcf2b9c4f4b : SUCCESS in 13m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ce8297c026274aa58326c1271180b58c : SUCCESS in 8m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc80f18bcf974c1d8d6e0b7bea0dc2c5 : SUCCESS in 1h 00m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ca20a3135e32431eb44c9c90e0858b2b : SUCCESS in 26m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fe2db57ae8254d698eaea88401a61d35 : SUCCESS in 36m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbdd483a66fe41b9a827bd2ae97b96b9 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0c8cedea7fd429aa47d397e499b21f7 : SUCCESS in 2m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/484d0fe4b2df40c99da84fa8f565566a : SUCCESS in 8m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d77b275782d543a4b31984ceccd12e4c : SUCCESS in 39m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75e860e1f1d74e9d942c789a75d14000 : SUCCESS in 34m 41s","accounts_in_message":[],"_revision_number":1},{"id":"bb578670382478c308af3bfb5c6a2221181d7831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 13:29:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f628fdbde26da0bb1a96185cec083831c0a7370","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 14:44:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f153270fa7ce4f2c877bd7cfe1732704\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e7ba4a3d5fb4ee19815f33b6fd3d08d : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/966d230e9e9a435c90fefe2b6f6a3051 : SUCCESS in 27m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/61c246708454468ca6d30de6f6f638fc : SUCCESS in 1h 03m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1b294d5704c540f18aea558a9d4ab8b9 : SUCCESS in 51m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/787c6553c5164414ab5242b3579d40cb : SUCCESS in 1h 03m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4d6a5aa072b04ec6ad4689da64309d1c : SUCCESS in 7m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4e5971c6935144b8a25ef2fc6e640fe6 : SUCCESS in 21m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/df8ba52978934206b9a20de178e34584 : SUCCESS in 19m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5e90689c0f2c43089dcb10ba87cb7c3e : SUCCESS in 25m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6be412a3ee404593b59ac5b22d39f5be : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a4bf1ab4487c49349efbe42456dc3191 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/65c73dca36564f63a4761c7c973c3779 : SUCCESS in 8m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3c255bfe277447a29b544b6361091b0d : SUCCESS in 4m 24s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1e541cfdc6b1470a9f2a742540c32327 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cfd7483bb2d4d648a83b64facff0aec : SUCCESS in 32m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eaebcd91cc6d4447b7e761f0e10f317a : SUCCESS in 30m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/be291f57e2ee4216acf01b299d520a2e : SUCCESS in 21m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93a461c859dd4af0b44611b21b7969f5 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd7f2499f984ccfa5e1773ebb046153 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a357da8f63d646ffaad8bd4ef9daee6b : SUCCESS in 8m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/163eb901633d4a2096846c5581903fc1 : SUCCESS in 29m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba3da2abfb6441e9a2d8c1247bbe67ed : SUCCESS in 27m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9fd1af70c38c2935c00b811f017eb8558d11d654","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 14:44:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b8ba5c8311af688a648b39837dc0ce0532182369","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 14:45:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acd48a37caa44140806b9fb559c1377f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce36208e7a34cd492b2baf16926aba8 : SUCCESS in 1m 02s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b6f6aa60ad894ba1a3450eac1d8ddb43 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e05ce306b7f20d7d486289fd3e90dd97e7fc1bec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 15:49:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaf8a0c9e5cc4e958d42cdd727c7621d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ace3bbbcd61489289653670444330f1 : FAILURE in 2h 59m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ef7059640f6441ab4c9e32193fec1a35876b263f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-21 00:36:02.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74274ec8bccc47c80f7302e3ce14e430826fe1ac","revisions":{"74274ec8bccc47c80f7302e3ce14e430826fe1ac":{"kind":"REWORK","_number":1,"created":"2025-08-14 10:31:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/12/957412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/957412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/957412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/957412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/957412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/957412/1"}}},"commit":{"parents":[{"commit":"bdb052b59b6b36ac57792248a637e81cd973bf95","subject":"Merge \"ssync-receiver: terminate session if subreq read times out\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdb052b59b6b36ac57792248a637e81cd973bf95"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-14 10:14:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-14 10:14:41.000000000","tz":60},"subject":"checksum.py: fail gracefully if pyeclib is broken","message":"checksum.py: fail gracefully if pyeclib is broken\n\nIf pyeclib dist is missing files then the isal loading would blow up\nwith \"TypeError: \u0027NoneType\u0027 object is not iterable\". This patch\nchanges that to a RuntimeError with a more useful message.\n\nChange-Id: I64a85eb739fb72efe41f1ee829e463167246b793\nCloses-Bug: #2120591\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74274ec8bccc47c80f7302e3ce14e430826fe1ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74274ec8bccc47c80f7302e3ce14e430826fe1ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
