)]}'
{"id":"openstack%2Fswift~957427","triplet_id":"openstack%2Fswift~feature%2Fmpu~Id3798c6c2b70cbdbe2730af13d9c580017231133","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-s3-compat-versioning","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-09-02 08:58:07.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Id3798c6c2b70cbdbe2730af13d9c580017231133","subject":"mpu: refactor object-versioning part 1","status":"MERGED","created":"2025-08-14 13:04:02.000000000","updated":"2025-09-02 09:38:09.000000000","submitted":"2025-09-02 09:35:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957427-p-mpu-s3-compat-versioning","meta_rev_id":"318530f4bcf5dd5973259071f8de547ff6079b43","_number":957427,"virtual_id_number":957427,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-02 09:35:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-09-02 08:58:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-09-02 08:58:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 19:40:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-15 07:23:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1562963a73b57377d97bd522f52afcac92876afe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-14 13:04:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f7be00edb3eb31da16aae7423fa716f5599cd1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-14 15:50:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"148c21d8631a35f6487cbdf0ffbc2fe9964dfb95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 19:40:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce2dd66620d84e8286996ae63f66d722\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f2c2eb27a8a40f1aa3ee18d5c9f4c14 : SUCCESS in 1h 56m 38s","accounts_in_message":[],"_revision_number":2},{"id":"d26a8168e3d4b710c6cbc4da24ca7f18be802e67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 07:23:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4ed58032664443aa00527f43795e91d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e14d41c27cdb44e8b231fbfcc16c6716 : SUCCESS in 5m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f1e21a783ab4b1aa787656962d52f25 : SUCCESS in 10m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09c3887b9424498098b66986fb9e8e87 : SUCCESS in 14m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c501cd21f5414ae899b63b7f61d8a810 : SUCCESS in 11m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/048fe109a35345c99126c9ecd71d993e : SUCCESS in 4m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ea9765ecfafc47f8849d0eb22646d678 : SUCCESS in 8m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/89cfe82855fd4f2490351a3eff8eaf3e : SUCCESS in 19m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2754bcf787e34801ae278005d79b15f4 : SUCCESS in 12m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c33716e9b4954b8b88ed79da7eb1dafb : SUCCESS in 11m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e7b0e218d34a4b3bafb39e1f4857aa5d : SUCCESS in 8m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4f92f76249cd4a97a253e54fd90e7115 : SUCCESS in 8m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7755e7a9459a42c7b112bad4858f5322 : SUCCESS in 9m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1262817d7a34426bb5e868eec6a5595 : SUCCESS in 3m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9977556a14354f78acbf20ad8beba3e2 : SUCCESS in 14m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7eff80dfd45a44348bf4ade2d5c8af7a : SUCCESS in 5m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4efcbd2b2b284120b30c3fd9caec7249 : SUCCESS in 35m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/699ebeec7fcf44a8bfe30c7c87d126e1 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83d4e4757ed6407a9a0f55d882222523 : SUCCESS in 2m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d46de1befb8e4db5bfdc0f161c45b18c : SUCCESS in 21m 47s","accounts_in_message":[],"_revision_number":2},{"id":"96eaaa21fc7c5fc76dc73aee4451f52b34b39bb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-15 13:34:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6c21770f09b1d7ae855852ea8ad20e9abe387e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 14:46:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f108875378c49f38f6ea8b1f7619331\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d0588c8a9b404585997e5549fc6c6e : SUCCESS in 10m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bdfccaa0db334965bac69dd9a9c4c45b : SUCCESS in 12m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5595d489bfc4fe2b9f7006eb1f4b7df : SUCCESS in 11m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d7e6b91c38c4d03b256a3aab0bb9922 : SUCCESS in 17m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4a04ecc9814a43999f2f914be30d8179 : SUCCESS in 3m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/226e3b2f8e814dfeb9bd419e7abe37ed : SUCCESS in 8m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6168e7c820aa44569f28257f47854d72 : SUCCESS in 7m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/117dd035aad4451e8223fc29e40e2d9a : SUCCESS in 10m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c536ffa5baa349edb233cbf33a054bea : SUCCESS in 10m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/815d167f5d6244bf84de5316c65f20d8 : SUCCESS in 7m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5cd9597274a346a1aa1e85634ab0faf7 : SUCCESS in 8m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9559bb8d8f564c158c660b56d4debb15 : SUCCESS in 8m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2cb74bcc51124b85986c25e885e6331a : SUCCESS in 3m 48s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eb29d4bf0d7540f3ac7c4a065f958fe3 : SUCCESS in 11m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6431d5288c064999b555315274d181b9 : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/04ed2793e08b425081b42bc830ee36e3 : SUCCESS in 54m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75b7837304954a2ab96621e50149ef33 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8d71463f1a94482bca719a2fe5cded3 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e2874c1ce6a4b0b9654182504a59fa3 : SUCCESS in 8m 44s","accounts_in_message":[],"_revision_number":3},{"id":"22a40be72120881715c5c7c515097a01b89a6f2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 18:03:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df371453a21146db9fea4cbb6ca09e7c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/65caf48650514c14b494fd302e53c1c4 : SUCCESS in 1h 38m 25s","accounts_in_message":[],"_revision_number":3},{"id":"a646307cfd79a10b8a46dfa14ecac77b345001cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-19 13:15:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"288bf8745f8b2fed19c06c4842a9783b80518680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 15:22:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1b718410d734891ad50ea940e20bac5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f699316ecaf4e6396183370fc0bc4ad : SUCCESS in 4m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0a4af77bcd244baa17403693eb94d7e : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bac94c3002234ef3bd544fe5e64be0a9 : SUCCESS in 24m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f8804013c984e8199104d6ef9d2585d : SUCCESS in 23m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5cec29b1fc8a4d2e905a372fa1190812 : SUCCESS in 4m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/92786d4bf6a7429c9ad51c9095e091e7 : RETRY_LIMIT in 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/91798104f81245bdaed7e45439284882 : SUCCESS in 11m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8314577da9f14ec6b1b5cda6c4ea9417 : SUCCESS in 10m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/58676a2925ca46f1b2259d96ca4d72c1 : SUCCESS in 11m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e8430665604d4552b7d68e8d16ebfe1c : SUCCESS in 8m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ceaabbb609554b178a410618e0b92e7f : SUCCESS in 19m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0e6d7c3e81fa4ad8ac8c7132da5e2e7a : SUCCESS in 8m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/980449c1ba7e4419b215035549d0c95b : SUCCESS in 8m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7f67cc88a8734344a2f258ecf94ec2e5 : SUCCESS in 13m 40s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/78c63701e8584006a1855f60437702d5 : SUCCESS in 9m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2d39b0487944a7792b88e8401c06e01 : SUCCESS in 1h 14m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/554c6c537d3b4a35a3e62de3d353b77d : RETRY_LIMIT in 8s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d794c039db54aa0b0f120ce325f0994 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3181a9c6dcf4454b837d8da47f26abb6 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":4},{"id":"81c1fc9399a473d3398d8741669f6c66fa0b9b99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 16:20:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd34572c060d4cce8b7ba16cb7c3dc60\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f0a4a4b0d634fd4951a1d6a6b5c7210 : SUCCESS in 2h 28m 36s","accounts_in_message":[],"_revision_number":4},{"id":"353338fe4c4e491a29a06eff2aa8b58f7ea6d1d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-21 15:30:03.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ca72661c001c044c29545bc577dcab9f5382dedb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 16:36:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cb721e77c3d4241ba1e6e862a1a64ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c443387dbb2b4424ad8af57a37aac861 : SUCCESS in 11m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9085678f4a7a4ad9af5b8ae10e8bf57e : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55b62aa7f0464c3b8f5f33c6fc07f9a0 : SUCCESS in 21m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45ed31a3b863455189d5673b8e1fff32 : SUCCESS in 14m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/acb8db06eb9c48d2b2edb6b4fe16aa77 : SUCCESS in 4m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8d13cb4a8f984d74b4fc12cc82be1b09 : SUCCESS in 27m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/341df42cb81049888cea6489bf769d8d : SUCCESS in 20m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/af0794254a094c329dce403c422191c7 : SUCCESS in 10m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5ee4655a823c4b719445b3c27f0e0e19 : SUCCESS in 20m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0876ed2b48db4abea66959373364d789 : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bf8312d5c57f4209957f19ef410cec5d : SUCCESS in 8m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ae25ec3fa43e425bb59a7fd6e58a037a : SUCCESS in 16m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b945fb519b1840929cf7c1ce02b5adf6 : SUCCESS in 6m 48s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd9b5041d175448a846135e0cf654147 : SUCCESS in 8m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9de1d5f98ba54981b38215a3ee945edb : SUCCESS in 8m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4eeca2ef3f84407b678d8f4474035ee : SUCCESS in 1h 01m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8a02b0e21cd41da816b61564be4b2b1 : SUCCESS in 26m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/954c771c947148e29074fcd851451aa1 : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d32fdc9363f14127a543957e1f99ded2 : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":5},{"id":"37ce6531ea8d49fb78b2cab3132478b1f85f88d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 18:43:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/430f369242ec4231ab1e5d99b0c86420\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/93bb68f386984b05a73dc076545c1b67 : SUCCESS in 2h 13m 27s","accounts_in_message":[],"_revision_number":5},{"id":"d1fa7b9add330b2cffb0015b51cc29d114f00070","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-02 08:58:07.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3c278216b70c6b241e7647fb3449461f1ea7c63c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 08:58:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f66100662e31dec5fca963df57badd3aff3f3b8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 09:35:52.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d50d1fec4ae64793b8390faa713cb92c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d6cc1b57204e068bb37eb8de694c0c : SUCCESS in 7m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5817ffcdad3a40aca2e7a7b57a893f00 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/72b65f388da646f4be22506c43aceedb : SUCCESS in 20m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/31cb88de56e0498eadb16ef71d57bf48 : SUCCESS in 8m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b8a01b31225847b591a711a92c2fc64e : SUCCESS in 21m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/001a09c621534982882f732eb0f3371b : SUCCESS in 24m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4b9aeea63df94ea49482719a55005e4b : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ba4c1876ac15424b8eb3358ebdf59a16 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/aeb1d2caef4747f992353898a437aa04 : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/19b6d3212d6d4505a7ebce625a8676d6 : SUCCESS in 3m 57s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e3e8aa78305c446885b5e0e59c3fc1fe : SUCCESS in 8m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/74b5dace6c164ef6a361b9e537831d4c : SUCCESS in 35m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd23690dec5642b5a92e4694d7628315 : SUCCESS in 26m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f204c604cb64b3abc024dfae1e09c2c : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3022ca9c71fb4ece87d0481d50cdc705 : SUCCESS in 24m 38s","accounts_in_message":[],"_revision_number":5},{"id":"71a0b0abc4695e216744dee2d23ece636ec1eab6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 09:35:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"318530f4bcf5dd5973259071f8de547ff6079b43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 09:38:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f1d0aef41d1423294092f6e5115f392\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54758585050c4dee9324194533de33da : SUCCESS in 2m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1b10fdaccd564130b546c5daf3ba6d57 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"13689f2d113688c2ff202abf9fc766d8a8fdeecf","revisions":{"f644c7aa9d87d27be64839b0c44cfbbaa48adf92":{"kind":"REWORK","_number":1,"created":"2025-08-14 13:04:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/957427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/957427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/957427/1"}}},"commit":{"parents":[{"commit":"95ac6e8178f7943492c52f29edb69cdbb4d45d80","subject":"mpu: refactor versioning s3api compat tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95ac6e8178f7943492c52f29edb69cdbb4d45d80"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-13 11:57:36.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-14 13:02:46.000000000","tz":60},"subject":"mpu: refactor object-versioning part 1","message":"mpu: refactor object-versioning part 1\n\nMake the concept of version id (which is based on timestamp) more\nabstract.\n\nChange-Id: Id3798c6c2b70cbdbe2730af13d9c580017231133\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f644c7aa9d87d27be64839b0c44cfbbaa48adf92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f644c7aa9d87d27be64839b0c44cfbbaa48adf92"}]},"branch":"refs/heads/feature/mpu"},"ac0bcd528239f0958fb3342f4d90ab93170b2140":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-14 15:50:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/957427/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/957427/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/957427/2"}}},"commit":{"parents":[{"commit":"355340801146e188aabf8a79e40a25ca3aa78d7b","subject":"mpu: refactor versioning s3api compat tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/355340801146e188aabf8a79e40a25ca3aa78d7b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-13 11:57:36.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-14 15:49:39.000000000","tz":60},"subject":"mpu: refactor object-versioning part 1","message":"mpu: refactor object-versioning part 1\n\nMake the concept of version id (which is based on timestamp) more\nabstract.\n\nChange-Id: Id3798c6c2b70cbdbe2730af13d9c580017231133\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac0bcd528239f0958fb3342f4d90ab93170b2140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac0bcd528239f0958fb3342f4d90ab93170b2140"}]},"branch":"refs/heads/feature/mpu"},"7b3f7611f6de34dd382bc3e56004d96e759c16fa":{"kind":"REWORK","_number":3,"created":"2025-08-15 13:34:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/957427/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/957427/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/957427/3"}}},"commit":{"parents":[{"commit":"7a0a975957b7b89bf1b0abb060dc129771f544be","subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a0a975957b7b89bf1b0abb060dc129771f544be"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-13 11:57:36.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-15 13:21:18.000000000","tz":60},"subject":"mpu: refactor object-versioning part 1","message":"mpu: refactor object-versioning part 1\n\nMake the concept of version id (which is currently based on timestamp)\nmore abstract.\n\nChange-Id: Id3798c6c2b70cbdbe2730af13d9c580017231133\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b3f7611f6de34dd382bc3e56004d96e759c16fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b3f7611f6de34dd382bc3e56004d96e759c16fa"}]},"branch":"refs/heads/feature/mpu"},"9bdeff24e552039847400570b3b6dd2791dcac65":{"kind":"REWORK","_number":4,"created":"2025-08-19 13:15:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/957427/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/957427/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/957427/4"}}},"commit":{"parents":[{"commit":"7a0a975957b7b89bf1b0abb060dc129771f544be","subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a0a975957b7b89bf1b0abb060dc129771f544be"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-13 11:57:36.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-19 13:15:31.000000000","tz":60},"subject":"mpu: refactor object-versioning part 1","message":"mpu: refactor object-versioning part 1\n\nMake the concept of version id (which is currently based on timestamp)\nmore abstract.\n\nChange-Id: Id3798c6c2b70cbdbe2730af13d9c580017231133\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bdeff24e552039847400570b3b6dd2791dcac65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bdeff24e552039847400570b3b6dd2791dcac65"}]},"branch":"refs/heads/feature/mpu"},"13689f2d113688c2ff202abf9fc766d8a8fdeecf":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-08-21 15:30:03.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/957427/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/957427/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/957427/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/957427/5"}}},"commit":{"parents":[{"commit":"42b2da8a6d63c5323a711113fdb858e618bfd0eb","subject":"Merge branch \u0027master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42b2da8a6d63c5323a711113fdb858e618bfd0eb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-13 11:57:36.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-21 15:29:52.000000000","tz":60},"subject":"mpu: refactor object-versioning part 1","message":"mpu: refactor object-versioning part 1\n\nMake the concept of version id (which is currently based on timestamp)\nmore abstract.\n\nChange-Id: Id3798c6c2b70cbdbe2730af13d9c580017231133\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13689f2d113688c2ff202abf9fc766d8a8fdeecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13689f2d113688c2ff202abf9fc766d8a8fdeecf"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
