)]}'
{"id":"openstack%2Fswift~957428","triplet_id":"openstack%2Fswift~feature%2Fmpu~If901c4e8fe9e23a9f6515e4626c6aac04194c98b","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-s3-compat-versioning","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-09-02 08:58:27.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"If901c4e8fe9e23a9f6515e4626c6aac04194c98b","subject":"mpu: refactor object-versioning part 2","status":"MERGED","created":"2025-08-14 13:04:02.000000000","updated":"2025-09-02 10:01:20.000000000","submitted":"2025-09-02 09:58:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957428-p-mpu-s3-compat-versioning","meta_rev_id":"7b6384c272acb87bf4132b034f62dcce9a16e94f","_number":957428,"virtual_id_number":957428,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-02 09:58:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-09-02 08:58:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-09-02 08:58:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 18:55:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-15 05:22:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9300c61195ab35bc88a9c880dd7024a5a8f10f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-14 13:04:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91249c5c25f32af138e33228c0d3d2321fa30c8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-14 15:50:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"150529c7378720ad35a442d16ac426f0b787c2a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 18:55:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7949c9cd09a4afeb97b19c54de4e65a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba8e65b1188a4970bba70c95a60c17a2 : FAILURE in 1h 53m 22s","accounts_in_message":[],"_revision_number":2},{"id":"b2ab531052b8c25ef9985e937c2f01381ed0d4fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 05:22:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/856f2829a3624d81a33cfff3295e765b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/403a874088a6466c80682fbde9ca334b : SUCCESS in 6m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51358fa0a3f64837b53c9f79897c5ca4 : SUCCESS in 14m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54fe281777d44783865168927327dc00 : SUCCESS in 11m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8adc46ff01d4464493daa3784e1db2ab : SUCCESS in 10m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/01acecbcabfd431b8924c93146cbd35a : SUCCESS in 4m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2b1d7671cba041ba85115316eaf26d89 : FAILURE in 9m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c4b368b107f34e7688bc31499ab04f6b : FAILURE in 8m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4af2fa94dd03427d934a93321d26bb6c : FAILURE in 15m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bb05b4a6f6364b97bf24980e3ca6709c : FAILURE in 11m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/92cd1dc9dac840c7b799cbe76a01310a : SUCCESS in 7m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cd679266ff5142f6af9f750281b884b2 : SUCCESS in 8m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/555e542e86174c4a98509d9f13640115 : SUCCESS in 8m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/52016e06ebc5484bb85ba15cadcd3db9 : SUCCESS in 3m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aebcdc724d514a2f83d57797c575f197 : SUCCESS in 9m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3ccd5f965e5b4eb889c7f134d67d415c : SUCCESS in 4m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9b904c5acbd4afe998a4603524c1e10 : FAILURE in 47m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adc67b792ee14a5db355e353e89814e2 : FAILURE in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f6a15828c34794b094e1bdb11dcc20 : SUCCESS in 2m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c37a4bc0b80e437cb2828e5f6649590e : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":2},{"id":"0c4c9853ea8143a9e056b200eac7e2d95f558390","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-15 10:22:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56add64f87a5626d040a702aa7b58177d5779c04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-15 13:34:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fe6bcad431512f285caaf536ae7ce3bcd0d3f776","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 14:20:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0388caf51160475ebb5ef9b0a9388a7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0db66f27f4a4c5080cf2572ad5a3823 : SUCCESS in 4m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cacf6f0328d4f2d964024d44445da76 : SUCCESS in 22m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5be22d24165147e7925851186aa4da84 : SUCCESS in 21m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab2b89ba484d41d6878f31e08a9c120c : SUCCESS in 23m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4c74591eb98943aca1d36fc3a7f0cb4f : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0d873a13ab444ec5b2bb2e2400c589b6 : SUCCESS in 10m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d8b271af2d884b8dad98fc49c0f3eeb8 : SUCCESS in 20m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e586309ae4a240d6a5e9bf81e45a703e : SUCCESS in 12m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/eb7b25b006b64119830dc1d0cd135448 : SUCCESS in 27m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7db223a2167c44a28735c54ba7773ed1 : SUCCESS in 7m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6b928a94d5854b229008584d734bff15 : SUCCESS in 9m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9618ecbc30834d80a4686ecd9434cd0f : SUCCESS in 8m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5f181aa0bb3f48d58d9a0ec5eb70f656 : SUCCESS in 9m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e74f6e2ce8b64b0f8d0c46e17e1ba171 : SUCCESS in 9m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e8e242695e9249b58cb43ea27f7a835f : SUCCESS in 4m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f4599de397d45fba188383f15fe3b49 : SUCCESS in 34m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b33f921714af4b68a5af4dd777454862 : SUCCESS in 21m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38e4664adc18420583724d9537235b57 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/601ac77ec4be4a09aa1ff2b61ab8d9c6 : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":3},{"id":"bd8e43754745dea6341049faa73e42baea87eb5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 15:29:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4053ad754ce341998b92e4049139f71f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d461ab0605947f7ba930ae08b4020d7 : SUCCESS in 1h 49m 14s","accounts_in_message":[],"_revision_number":3},{"id":"a3b09bd66de4fbaac1fcfcd14595930f49f1f8db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-19 13:15:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"86e4031b5f430512df9aaacb3263e78cf1bc8f4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 14:29:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a635da3ab15e4662ba147a98a7568b27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec1225a7f1e459fb784517e310dae3a : SUCCESS in 5m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3baa4c01682438e939da9a61613a330 : SUCCESS in 23m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9a7d8a48797433db4d873525c3136bc : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1d030a56a5846d78eacaf91931e6c60 : SUCCESS in 23m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e28b615e03a44240a32e2f5889b0c7c0 : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f04244db81e34d05bbdd6318a004f8e8 : RETRY_LIMIT in 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c60b80ee76a248dfafce4232a3058e4a : SUCCESS in 8m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f43a5dec1211417e80e570b2132d7fbf : SUCCESS in 27m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/44eab68ddb8b4742889742fab73a8a7d : SUCCESS in 27m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/541c9aca183c449fb27ef7d3dd66ad68 : SUCCESS in 10m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/28df4d8beed140b090616c6edf52886d : SUCCESS in 16m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f6768de2142246de94a6af60688987b4 : SUCCESS in 19m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a6ba67c5fb7466e96e264eab6aaa29a : SUCCESS in 6m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/791b4c9488a14092b78481c6d5676147 : SUCCESS in 9m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/702aa40d0b5445f083893d1970815bda : SUCCESS in 6m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b904db7556be4b4e88b78c899774d5af : SUCCESS in 1h 01m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9e0d5b0d6c746a297def030685b3685 : RETRY_LIMIT in 9s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adf3e0ee58a04a9e8aedf3be8ea322dc : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d26cec123b2f45c1b8658eb73203e4ab : SUCCESS in 23m 38s","accounts_in_message":[],"_revision_number":4},{"id":"8294b6ec27f0503078401a3651a144f311039a6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 15:47:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaa652b7298b49ad93b47a99e9a1f10d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/03c854db7b3945959b97c9cd08bfa844 : SUCCESS in 1h 13m 46s","accounts_in_message":[],"_revision_number":4},{"id":"d40ff6f8e123d2befb6736678d6545cdd59ebbae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-21 15:30:03.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"34938832d917115230f808a78ece101bdcad633f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 16:37:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ebbd097f59d4e7ea9019298811c17a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f20a9c976864677a5f07a20cd1788ac : SUCCESS in 10m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2a36d632db2440aa9eb3663318438e4 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a2a0b547cc84a4cb3f54832540ce4af : SUCCESS in 21m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db6178566021453eb2d453a9aafd4349 : SUCCESS in 22m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2599c4abdd9a4442b392da36df9b697a : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a0062594c545415f90c9cd5011c98aa0 : SUCCESS in 27m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d18ae6da2e3a4e04a8b857a58a3edcb1 : SUCCESS in 21m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6d80b8a058fd4b5481f0e9e61d2d1e2c : SUCCESS in 29m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/70932d3611814d17816c8401edaa41ea : SUCCESS in 26m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/76518fc8cf79484ebdb63a588902ff64 : SUCCESS in 19m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/99e632e9ba84447eb211118e1e82b65a : SUCCESS in 17m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8c8250b1a88641e18a13b947dc5fbfd5 : SUCCESS in 18m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/97440edfaa5d4cbf87952498583f957e : SUCCESS in 5m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/56f600993ed24a9780e181568f77c118 : SUCCESS in 14m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/35f839c4a0a8431bb390951b6c7c4597 : SUCCESS in 8m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca47dfb87f904fba9b7dbc4717872aaf : SUCCESS in 1h 02m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc699350cb004719b4529eea6898020d : SUCCESS in 26m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e69bf9f2a540b38f7d3fea894c6c13 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b5b3b1e80ef54d1bb066fd3a2b543f58 : SUCCESS in 17m 47s","accounts_in_message":[],"_revision_number":5},{"id":"4df7dd810a899158782d000152106cb9d5046968","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 20:42:42.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8589aad7bb7e4cb5bcaf24c232ce1e42\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/76c99a9c04bb4df6823942963945db41 : TIMED_OUT in 3h 01m 15s","accounts_in_message":[],"_revision_number":5},{"id":"02692ca97645cb04d92d9de9d8e60441536d381d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-02 08:58:27.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0a22ed856e348af09afc37dfcd8882c82b017154","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 08:58:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0e6ad883dd3de5636de81766ef65bd985731cccd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 09:58:50.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4da2bef82338481f9608e24f3253bf75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b125637e8ca4702a5dc218219dbfadf : SUCCESS in 9m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bef239ce743b44c597df6c2bb3b6e073 : SUCCESS in 4m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e9a6adce2e924b4abba02a13bd5c05fc : SUCCESS in 21m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3a930e899d5345318d8653257eeaa595 : SUCCESS in 19m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/05519c00891044bcba68af7bd52a0a22 : SUCCESS in 20m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/61d230cb3e594940829947e09afd94bf : SUCCESS in 12m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fb1dce3a4e494c9580667055dcfeebf5 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b126f7a7a1884a3894dbd2be09542e67 : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8d14622cb1c645c18d018ff1611fdf89 : SUCCESS in 9m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a29a7275e56d46769944779cfbc6f307 : SUCCESS in 4m 01s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ea7cbd5b0ab84b6abd3730196bb28189 : SUCCESS in 5m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd05d5c0d25448f9981d5ab96bc421c2 : SUCCESS in 36m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14260bbc8c8247a9b52a89980c62c7cd : SUCCESS in 26m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8c04d64bd7b45d19b3f33c50e99f4b8 : SUCCESS in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3cee84c958284654a4b5eafe21272088 : SUCCESS in 25m 00s","accounts_in_message":[],"_revision_number":5},{"id":"720527368a242f4443ac36f5fff8457f576dc149","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 09:58:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"7b6384c272acb87bf4132b034f62dcce9a16e94f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 10:01:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2872ca7a4544424d96262c5fdc8ff4c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d1734a08e642e885ae2a8d29be5f78 : SUCCESS in 2m 19s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f51e9c0330244c0784c94771056fb9a8 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c5773a10c99bc45f23448dd9d4da8be6e04d9254","revisions":{"cf9cac7093b508048090d9b46a1cb0575de247d3":{"kind":"REWORK","_number":1,"created":"2025-08-14 13:04:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/957428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/957428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/957428/1"}}},"commit":{"parents":[{"commit":"f644c7aa9d87d27be64839b0c44cfbbaa48adf92","subject":"mpu: refactor object-versioning part 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f644c7aa9d87d27be64839b0c44cfbbaa48adf92"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-13 14:48:09.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-14 13:02:46.000000000","tz":60},"subject":"mpu: refactor object-versioning part 2","message":"mpu: refactor object-versioning part 2\n\nGive ObjectContext some variables to avoid passing them around between\nmethods.\n\nChange-Id: If901c4e8fe9e23a9f6515e4626c6aac04194c98b\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf9cac7093b508048090d9b46a1cb0575de247d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf9cac7093b508048090d9b46a1cb0575de247d3"}]},"branch":"refs/heads/feature/mpu"},"80767c481df42f7b30fc180e708686497d7a384b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-14 15:50:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/957428/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/957428/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/957428/2"}}},"commit":{"parents":[{"commit":"ac0bcd528239f0958fb3342f4d90ab93170b2140","subject":"mpu: refactor object-versioning part 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac0bcd528239f0958fb3342f4d90ab93170b2140"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-13 14:48:09.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-14 15:49:39.000000000","tz":60},"subject":"mpu: refactor object-versioning part 2","message":"mpu: refactor object-versioning part 2\n\nGive ObjectContext some variables to avoid passing them around between\nmethods.\n\nChange-Id: If901c4e8fe9e23a9f6515e4626c6aac04194c98b\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80767c481df42f7b30fc180e708686497d7a384b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80767c481df42f7b30fc180e708686497d7a384b"}]},"branch":"refs/heads/feature/mpu"},"71480ee675544d6157e6c52f05684d72d439ce01":{"kind":"REWORK","_number":3,"created":"2025-08-15 13:34:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/957428/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/957428/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/957428/3"}}},"commit":{"parents":[{"commit":"7b3f7611f6de34dd382bc3e56004d96e759c16fa","subject":"mpu: refactor object-versioning part 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b3f7611f6de34dd382bc3e56004d96e759c16fa"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-13 14:48:09.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-15 13:27:49.000000000","tz":60},"subject":"mpu: refactor object-versioning part 2","message":"mpu: refactor object-versioning part 2\n\nGive ObjectContext some variables to avoid passing them around between\nmethods.\n\nChange-Id: If901c4e8fe9e23a9f6515e4626c6aac04194c98b\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71480ee675544d6157e6c52f05684d72d439ce01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71480ee675544d6157e6c52f05684d72d439ce01"}]},"branch":"refs/heads/feature/mpu"},"35060c40f5de744a2314f727c4b61127ca8f8a17":{"kind":"REWORK","_number":4,"created":"2025-08-19 13:15:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/957428/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/957428/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/957428/4"}}},"commit":{"parents":[{"commit":"9bdeff24e552039847400570b3b6dd2791dcac65","subject":"mpu: refactor object-versioning part 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bdeff24e552039847400570b3b6dd2791dcac65"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-13 14:48:09.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-19 13:15:31.000000000","tz":60},"subject":"mpu: refactor object-versioning part 2","message":"mpu: refactor object-versioning part 2\n\nGive ObjectContext some variables to avoid passing them around between\nmethods.\n\nBreak out/rename some methods for clarity.\n\nAdd an is_versioning_enabled helper function.\n\nChange-Id: If901c4e8fe9e23a9f6515e4626c6aac04194c98b\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35060c40f5de744a2314f727c4b61127ca8f8a17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35060c40f5de744a2314f727c4b61127ca8f8a17"}]},"branch":"refs/heads/feature/mpu"},"c5773a10c99bc45f23448dd9d4da8be6e04d9254":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-08-21 15:30:03.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/957428/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/957428/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/957428/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/957428/5"}}},"commit":{"parents":[{"commit":"13689f2d113688c2ff202abf9fc766d8a8fdeecf","subject":"mpu: refactor object-versioning part 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13689f2d113688c2ff202abf9fc766d8a8fdeecf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-13 14:48:09.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-21 15:29:52.000000000","tz":60},"subject":"mpu: refactor object-versioning part 2","message":"mpu: refactor object-versioning part 2\n\nGive ObjectContext some variables to avoid passing them around between\nmethods.\n\nBreak out/rename some methods for clarity.\n\nAdd an is_versioning_enabled helper function.\n\nChange-Id: If901c4e8fe9e23a9f6515e4626c6aac04194c98b\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5773a10c99bc45f23448dd9d4da8be6e04d9254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5773a10c99bc45f23448dd9d4da8be6e04d9254"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
