)]}'
{"id":"openstack%2Fswift~957870","triplet_id":"openstack%2Fswift~feature%2Fmpu~I8ff0ea9fd87b1ff800f7b8e4e4f60a05d0ae0796","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-08-20 08:55:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8ff0ea9fd87b1ff800f7b8e4e4f60a05d0ae0796","subject":"Merge master into feature/mpu","status":"ABANDONED","created":"2025-08-19 10:43:12.000000000","updated":"2025-08-20 08:55:20.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2efae0a9294101c46befc0ab6724e17525b5e432","_number":957870,"virtual_id_number":957870,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-19 11:21:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-19 11:21:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7f507fc35c8800ae5813ed61deeaf8f38546274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-19 10:43:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f48d9868605819f7996af26c4ba09aa5a6827186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 11:21:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd6d8111cfc444f8b06b2f5f62ea81c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c29778bfe97a476ca4a30a3ab076af5f : SUCCESS in 11m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b7b6b6517824e9e8e6d55e6dd9cacc0 : SUCCESS in 22m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e409d6ad47ef4cb29439bc57f7b502f7 : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d40f35228a7b4745986851abf516c4a8 : SUCCESS in 22m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/220b3920f1914666a4cca028ce2c26bb : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/490033fda82546559b92e54bfd4b13f5 : RETRY_LIMIT in 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/86dbc74d15e142779e0cbafe6c5a34ec : SUCCESS in 19m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d3391237042b4b5ca1863a54d4e5ff92 : SUCCESS in 23m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8aa60adc0d1648a1b8a6f97eca35033d : SUCCESS in 25m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/42937a4a77f645408b571907d09ac751 : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bf6a9c65bacf40c5846169c790896e19 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f72ea7ed8e874ccc950cf3654a46a528 : SUCCESS in 17m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ae31c60e9e674456b8d262036792d97c : SUCCESS in 8m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b476510b281f4bf68fe9085b296b3857 : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e68bb28275894440998a9d6a457f252c : SUCCESS in 11m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8edc7c401598447dba4f624937066bc8 : SUCCESS in 33m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7883890b264241ae85c0076febf3b385 : RETRY_LIMIT in 9s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c4d94558ec543cab5dafe082ec0d196 : SUCCESS in 3m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/822af39431f14f08bb8f6afee55e5ae1 : SUCCESS in 26m 57s","accounts_in_message":[],"_revision_number":1},{"id":"09f3404f241cd1347db1d11b9683bb18ac977a2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 11:51:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d77daae5fa1d490da1c4aec9083b6fe9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c898dad666cd427e8682bbfe0488f4ca : SUCCESS in 1h 05m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0c82b99d7986af0c6a0935c9186b67c4a3f8c71f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-19 12:02:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dac53ccde60d490214fc75f63ff6d5bf64db36ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 13:08:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61b1ac38501d4fad99925bbe62aef380\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc90d2f6492d4158a0685677801bde94 : SUCCESS in 6m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53906f557f614bb68856bd03f0bcf153 : SUCCESS in 22m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48f31bdd46ed4e8e909049c6bc87f711 : SUCCESS in 10m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1c093349b19412d8cb57d6eed4eeb41 : SUCCESS in 23m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9371621c98c64de783ac31b7220f04d3 : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/564adb3e974a4adb9a3bb49aab9765ad : RETRY_LIMIT in 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/73b8008e30b141c9bdd9b63c90069c6d : SUCCESS in 8m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/40375aeb7c6847fbb805545a67653a96 : SUCCESS in 10m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c4d074838793441b8e67ec4c11a9c7ae : SUCCESS in 10m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6c2a19b1ec6a431693471488950c966b : SUCCESS in 20m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/931b10640f0d4ec09bd8f8cbe94032c0 : SUCCESS in 9m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/56a4f189031449d5aab1e42084540761 : SUCCESS in 17m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/46c8003d4fec4f4e87853bb3863ed837 : SUCCESS in 4m 48s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/42bb423fba1743ba8e10be156f0db3bc : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cb57573d5cef493dbc1cb0547e2cf0e9 : SUCCESS in 11m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0baf674847234527961fe64a9d054acf : SUCCESS in 1h 00m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7e94134e8654f4e8b34db3710830c0e : RETRY_LIMIT in 8s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab9d93a28c0a4eca9d4585da56d5a960 : SUCCESS in 2m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5bb62e64c1664066a08f802cd0581f17 : SUCCESS in 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b383f83c66ae2de658b091f923970ce215f187f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 14:04:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52ffaca082a649b390efd9d0acee9cc2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f02bd18b7056492f86e82bce18a8e818 : SUCCESS in 1h 48m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2efae0a9294101c46befc0ab6724e17525b5e432","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-20 08:55:20.000000000","message":"Abandoned\n\nsuperseded by https://review.opendev.org/c/openstack/swift/+/958006","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"419bf591f90fbb111e8829b33f11c6f62db6e399","revisions":{"419bf591f90fbb111e8829b33f11c6f62db6e399":{"kind":"REWORK","_number":1,"created":"2025-08-19 10:43:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/70/957870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/957870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/957870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/957870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/957870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/957870/1"}}},"commit":{"parents":[{"commit":"7a0a975957b7b89bf1b0abb060dc129771f544be","subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a0a975957b7b89bf1b0abb060dc129771f544be"}]},{"commit":"28a56e3f1b3527dabe9658c77d78eee400c4e9c2","subject":"Merge \"zuul: run py3.13 unit tests in the gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28a56e3f1b3527dabe9658c77d78eee400c4e9c2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-19 10:42:13.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-19 10:43:03.000000000","tz":60},"subject":"Merge master into feature/mpu","message":"Merge master into feature/mpu\n\nChange-Id: I8ff0ea9fd87b1ff800f7b8e4e4f60a05d0ae0796\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/419bf591f90fbb111e8829b33f11c6f62db6e399"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/419bf591f90fbb111e8829b33f11c6f62db6e399"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
