)]}'
{"id":"openstack%2Fswift~957936","triplet_id":"openstack%2Fswift~master~I16af4fc74974979bdf68390b42b2f571a8183a05","project":"openstack/swift","branch":"master","topic":"p-zuul-py37","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-08-19 20:54:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I16af4fc74974979bdf68390b42b2f571a8183a05","subject":"zuul: pin py37 job to ansible 9","status":"ABANDONED","created":"2025-08-19 15:05:33.000000000","updated":"2025-08-19 20:54:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1064a3d06b32dfb8e0595652f91772ead037f792","_number":957936,"virtual_id_number":957936,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-19 18:03:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-19 18:03:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f1b18fd7b057a2413f3f85daa433a0bd8ba6bd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-19 15:05:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a20a0c11d9463763f6ed569f004270d2ce5cad6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 18:03:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cad5e39762e4781a3ff67c2ef64ec47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2df9dc3729264a00a9acfc16e5573476 : SUCCESS in 5m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/f0f6a355f3ce45d18355340259b364b4 : SUCCESS in 1h 02m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b9e243c020f4613b75a656bc11d9a77 : SUCCESS in 1h 03m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/edb9eec7c2f94cbb8640c80deca22413 : SUCCESS in 30m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6642804af70845819a4fa69ba4ae62fd : SUCCESS in 1h 04m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34ce6636750240f3ab194dd131baccba : SUCCESS in 21m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0c54197109b428a8642b7b5b5091322 : SUCCESS in 10m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/603e30559afb4a419136c2f836fdfab2 : SUCCESS in 21m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bea70315281e44068ff20778c15e58df : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b0dd3870f78c43d1816d8eb17972e43f : SUCCESS in 8m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/31e6aa518d0e45eaa41c009184e95052 : SUCCESS in 18m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/842283e20dc44bf2b8cb1e8000bfe21e : FAILURE in 26m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/732f1c7dd4924fe492bd8442fbca0612 : SUCCESS in 25m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a196c7d9d74448958720580b1b8f7b17 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6dcae876384544d0ad060c1dfb02c543 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4343b93facba420ea838eb91cc8b6064 : SUCCESS in 18m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b266df89ef5427da74fdcbeace6cd0c : SUCCESS in 3m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5854248b3b904694aa062c87cbddea7b : SUCCESS in 7m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5ae7d150fdd14dda8dd2eccad32aa263 : SUCCESS in 8m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/59242ec7dc1646e3b18895a51febae04 : SUCCESS in 33m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0d4f4359bff74f13b1bcf0aa804b944d : SUCCESS in 56m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/35bcb40108484e2ea642f19987a3fed8 : SUCCESS in 56m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c67f4706f3c145aab3df6064dd8d95e4 : SUCCESS in 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88895f41f08144e883a078c38e010d99 : SUCCESS in 3m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d31cd3adf6842a0855d1e27401aa908 : SUCCESS in 18m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7fc559227ca42cd963aacbe6b1fa9c4 : SUCCESS in 42m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c129e0b64cc44f38ca397045389ffef : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b564590ad76f61367b97b9b6bee04230027b21c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 18:53:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/338260165fd040d7b75dae6e426cb4aa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64370dbe10ec453cb9a9f5983ef9c46a : SUCCESS in 1h 18m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1064a3d06b32dfb8e0595652f91772ead037f792","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-19 20:54:48.000000000","message":"Abandoned\n\nThanks for bringing this to my attention! I think https://review.opendev.org/c/openstack/swift/+/957959 might be a more sustainable approach, though.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bd36888024bb90224c29bd91b9600d3ad933399","revisions":{"4bd36888024bb90224c29bd91b9600d3ad933399":{"kind":"REWORK","_number":1,"created":"2025-08-19 15:05:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/957936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/957936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/957936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/957936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/957936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/957936/1"}}},"commit":{"parents":[{"commit":"28a56e3f1b3527dabe9658c77d78eee400c4e9c2","subject":"Merge \"zuul: run py3.13 unit tests in the gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28a56e3f1b3527dabe9658c77d78eee400c4e9c2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-19 15:02:48.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-19 15:02:48.000000000","tz":60},"subject":"zuul: pin py37 job to ansible 9","message":"zuul: pin py37 job to ansible 9\n\nzuul ubuntu-bionic runners don\u0027t support the now default ansible 11,\nso pin ansible to 9 for those jobs that require py37.\n\nChange-Id: I16af4fc74974979bdf68390b42b2f571a8183a05\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bd36888024bb90224c29bd91b9600d3ad933399"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bd36888024bb90224c29bd91b9600d3ad933399"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
