)]}'
{"id":"openstack%2Fswift~958122","triplet_id":"openstack%2Fswift~master~Ib2e57c29abd61279e6436818abd3219c5a8aab7f","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-08-20 20:53:41.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib2e57c29abd61279e6436818abd3219c5a8aab7f","subject":"tempauth: remove un-tested mis-feature","status":"NEW","created":"2025-08-20 20:44:04.000000000","updated":"2025-09-02 22:45:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"72952092f2ccebca7bef222d7b18cef343b4a8d2","_number":958122,"virtual_id_number":958122,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-20 21:59:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-09-02 22:45:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-20 20:53:41.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-08-20 21:59:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-02 22:45:50.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"25139f8cd90120546b00b32b2e1e0f12630114dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-20 20:44:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe399f1b89623e195d56ee88ad16bf112268eb13","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-20 20:53:41.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a3ce299c452da0ddb9eb6bd17c8eeb7a7c41e35e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 21:59:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfc6399d26dd41b494f5bd3e4cc96951\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb3210c2ceb42078b9c5640622e954d : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/d03c2a0920974147a196e78917f63da9 : SUCCESS in 29m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/036f9621943f43e29b4346cb6d6c3c0f : SUCCESS in 1h 10m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45631cb5248e44ebad1e8b8123100c02 : SUCCESS in 35m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e1360a146a04d99adb49a9ac774ce4e : SUCCESS in 59m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a1b7dc4d7464a59910299af6d93af5b : SUCCESS in 14m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8b5db5ffd1748a7a5a2bd38eac6c621 : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c005e703c9be4e9d8845182051ddfc5c : SUCCESS in 21m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/044ac269ee384ac08e80befd9fab2e56 : SUCCESS in 7m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5110f1b5cc8d4e38a82437287ea61981 : SUCCESS in 30m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c2c5faf0b2d04d1fbb3d5ebdc7f43477 : SUCCESS in 24m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a5281d1c219145189e232a89798d364e : SUCCESS in 28m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e858a7eaa9864b1492f143bd876dd929 : SUCCESS in 11m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/74cfd26082b645fd9553da723b5e0841 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/85fb296cc38f458791aa0d64943fd612 : SUCCESS in 15m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1496f5386a624a9cb3c92b04a9a6a5d3 : SUCCESS in 8m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/08b8234390be4a698266ea1f940af6a9 : SUCCESS in 8m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1a51bb7d4c83454b9c5e9dd8943416f5 : SUCCESS in 8m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/76bf5eeee3014c88baa620822d1e02b6 : SUCCESS in 10m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6001781f1e94c38b096dfbdb6eea0b7 : SUCCESS in 54m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6bd1ae7bcf5c42c8bd515e7d66234f06 : SUCCESS in 54m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e64b3607e7d43f88a52e7a347b79dbd : SUCCESS in 49m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b6f10afc67e444097e8fff201a8e156 : SUCCESS in 27m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/860f58ddf6804b1aa8c53bd0e52d53a6 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/283a7fdc5fdb44c3abb88243607e9872 : SUCCESS in 9m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba493f51addc4f08a5eaeca802918ef2 : SUCCESS in 53m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d1f60c49d8f497d8910075e5ccc717a : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2377bfd329436b15d924d75e0c08ca92f1d49f97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 00:07:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abf4840c04e641e9a1d632dc7a3381bb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bcf79cea141e4e99b50e91d7ad36635d : SUCCESS in 2h 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"72952092f2ccebca7bef222d7b18cef343b4a8d2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-02 22:45:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"89912f3f33538d23fbb3468900d7031e3fc6088b","revisions":{"89912f3f33538d23fbb3468900d7031e3fc6088b":{"kind":"REWORK","_number":1,"created":"2025-08-20 20:44:04.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/22/958122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/958122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/958122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/958122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/958122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/958122/1"}}},"commit":{"parents":[{"commit":"4d2426591e8fe47f0c7cee5d73cf4bc0004e33a0","subject":"CI: Stop using bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d2426591e8fe47f0c7cee5d73cf4bc0004e33a0"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-08-20 20:39:29.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-08-20 20:43:58.000000000","tz":-300},"subject":"tempauth: remove un-tested mis-feature","message":"tempauth: remove un-tested mis-feature\n\nThere\u0027s test_tempauth tests that use paths like `/auth/v1/act/auth` -\nbut as best I can tell they\u0027re all negative tests?\n\nI had no idea swob or tempauth could `path_info_pop`?\n\n```\n(Pdb) req.path_info\n\u0027/auth/v1/act/auth\u0027\n(Pdb) req.path_info_pop()\n\u0027auth\u0027\n(Pdb) req.path_info\n\u0027/v1/act/auth\u0027\n```\n\nThe docstring for get_token suggests these are supported on purpose:\n\n```\nGET \u003cauth-prefix\u003e/v1/\u003cact\u003e/auth\n    X-Auth-User: \u003cact\u003e:\u003cusr\u003e  or  X-Storage-User: \u003cusr\u003e\n    X-Auth-Key: \u003ckey\u003e         or  X-Storage-Pass: \u003ckey\u003e\nGET \u003cauth-prefix\u003e/auth\n    X-Auth-User: \u003cact\u003e:\u003cusr\u003e  or  X-Storage-User: \u003cact\u003e:\u003cusr\u003e\n    X-Auth-Key: \u003ckey\u003e         or  X-Storage-Pass: \u003ckey\u003e\nGET \u003cauth-prefix\u003e/v1.0\n    X-Auth-User: \u003cact\u003e:\u003cusr\u003e  or  X-Storage-User: \u003cact\u003e:\u003cusr\u003e\n    X-Auth-Key: \u003ckey\u003e         or  X-Storage-Pass: \u003ckey\u003e\n```\n\n... but I think it\u0027s a mis-feature, and the entire block and related\nnegative tests are actually some kind of vestigial cruft accidentally\ncarried over during some shuffling from swauth/testauth \u003d\u003e tempauth\n\n```\ncommit f68b6354e28f840d4a07f633888751c782267662\nAuthor: Greg Holt \u003cgholt@rackspace.com\u003e\nDate:   Thu May 26 02:24:12 2011 +0000\n\n    Renaming TestAuth to TempAuth because nose hates anything with the\nword test in it.\n\ncommit 3ee4a0110067ce256a27802c6917f10d87bc3ade\nAuthor: Greg Holt \u003cgholt@rackspace.com\u003e\nDate:   Thu May 26 02:17:42 2011 +0000\n\n    Remove swauth; update references from swauth to testauth.\n\ncommit b24b063552669e089d64463f29dd9ff8b129e903\nAuthor: Greg Holt \u003cgholt@rackspace.com\u003e\nDate:   Thu May 26 01:19:03 2011 +0000\n\n    Created testauth WSGI middleware to replace Swauth as development\nauth service.\n```\n\n... but it\u0027s hard for me go back and run code that doesn\u0027t work on py3.\n\nIf there is someone that\u0027s relying on some kind of \"tempauth multi\nprefix\" feature they can come back revert this change and add some tests\nthat make it supportable.\n\nChange-Id: Ib2e57c29abd61279e6436818abd3219c5a8aab7f\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89912f3f33538d23fbb3468900d7031e3fc6088b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89912f3f33538d23fbb3468900d7031e3fc6088b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
