)]}'
{"id":"openstack%2Fswift~958131","triplet_id":"openstack%2Fswift~master~I2791566b208327b1fb536fb56a363337ab3f3941","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-08-21 16:37:06.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-08-21 21:34:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2791566b208327b1fb536fb56a363337ab3f3941","subject":"checksums: Ignore broken pyeclib installs","status":"MERGED","created":"2025-08-21 00:35:03.000000000","updated":"2025-08-21 21:35:11.000000000","submitted":"2025-08-21 21:34:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958131","meta_rev_id":"e831fd1305d8dd087e004bdc817d24de95b1f22e","_number":958131,"virtual_id_number":958131,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-21 21:34:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-08-21 16:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2025-08-21 18:25:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2025-08-21 20:24:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-21 01:41:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-21 08:10:14.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-08-21 16:37:06.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-08-21 18:25:10.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"b9d6adb6b8b93791fb8627756dbcbced62d6b7dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-21 00:35:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0794fec478bb1aef6fee8de59228ae9d70934670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 01:41:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4cc02d8bd6a4509bde987531f53873f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e5a37f6624b4e1a979ae4cf58caf2f2 : SUCCESS in 7m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/c695748d3f3d4a5e8c70d730ba3e0c38 : SUCCESS in 54m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/48a13e0c37474c6ba50c7f34ad640cc4 : SUCCESS in 1h 01m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/192646bbf3ef471d9c23c9dc287eaf0d : SUCCESS in 55m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ca0023b141547a487bf124aed0c81cf : SUCCESS in 39m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1fa89dfea944d3eb75c1f338607a9c8 : SUCCESS in 10m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/959f6f213fe14998903f43903297a45f : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8944d917f29f43e2b70334977d7fcfb9 : SUCCESS in 14m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c81a242ce4cb41959e522cf252866459 : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ea172e084804430b8b9bbbce2010ea41 : SUCCESS in 29m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/40fdd5523d9d45588955199ddbf03e4b : FAILURE in 21m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/091f9d06a3e24e57a23806938932c9c6 : FAILURE in 30m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/16718862ab6b44009b7a82c772a92881 : FAILURE in 24m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0be0e155a6294ae3920b6bed7afd1628 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5b653d7e6c5448be85547475fb687b58 : SUCCESS in 10m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/26a14ffcabc64f6a9839f1da7de173cf : SUCCESS in 13m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6f03ae3dc9324d4cbabafa71d73d4e1e : SUCCESS in 7m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8944c3d9c54f4c95bc4dc11a9b1aebdb : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e87f814f6e64c4a804475438795e08f : SUCCESS in 10m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/11e961eb82b345b3bd82b3fae3677fad : SUCCESS in 40m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4185bae68bff49b1bf300d0591db0b7e : SUCCESS in 29m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6cf85cb1a6964b679f1f2ef79047a530 : SUCCESS in 57m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87ca794c42a2467c9fc9612ea9b17f7b : SUCCESS in 27m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/587a23c42e2b43feaf51430a78f857c2 : SUCCESS in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/265da49ce9d546bb80e69faa67241e0f : SUCCESS in 16m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78b39ed5636b420f942ac5f9ad6d13cc : SUCCESS in 52m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65cea5bf593d4e88b41c58dedaa187a8 : SUCCESS in 1h 01m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e1a32b9816715d448c36e5a5f49ff398675c0ed2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 02:16:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8611bbf8e0484a14a6b16a0c353f713d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7908d4cb27934b4eb43c6b8611236558 : FAILURE in 1h 39m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e9d5526a9c050687cab2c328a74513bf173c6669","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-21 08:10:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58a4b381eda58f15c9a2fe1a3a4bb75d37a91b72","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-21 09:58:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"616e0e6d4dc09b8c6c1ba2ee2bf9116cc1870d20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-21 15:42:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"89ad9ee7b70af64898c67da9f60d0ac548cbc59c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-21 15:43:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3813a2836eb14c9dd75e3cd148dcf5d588e7c43d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-21 16:37:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"24cf336176b024e19c13942c2e4a05344cc970d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 16:49:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dd141ae2a594f29bf1d58c9b2aec1c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/623aaf38b0994559bdc1fab0e583ba8b : SUCCESS in 10m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/792c6947083b4472aaf94897925a8b1b : SUCCESS in 1h 01m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d938c615d75a48c39e935e6b4193e3ab : SUCCESS in 51m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a0dbefd0d0a04f4880b04de55da22855 : SUCCESS in 49m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/761449c583ea4324b4faa901820ebf04 : SUCCESS in 51m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/735ebe73759e43728de08de30c46d405 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebe2e18f80314c979d8403bb1647929c : SUCCESS in 11m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fdcad05874940c6a474a64ebdd867c7 : SUCCESS in 10m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ca1d71fdb0443f0a6946616683f88d1 : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7ae5fff89f57453baa75c54d79379c1a : SUCCESS in 15m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c0e365712c904d6a8b970b87fb80414e : SUCCESS in 24m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1ccfc861d9cd4a6f9f5e314d91341a96 : SUCCESS in 20m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/218680d6afdd4a679396b00fa2e3f744 : SUCCESS in 20m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e349927224254827a58faa78cba90afa : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d8780204dfae47108eb96994eb69f114 : SUCCESS in 9m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fb0fbdbe8cd14834a69aea6a3fe456de : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5c22eb8c59d34d8b9658e675152bd504 : SUCCESS in 8m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/92f0ee92915847ee8f0d3231bd0b6a44 : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3fb302eb2cc643368b8ffd70e18a92db : SUCCESS in 11m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2dafbd41092149619cb0f87fdbda92aa : SUCCESS in 33m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a4761e19380e4deba83e27e35957c283 : SUCCESS in 54m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5604b624ed5a48ae9a7853ff8a424453 : SUCCESS in 22m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bfe9809b52747cba1b1d687a7a2256a : SUCCESS in 29m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4b09e7f99624440a794b9e85fb8ed8e : SUCCESS in 3m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2738f962662644a4844ab99000c258e4 : SUCCESS in 9m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71da5f7a081a4a1f8c2b349781fd3014 : SUCCESS in 57m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e06ee0ef2eaa47baa1d4fb64d7a3959c : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5e48ac2f4e70550fb6fe3b54087f40f1f09a1365","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-21 18:25:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7f1de7058e32cfeff06445fcbc377e72f3ce065","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 19:06:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ec1f2b99eb1485183fa4e73456700d3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80c99ee42f1a4b2bae744748659aa166 : SUCCESS in 1h 24m 00s","accounts_in_message":[],"_revision_number":2},{"id":"2c52dcae0ad6455265b0bb333d2e798732857b36","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-21 20:24:49.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"08268c3a59e0faaf592ecfedc7cbc1df644dd4dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 20:25:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4d43e34fe64867fab678b01638d90992af6b299d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 21:34:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d72103714274e089f43e5fbc14e8099\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3418d554c534772b151570d21f5cf94 : SUCCESS in 6m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/55672e21e5b14dc89b8fa55e5c661c1a : SUCCESS in 46m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/97164238a23f4affb8956a65c745f0bd : SUCCESS in 1h 04m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/93e9b1740b7e486c83ce2aac559b41f2 : SUCCESS in 36m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21a1c6d0174b453c884f422ec244f0ec : SUCCESS in 59m 04s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/733e6218854544548c434ff4c7892c7b : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/928ef3ea46724e0884828ac7f4dba399 : SUCCESS in 24m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b53778de6b1c4be2913c72b8ab8d969c : SUCCESS in 15m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e4ffc76bbae6458f859c915b8d5c94f2 : SUCCESS in 10m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ca148e9c5e6b4836973beb325f205f5b : SUCCESS in 13m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/19c30df80d384e32bf6c99217076cdc6 : SUCCESS in 14m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7ce919dfa71c4d269616465c01628c1c : SUCCESS in 14m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7bd8c72e30b34056b0ce79136fed4e7e : SUCCESS in 15m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e7f9c3f1f93e4435b855050f84acd19a : SUCCESS in 9m 53s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dd0ddf54276348e981701b7e082de792 : SUCCESS in 6m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/09319097f7714f58bb0decb8ece05c36 : SUCCESS in 58m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e3d3501cb61f4fe983e6f57e6b132c3c : SUCCESS in 35m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5379efe965a485d93b1fe8f98cae26f : SUCCESS in 50m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2db5a5c0e8e4c168e0668aad8a27482 : SUCCESS in 21m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3c1e2d60be84c2cb24cd1a0d8156b1e : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c236a96d288b4f99b311837270232626 : SUCCESS in 27m 53s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bd7ee5042ef40c69e2102a0666d925c : SUCCESS in 29m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc6b77b06399466b9f116b936f8c72d8 : SUCCESS in 43m 20s","accounts_in_message":[],"_revision_number":2},{"id":"9b6e6bf6005b922bbee38878e2f31523bd02f37c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 21:34:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e831fd1305d8dd087e004bdc817d24de95b1f22e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 21:35:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7444b20074d48f88e6083f1f72e4bb9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1412beda35b84d19885374b2a2fe3a97 : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/652dce3c85c349a9b5b30835eb381748 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81df05a9c4702204a8347ab2ba43e7ff75c328aa","revisions":{"c509ef13881fb3c24b5c1f2f9a8097b86612906a":{"kind":"REWORK","_number":1,"created":"2025-08-21 00:35:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/958131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/958131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/958131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/958131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/958131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/958131/1"}}},"commit":{"parents":[{"commit":"dfab57be74cfe20523dccda98dceb2a8fc507991","subject":"Merge \"Revert \"Remove unused MANIFEST.in\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfab57be74cfe20523dccda98dceb2a8fc507991"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-21 00:26:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-21 00:34:51.000000000","tz":-420},"subject":"checksums: Ignore broken pyeclib installs","message":"checksums: Ignore broken pyeclib installs\n\nThe RuntimeError gives more of a hint than the TypeError, but we haven\u0027t\nreally solved the issue. If there\u0027s a busted pyeclib install, it\u0027s safe\nto assume it won\u0027t have ISA-L alongside it; just proceed assuming it\u0027s\nnot available.\n\nCloses-Bug: #2120591\nRelated-Change: I64a85eb739fb72efe41f1ee829e463167246b793\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2791566b208327b1fb536fb56a363337ab3f3941\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c509ef13881fb3c24b5c1f2f9a8097b86612906a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c509ef13881fb3c24b5c1f2f9a8097b86612906a"}]},"branch":"refs/heads/master"},"81df05a9c4702204a8347ab2ba43e7ff75c328aa":{"kind":"REWORK","_number":2,"created":"2025-08-21 15:42:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/958131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/958131/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/958131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/958131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/958131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/958131/2"}}},"commit":{"parents":[{"commit":"dfab57be74cfe20523dccda98dceb2a8fc507991","subject":"Merge \"Revert \"Remove unused MANIFEST.in\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfab57be74cfe20523dccda98dceb2a8fc507991"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-21 00:26:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-21 15:42:05.000000000","tz":-420},"subject":"checksums: Ignore broken pyeclib installs","message":"checksums: Ignore broken pyeclib installs\n\nThe RuntimeError gives more of a hint than the TypeError, but we haven\u0027t\nreally solved the issue. If there\u0027s a busted pyeclib install, it\u0027s safe\nto assume it won\u0027t have ISA-L alongside it; just proceed assuming it\u0027s\nnot available.\n\nCloses-Bug: #2120591\nRelated-Change: I64a85eb739fb72efe41f1ee829e463167246b793\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2791566b208327b1fb536fb56a363337ab3f3941\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81df05a9c4702204a8347ab2ba43e7ff75c328aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81df05a9c4702204a8347ab2ba43e7ff75c328aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
