)]}'
{"id":"openstack%2Fswift~958133","triplet_id":"openstack%2Fswift~master~I4d04415601f43b87f528993ffadcdd29cb401a24","project":"openstack/swift","branch":"master","topic":"db-replicator-log","attention_set":{},"removed_from_attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-10-02 20:05:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4d04415601f43b87f528993ffadcdd29cb401a24","subject":"container-replicator: include container and db paths in logs","status":"ABANDONED","created":"2025-08-21 02:39:32.000000000","updated":"2025-10-02 20:05:30.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"afa87c63fb3dff5c2ba8f3996154be38923101c2","_number":958133,"virtual_id_number":958133,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-23 16:55:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-21 03:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 15:49:18.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"60ecd57f0702cd1e0e465ee2676e45f284a471ac","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-21 02:39:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccbcd0cdc9959b231e42231657c1574024275b2b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-21 02:39:42.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"691138e36c7b536f0f00ee1d25fab374d0b8302e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 03:55:39.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a57d2e6524424aa0bbde50fde8bcae24\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5adc5e01020246549191b7d7297bd5b8 : SUCCESS in 11m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/d7976f29c8284d899d05130ddf1ba8d2 : SUCCESS in 1h 01m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de93131448654c1fbbd11636fb60c355 : SUCCESS in 1h 11m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a7c53deb7b84af78a645bc143a968ba : SUCCESS in 46m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26f87c22486d4f57970698872d95602b : SUCCESS in 1h 02m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b2f5563223e4d59a81799b90050ddc2 : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/feec0781ccc5400585c6d382a247fcc4 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4b2217aa8944e058c09e44e8ce91228 : SUCCESS in 23m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/428a1a1c9f954d2e96de6804e0753dbf : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b14a05f56f844fc496fb4f3d3ddf153a : SUCCESS in 22m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fa5f678b61a642f9b67ae191fe603440 : SUCCESS in 15m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9fc762147ebc488a9ffbc852cc9182e6 : SUCCESS in 23m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a809cda13a584db2a91a16a81443e4f5 : SUCCESS in 28m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f28877dcf9b544c5b1d28f07ec8821ae : SUCCESS in 9m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2237adbbcc664b6f9ee1fdcc2cd9f10f : SUCCESS in 16m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eac06941ad0d491cb36799629a17416d : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/058754f62f0142eca6045e684795dcf1 : SUCCESS in 6m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/483ee977238f43bc8eb59a3f070f0964 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6b7e1d81a95d4f5e8c588a4afefe5f71 : SUCCESS in 11m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a749dee13cc492887b521f90ccc1811 : SUCCESS in 50m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8b8b4bb19759445e9c6554167c7c9dd0 : SUCCESS in 51m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6d4022d53c6c480f93dbadd0f71ab5cd : SUCCESS in 41m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85cf3efd978a485c97b36ded1169074b : SUCCESS in 25m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/706f23ae9c6b4accb1d9057b05e3b216 : FAILURE in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1089df83069d426581112efab77df1cd : SUCCESS in 27m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee2f679d1d6d43969eda11c968e92aaf : SUCCESS in 1h 05m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e25c214045345bf862ee5b2f2180e66 : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ed84670ed8b8afff21ed7998e0adae5e30bc532a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 04:29:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aec272291a944869a9d067643608425\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe31195f0b8c40f090d16144373eb6ae : SUCCESS in 1h 47m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d17b00c49ea55a86628c4f5c7b530b9d9e563141","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-02 18:49:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0088f342ceebc54b00f70819d28166c97e7f0d34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 19:56:14.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ecf32b7e4b74b569f2d3cdad3b13e0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67016cc676ae4b40b88d78bb6e0bc273 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/db1dfc5abc434f64b97cb2b8e2eefde5 : SUCCESS in 56m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/138150ccf0084293b5abbaffaef253c9 : SUCCESS in 59m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d673ed64daba42528012f37b44117888 : SUCCESS in 34m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2eaf0f86cd4e474ba09542d6557f68ca : SUCCESS in 56m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88108042f7594da6be8cd79f3a07bea5 : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1e583810c234538801570d9bb7f89ab : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1fa04ddef36b422abee6a79ab654addd : SUCCESS in 10m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/287446b808c4491a854483d5b609ab19 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/15968f0be24d4871adc0bf669a50df28 : SUCCESS in 26m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/12a6c1f8e9454251a2a70aa729fd7312 : SUCCESS in 20m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eed91e5df5c3443abba2462a957b1e2c : SUCCESS in 20m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7cbbc0f84eee408c916e1df1603091de : SUCCESS in 30m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fdcc9a6d01c24328a676a28b39ddf95f : SUCCESS in 13m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/069261ae44a9452a93fb3f0c1203e560 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4d0b8f01db114ae9acff217943fea899 : SUCCESS in 16m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9ac539eb0ba400fa91cb29d1374389a : SUCCESS in 8m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1106e5dbe50c419ebe476954dabab954 : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/68f0d810325541a9acc2c350bf2219f8 : SUCCESS in 8m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f73be0a54c7c4025b335a75a819ca615 : SUCCESS in 49m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/498ef878b7b44801a3dccded96a08474 : SUCCESS in 55m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4435d2844e284b18984543d6f839b7af : SUCCESS in 53m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca75d6142d184467b332290acdcfc0d9 : SUCCESS in 21m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dbe5ce8ba06418eac78d3533616f592 : FAILURE in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/44800f50658846ee92c1027324882441 : SUCCESS in 17m 46s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ed6aaa5bb9c4ac6b8aeeab78601e9f7 : SUCCESS in 42m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00fb4dd818e3498492fc127829da1f9d : SUCCESS in 56m 15s","accounts_in_message":[],"_revision_number":2},{"id":"6702cce57390c7312aafaeba8a54f1310206c49b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 20:23:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bbfcc11a22045b38a5ace30c6e529fd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/50ae4845504646dda6c8d816c5af4751 : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4ba7930b1e30502585b4b6fb720352a783abe0d6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-05 17:38:13.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/933379","accounts_in_message":[],"_revision_number":2},{"id":"e2ea6aac3e7295f9a7be99a1a19b9811c2691723","tag":"autogenerated:gerrit:restore","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-20 14:36:08.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"4548ef57d98e54aee264e573dccb4b65157c7a86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-23 15:47:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"47ebbdf937460c038c136048cf196f453cec3b68","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-23 16:02:56.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"0f082ab956f2565870b95edee55c93f2c91fbe1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 16:55:14.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16bd110f0514462f8ed67c8ba20bbd20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6352e3d650ac454cba50237e46147bd8 : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/051c1119d44b439abc6c8bc0eff80e33 : SUCCESS in 53m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2e808e8bd83c4bdd9a3fc770c96deae8 : SUCCESS in 56m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ebac10b4b7e4df0859a83ca25562f9c : SUCCESS in 55m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c3c1868cc27469689578e0e2d446833 : SUCCESS in 1h 05m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42d939d23d814be091263617ca00e9ad : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d86962555e2457e8ade6ae3343c2b47 : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/118f02aa89274242b737e5f209f3f495 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/765f29fcfcf54acfa594476167c5bf99 : SUCCESS in 7m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a8af18e108e54571b0755678a00b27a6 : SUCCESS in 22m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/38b46fbb423045efa467c67c757b1a22 : SUCCESS in 20m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4418edfa24344e509649806f9beeaff4 : SUCCESS in 21m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e51f485387754c8886f180f08edf7347 : SUCCESS in 13m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/41975a9553a847f691a8a427f02c990e : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e8c2fdf8ddc448a99346a13e8141df32 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e07d2b9d1caa40f38dca484bf8d788fb : SUCCESS in 15m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ac969aad84dd42b2bcfacd051058cf57 : SUCCESS in 8m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/568035a3e9014c25be5d4a92725afef8 : SUCCESS in 15m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/aae750fe55164b8d94ecd344e461d7e0 : SUCCESS in 8m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c5f167a3dc24385b3a7316db4e11e6f : SUCCESS in 55m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ec54a1df8a8b43a4b1bed39d691bc49e : SUCCESS in 52m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/539f4e9f9ef345f2a79b7a71dbf783a5 : SUCCESS in 54m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97db1a0f013545238aba3c13b96728dd : SUCCESS in 27m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bae82addfac1450cafbc6201538a2eb8 : FAILURE in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8d735d002e704826a2992122dceb9212 : SUCCESS in 19m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66da9881de3e41e38a5c2478a39cd14a : SUCCESS in 55m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6109e78d13d64fd1afd8f5ba94aa72fb : SUCCESS in 55m 36s","accounts_in_message":[],"_revision_number":3},{"id":"9f93776d80368070b3e319eed0aef77a5ad63465","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 17:01:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f8c532a1d26414fa442841f2c831e2e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/af0bdaac48b54f31aff3a5deac0fd74a : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":3},{"id":"587b3e39c0d2826884efd9629d7929d5f5a9ca70","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-29 15:49:18.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"afa87c63fb3dff5c2ba8f3996154be38923101c2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-02 20:05:30.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/933379","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6310b926409092e5cd1b18c4f0c8b3ed43fb8edc","revisions":{"ed956678d092f33e9ee2ae9b34e98e7febb9034e":{"kind":"REWORK","_number":1,"created":"2025-08-21 02:39:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/33/958133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/958133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/958133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/958133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/958133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/958133/1"}}},"commit":{"parents":[{"commit":"941e41427c390d407c85fe5f033dcc42aa367527","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/941e41427c390d407c85fe5f033dcc42aa367527"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-21 02:36:36.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-21 02:38:50.000000000","tz":-420},"subject":"container-replicator: include container and db paths in logs","message":"container-replicator: include container and db paths in logs\n\nChange-Id: I4d04415601f43b87f528993ffadcdd29cb401a24\nRelated-Change: I289d3e9b6fe14159925786732ad748acd0459812\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed956678d092f33e9ee2ae9b34e98e7febb9034e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed956678d092f33e9ee2ae9b34e98e7febb9034e"}]},"branch":"refs/heads/master"},"84f49a99f034df84985441b3e64aeb893188936e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-02 18:49:25.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/33/958133/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/958133/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/958133/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/958133/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/958133/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/958133/2"}}},"commit":{"parents":[{"commit":"27db680ae44290d8715f127777def65a76012fd2","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27db680ae44290d8715f127777def65a76012fd2"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-21 02:36:36.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-02 18:49:25.000000000","tz":0},"subject":"container-replicator: include container and db paths in logs","message":"container-replicator: include container and db paths in logs\n\nChange-Id: I4d04415601f43b87f528993ffadcdd29cb401a24\nRelated-Change: I289d3e9b6fe14159925786732ad748acd0459812\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84f49a99f034df84985441b3e64aeb893188936e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84f49a99f034df84985441b3e64aeb893188936e"}]},"branch":"refs/heads/master","description":"Rebase"},"6310b926409092e5cd1b18c4f0c8b3ed43fb8edc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-23 15:47:43.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/33/958133/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/958133/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/958133/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/958133/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/958133/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/958133/3"}}},"commit":{"parents":[{"commit":"4be903c5763f71af495fd53e858e6b7b1c99557b","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4be903c5763f71af495fd53e858e6b7b1c99557b"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-08-21 02:36:36.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-23 15:47:43.000000000","tz":0},"subject":"container-replicator: include container and db paths in logs","message":"container-replicator: include container and db paths in logs\n\nChange-Id: I4d04415601f43b87f528993ffadcdd29cb401a24\nRelated-Change: I289d3e9b6fe14159925786732ad748acd0459812\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6310b926409092e5cd1b18c4f0c8b3ed43fb8edc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6310b926409092e5cd1b18c4f0c8b3ed43fb8edc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
