)]}'
{"id":"openstack%2Fswift~958162","triplet_id":"openstack%2Fswift~master~Ic05c148d5dc1c9716e769a5b265481d48eb3bbe8","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ic05c148d5dc1c9716e769a5b265481d48eb3bbe8","subject":"sq? checkum: fix find_isal tests","status":"ABANDONED","created":"2025-08-21 09:57:12.000000000","updated":"2025-08-21 15:42:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c84a88e3cb7ebcdb602547e16b6011ec11fcb63b","_number":958162,"virtual_id_number":958162,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-21 10:43:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-21 10:43:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"858f3aea856b919345cdde5c988f09a8cd2edbc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-21 09:57:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d4b6d82408b0907d3d2d5e222ec23f86791982b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 10:43:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35e43b6c8e354d17b298ddca5f5e164f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203480b86d994c429a08ea244c823fa5 : SUCCESS in 10m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be3861d6e3af48d1b123e27250ceb0f7 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28cda1d8ac014283960c08144a6a308e : SUCCESS in 10m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db34f001a64e43f9bf1f424ff2aecdf5 : SUCCESS in 26m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f6d6893756b2485d9dca44eca52d867a : SUCCESS in 3m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a30a8a11760a476780c122b4fb7998bb : SUCCESS in 20m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/32c75e16913c48689125b0d69a03bd86 : SUCCESS in 20m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3926ed3dabc8408680efd25f0faee8db : SUCCESS in 33m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e80afeb623284f1982387a982cf17468 : SUCCESS in 21m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0ac174aa955c441f82e2d91288275ae7 : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d935607a8a114db9922726b123bd9293 : SUCCESS in 8m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e137d4ed5d3c4275a84ebcfa64d15930 : SUCCESS in 15m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/abe8ad6026eb4d0ba064854db49360ea : SUCCESS in 21m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f528c5bd1ce482388424a3b172ea163 : SUCCESS in 44m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/466198c8131c4500a42172425c6f614d : SUCCESS in 24m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d153792525f448359cce57c079a334dd : SUCCESS in 3m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/71eb81ff286c4920ac779683aebee82e : SUCCESS in 11m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b107789276858552dd5a1c403e989946b41374f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 12:47:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92237a9278514564a5abd3b321899ab0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/838f104936bd4e748c89efa039cd129e : SUCCESS in 2h 19m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c84a88e3cb7ebcdb602547e16b6011ec11fcb63b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-21 15:42:31.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e44bd0a2378d49b3dfbfa1865424dc7d897ea0ae","revisions":{"e44bd0a2378d49b3dfbfa1865424dc7d897ea0ae":{"kind":"REWORK","_number":1,"created":"2025-08-21 09:57:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/958162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/958162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/958162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/958162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/958162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/958162/1"}}},"commit":{"parents":[{"commit":"c509ef13881fb3c24b5c1f2f9a8097b86612906a","subject":"checksums: Ignore broken pyeclib installs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c509ef13881fb3c24b5c1f2f9a8097b86612906a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-21 09:56:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-21 09:56:41.000000000","tz":60},"subject":"sq? checkum: fix find_isal tests","message":"sq? checkum: fix find_isal tests\n\nChange-Id: Ic05c148d5dc1c9716e769a5b265481d48eb3bbe8\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e44bd0a2378d49b3dfbfa1865424dc7d897ea0ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e44bd0a2378d49b3dfbfa1865424dc7d897ea0ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
