)]}'
{"id":"openstack%2Fswift~958238","triplet_id":"openstack%2Fswift~master~I86018a29ecbdfa33246516de1289187ccaa52e22","project":"openstack/swift","branch":"master","attention_set":{"38213":{"account":{"_account_id":38213,"name":"Vitaly Bordyug","display_name":"Vitaly","email":"vbordug@gmail.com","username":"vitb"},"last_update":"2025-08-21 21:15:29.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I86018a29ecbdfa33246516de1289187ccaa52e22","subject":"[proxy_logging] Add acl_required similar to aws aclRequired","status":"NEW","created":"2025-08-21 20:08:33.000000000","updated":"2025-08-21 21:15:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dfea76de969ba715c80901d5c1da151ce4903638","_number":958238,"virtual_id_number":958238,"owner":{"_account_id":38213,"name":"Vitaly Bordyug","display_name":"Vitaly","email":"vbordug@gmail.com","username":"vitb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-21 21:15:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-21 21:03:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-21 21:15:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46a35d3c93843f642281a7a9b75fa174ff3454a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38213,"name":"Vitaly Bordyug","display_name":"Vitaly","email":"vbordug@gmail.com","username":"vitb"},"date":"2025-08-21 20:08:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8431c796322323a1e0b79ba460a7ab819111031f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 21:03:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b5135728a7a4ffc95efb50ea0362f79\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/369a5bcd9ccb4a2bb373cd6667e64fd3 : NODE_FAILURE Node(set) request 43a6c39dff0e489db1ab3d66e89b400a failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"dfea76de969ba715c80901d5c1da151ce4903638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 21:15:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d1ad8d2d03a4835ace04cd0048b4944\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4960eb674d042a7bedf77afff3d40ad : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/bae971945b784744b3abf252b2169987 : SUCCESS in 1h 02m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00fc3a0e497549d0837cc75e7360d5f0 : SUCCESS in 55m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/647686af65774fa38fa67605352596aa : SUCCESS in 54m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/baa29a6bf9cb447aafbfe8485586bc53 : SUCCESS in 1h 05m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0e6896be5e845d4b056ca84d85ad26b : SUCCESS in 12m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35b25717db0c458b84ddabf81e015571 : SUCCESS in 22m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71f12ec128434c5e9b9db657b3d265d8 : SUCCESS in 19m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a2a9289ae5eb47dfbcf0b24c88e07a4b : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/734b0ba320784825b8b5ab5e9c56320a : FAILURE in 30m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b63bf8b4440140c4968ab45ccc5f630e : FAILURE in 22m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1557ec034f864554b1c2425d7644e230 : FAILURE in 28m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a7ae559f34b4421ba10203189ffd7ad9 : FAILURE in 31m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c8d64e1575954af9b45f7cfe07d3e183 : SUCCESS in 11m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ab8294eaf9c94c46b8ee7d2a60d34c39 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/462f74b99d704589a43cb64a6a15d865 : SUCCESS in 13m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7bd1617d6a6d496ebe4667e9b04da6ab : SUCCESS in 6m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ac72c784cbc446678b6b29624cbebab8 : SUCCESS in 15m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4b11d121239a431ca3d235269618c16a : SUCCESS in 6m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0617f9d184742cc9485ea31ff210825 : SUCCESS in 54m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/78a820df96bb49a4bf956a2f515bdea0 : SUCCESS in 26m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1e4811564e546f18f7a52a2bd73e96e : SUCCESS in 51m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2de221a479c3468aa2a641dd2d49526f : FAILURE in 29m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd276c6de214519a42487a2dc900c9c : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0d163df757d437bb71e5f092c1f8510 : SUCCESS in 20m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66c30f2ca65e46a3817e6e4a02fd9e09 : SUCCESS in 51m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae70f4e6c5ac4ea0bb71a0b756cb97ae : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6ea53a6f27d46515df91f8f25ba76b08aac95a6","revisions":{"e6ea53a6f27d46515df91f8f25ba76b08aac95a6":{"kind":"REWORK","_number":1,"created":"2025-08-21 20:08:33.000000000","uploader":{"_account_id":38213,"name":"Vitaly Bordyug","display_name":"Vitaly","email":"vbordug@gmail.com","username":"vitb"},"ref":"refs/changes/38/958238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/958238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/958238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/958238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/958238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/958238/1"}}},"commit":{"parents":[{"commit":"0c68af898666c71fa96ea36b0c9ef886a791808e","subject":"proxy-logging: create field for access_user_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c68af898666c71fa96ea36b0c9ef886a791808e"}]}],"author":{"name":"Vitaly Bordyug","email":"vbordug@gmail.com","date":"2025-08-21 19:29:16.000000000","tz":120},"committer":{"name":"Vitaly Bordyug","email":"vbordug@gmail.com","date":"2025-08-21 19:32:58.000000000","tz":120},"subject":"[proxy_logging] Add acl_required similar to aws aclRequired","message":"[proxy_logging] Add acl_required similar to aws aclRequired\n\n  Implements an acl_required field in Swift access logging similar to\n  Amazon S3\u0027s aclRequired field. The field is set to \"Yes\" when ACL\n  validation is performed (account, container, cross-tenant, or\n  role-based ACLs) and \"No\" when no ACL validation is required.\n\n  Changes:\n  - Add acl_required to proxy logging template and documentation\n  - Update tempauth, keystoneauth middleware to set the field\n  - Add unit tests for affected middleware components\n\nSigned-off-by: Vitaly Bordyug \u003cvbordug@gmail.com\u003e\nChange-Id: I86018a29ecbdfa33246516de1289187ccaa52e22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6ea53a6f27d46515df91f8f25ba76b08aac95a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6ea53a6f27d46515df91f8f25ba76b08aac95a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
