)]}'
{"id":"openstack%2Fswift~958251","triplet_id":"openstack%2Fswift~master~I4c96a76b6124544023c6281b48f21681dd8e3310","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I4c96a76b6124544023c6281b48f21681dd8e3310","subject":"sq? tests and docs","status":"NEW","created":"2025-08-21 22:56:58.000000000","updated":"2025-12-18 21:05:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d13e7ea5cd8a547b633b069d28f447c93cfa04fa","_number":958251,"virtual_id_number":958251,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-22 00:05:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-21 22:58:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-22 00:05:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-18 21:05:06.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"d82c241bbcb29b54a7b04744cbf303683d314948","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-21 22:56:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c287eab039f659e87cd782dcb99cb8a58f1da9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 22:58:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe1bf3bd7e2f41d7b99a38a0b9721dce\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/382e5f56bb9b4c43ad77185d8472c1a0 : NODE_FAILURE Node(set) request aa8762a837fa4014bfc1803c91dcb16f failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"8be398795700977e65eab46e25a1f33423bd5bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 00:05:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5073969a5dd45029e0ff7857da2222d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a613c7d84caa42d7aceb752d6ba22ee1 : SUCCESS in 6m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/ebefc4787dfc4189b50ccdeb05494573 : SUCCESS in 58m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6ca49ceed674745919b933125e8a9f1 : SUCCESS in 1h 05m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e82d90998e648e39f542a96c08fda35 : SUCCESS in 49m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dc7a2a5db884e07910d13f19b2fe02e : SUCCESS in 44m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afeea23ea10e443cb70647fe87941c95 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a730386f2794cf09e433125b50cb84b : SUCCESS in 16m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3a33646a3bb4858a26c0417fd809f9b : SUCCESS in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ea308295a2b74d0aac9415f2c15c4f8a : SUCCESS in 7m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d872e0161c774f83ad64f514821be7a6 : SUCCESS in 28m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6e0df6ff4be14984b0dea42e8ea01376 : SUCCESS in 21m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/28e051c4444c4231a6f7aeacff94b7ed : SUCCESS in 28m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8a4fd6cb919a418b9df8457b9480c9d6 : SUCCESS in 29m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/41bd9a1e77ce43c7a30eada1c2dd60c6 : SUCCESS in 11m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1aea747692f74e4384ac269ea16df965 : SUCCESS in 9m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/738e642f5a034bdf91d624ed242b49c4 : SUCCESS in 13m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee6fe02f4c9247aabcb431e157622e41 : SUCCESS in 4m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/408e24bc81964e8ea29cabbbe887eb29 : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d923e1fe5f654481a9620300ebd3fc21 : SUCCESS in 6m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a68ffb84af3040c188e19c89c71f017d : SUCCESS in 1h 02m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/63c13f15ab98408e9542126c771e9937 : SUCCESS in 51m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/833af46086bc46a985c3b6b539216a2d : SUCCESS in 55m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce711ca8361245fb8490d17860c64b00 : SUCCESS in 22m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3288174c90c648dc9dd6a56f470aa959 : SUCCESS in 5m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d36f8ca81daa4265ace6d8908e3394d9 : SUCCESS in 17m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/637aaa570d0e4e4cad65cafdd8d6f9b3 : SUCCESS in 51m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90aa370a7d024d368d4c6571cc77a4b8 : SUCCESS in 41m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d1dc0f0b904076fe0f6b98b72abd26715bd6942","revisions":{"3d1dc0f0b904076fe0f6b98b72abd26715bd6942":{"kind":"REWORK","_number":1,"created":"2025-08-21 22:56:58.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/51/958251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/958251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/958251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/958251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/958251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/958251/1"}}},"commit":{"parents":[{"commit":"0c68af898666c71fa96ea36b0c9ef886a791808e","subject":"proxy-logging: create field for access_user_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c68af898666c71fa96ea36b0c9ef886a791808e"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-08-21 22:56:08.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-08-21 22:56:45.000000000","tz":-300},"subject":"sq? tests and docs","message":"sq? tests and docs\n\nChange-Id: I4c96a76b6124544023c6281b48f21681dd8e3310\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d1dc0f0b904076fe0f6b98b72abd26715bd6942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d1dc0f0b904076fe0f6b98b72abd26715bd6942"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
