)]}'
{"id":"openstack%2Fswift~958291","triplet_id":"openstack%2Fswift~master~Ifab08cfe72f12d80e2196ad9b9b7876ace5825b4","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-10-08 15:56:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifab08cfe72f12d80e2196ad9b9b7876ace5825b4","subject":"s3api: fix test_service with pre-existing buckets","status":"MERGED","created":"2025-08-22 15:45:06.000000000","updated":"2025-10-08 15:57:43.000000000","submitted":"2025-10-08 15:56:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958291","meta_rev_id":"10a08ba0befd0b619d4c31ee6dfbfa7c4542ad95","_number":958291,"virtual_id_number":958291,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-08 15:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-10-08 14:47:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-10-08 14:47:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-22 16:51:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 16:15:31.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2025-10-08 14:47:52.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"15592c5071e4d88cf1dd0293f25aba4305a958a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-22 15:45:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9c6ba08a5ffc821947d4116400084a5281bb6b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 16:51:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be8b231078204c939f5b6e003361525b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/119a4049abf141ddad94e4fd15ce5028 : SUCCESS in 6m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3bcf1b6995d4c9ab4da6df06c65139b : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a27ab86bb5554899a041a2ebde62dda7 : SUCCESS in 10m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51dedc2b65fa47fd9e8d454cf0bfbdf2 : SUCCESS in 23m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0178ca5f6d2a4f368d2f940366a9943a : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/847c71c1f70246e09220fb6cb96dc5a4 : SUCCESS in 31m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/800dfb51b4c345208214b1d8214c362a : SUCCESS in 22m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/887c44bdad9c484abfd7bea26a9cd137 : SUCCESS in 31m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/36f7c7e1312e4781b766ce5d9e122880 : SUCCESS in 44m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/28f78d4adbba416c8a73108488b35647 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d0ff948620844dfaa30f0e4f76d95ed5 : SUCCESS in 14m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/13a97655cf9f463d9a52fa8a04d649e7 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b6d66a5697f44162ae18b16a0d5ffce4 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/525c9bd54cfb441fbe971b9cf2a587c1 : SUCCESS in 4m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c88904afbb1e403589dbe7ab24ca957d : SUCCESS in 37m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/560c3ff0374d4991b8492da717c5cc8d : FAILURE in 33m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fab9967bb5224be8a0f1198d5831703b : FAILURE in 14m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de2ecd45f9dc4f51a1ba36705cfff145 : SUCCESS in 22m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838ec26cf88649eebb69298d7a98d457 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/be60529768ef4070862fd93738231773 : SUCCESS in 18m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ef3106d3fe20f47db7d32012e97832e773fa85e8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-22 17:47:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8bd3822a55ce0a7fa69fbe8f0967c7458905bfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 19:08:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7466c98c8eb14456b1c4b4bc792f1176\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3cf4e677aea24c6a90cd9458d612bd43 : SUCCESS in 1h 27m 44s","accounts_in_message":[],"_revision_number":1},{"id":"518a8d8907fb4a33a84ba6b11c3191b45f8df146","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-25 14:54:29.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3b701d596b43b2011a3c3dd27e2c014bc7de476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-03 17:01:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4957d70d545098ace384dd0a9b5d738758bb0c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 18:11:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3184ca603a24da8911d2697f1550cab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a8d11948de4ec8ba2e7ef17481f5b3 : SUCCESS in 11m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/3f1522bde32d473b8517e66f40624a29 : SUCCESS in 28m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c4f476e386df4a1ebd58844395fe7585 : SUCCESS in 1h 05m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b2c1713cc32a4879a7125289f5be5002 : SUCCESS in 34m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbb7a45bfa9f4b3081737af66357c616 : SUCCESS in 32m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df25529c8db04044a02561b7bf986c62 : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8e690547c49472bb34993a48a45c9b6 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/807b91dab2344735bd12f9b959eb5267 : SUCCESS in 21m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cceae03c790044519ae0f5ef5cecd255 : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6f96c5e4bb6a400ebbb3eb50557ce939 : SUCCESS in 11m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7000ac0614ee4f3593e26cd3ec788669 : SUCCESS in 18m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c2461d54e6344632a914018fe05758cf : SUCCESS in 19m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/92485efb017942308c5aa722dc94003f : SUCCESS in 29m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/72724174e97548808795291609918d06 : SUCCESS in 13m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cb12e81077624f2e9642f3b826cd0501 : SUCCESS in 14m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6ad6e2bba04a4f9cbcfe567edd472138 : SUCCESS in 15m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/762295eb73c040fabc3bcd038e5cf40f : SUCCESS in 6m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0e245b337ca64cf49f7739dfe19f2300 : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/41d0ed29347a44d8afa78343f64985d9 : SUCCESS in 7m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ce7ea8d54b64d28863766ba1636c0ef : SUCCESS in 34m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7da8587550eb473fa7f3d15864094e81 : SUCCESS in 48m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0167eac59e7c42b2a907ae9173e4951e : SUCCESS in 39m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec24e0874ef34036bbc2124a5c8f4ba2 : SUCCESS in 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a503640a1fb4b6bba7efe066cc4462d : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf6ed153927a493180afa7cd2e9f4f7d : SUCCESS in 25m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e1d5b39f440404a8ed30553774dc12f : SUCCESS in 48m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/500d522e622a4d168f08b2c39b1ac4d4 : SUCCESS in 47m 39s","accounts_in_message":[],"_revision_number":2},{"id":"486c9d63a6401cb2e2acc867908e69fe72e3268d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 21:05:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9105b18b093d4d90b5fc2f1f88ef2140\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d68f864dc57e475aaecfc5f2ff3cbde9 : SUCCESS in 36m 33s","accounts_in_message":[],"_revision_number":2},{"id":"473c56fca3f1727d3de22d80216545287245bcdb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-08 14:47:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abeff96d973e58f572dff1b02ecbd469abb51069","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 14:48:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"27dabb721bf0fb91ab693da538f7582ebc0afa72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 15:56:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ac19503ea2f476faf6bf627e9ca2ed9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b934fbb92a4117a3e0b0e96e4c6874 : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/1f9e64730ffa449b8e220ddccc4b7f73 : SUCCESS in 1h 00m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9828cb5254484531be7c6c029c7b13b3 : SUCCESS in 1h 04m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0af368dbd5824aa3b9583801873db4ed : SUCCESS in 42m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/763837fc24a04c1faeb1c89116ff02b2 : SUCCESS in 1h 02m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0ae671b69e6f4d50bfadffaffae8bc88 : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e97dd18c9030418cb3c069882b45bc15 : SUCCESS in 25m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/301457341b934122a12379d0c1eab40b : SUCCESS in 22m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2ec67459dd5d484382177a777ebed65b : SUCCESS in 26m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6fed2060da8d4e03adb9ec6d6414415c : SUCCESS in 23m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8ca5075d60364f5aac67342e7dd944c4 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/72440f315fe94670904834c1b70d85c2 : SUCCESS in 14m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f5b0cf3d82954e2cbd286ab019cf9732 : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b4e61e691ff14d648dd7e16a3fe24521 : SUCCESS in 8m 37s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fa318b90de924b098b36ffbe0e04bf66 : SUCCESS in 7m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/23424566f4c446a783c214cd163c5855 : SUCCESS in 58m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b1609070e8ce41a9b85a94b1042ac3fa : SUCCESS in 52m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a17a58a156745c6b1657dc2c1546a35 : SUCCESS in 51m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/762ef64a8c134614827497d53f5be20d : SUCCESS in 18m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e7bb401c04d41ebb606cd3f609b0fda : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/682208b5cc0a497689b1e82aeda4e0b4 : SUCCESS in 18m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d868751776d34bc09bdea70f1c1e84b7 : SUCCESS in 44m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2338989514c4833958976264cb485f7 : SUCCESS in 40m 23s","accounts_in_message":[],"_revision_number":2},{"id":"0764da2946c0e05ee209d60cb2904ecdd6b4d8ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 15:56:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"10a08ba0befd0b619d4c31ee6dfbfa7c4542ad95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 15:57:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bc4c5f9a6b440c38226f11bf3ac1144\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c6517bd8d8d452ca226987437653039 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c98d4fda69de41418dbf4ef6b9e9c75c : SUCCESS in 1m 04s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b5e6964a223eb7b5bd3225c6da66ddced11bcd96","revisions":{"39983167ea0291ff5c768c49d7b53f0161523ed1":{"kind":"REWORK","_number":1,"created":"2025-08-22 15:45:06.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/91/958291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/958291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/958291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/958291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/958291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/958291/1"}}},"commit":{"parents":[{"commit":"c4cc83c5e7083cad04f8fb9f620b03b5b73e3087","subject":"s3api compat tests: stop asserting DisplayName in Owner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4cc83c5e7083cad04f8fb9f620b03b5b73e3087"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-08-22 15:35:23.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-08-22 15:44:30.000000000","tz":-300},"subject":"s3api: fix test_service with pre-existing buckets","message":"s3api: fix test_service with pre-existing buckets\n\nThe s3api cross-compat tests in test_service weren\u0027t sophisticated\nenough to account for real s3 session credentials that could see actual\naws s3 buckets (or a vsaio you actually use) - however valid assertions\non the authorization logic doesn\u0027t actually require such a strictly\nclean slate.\n\nDrive-by: prefer test config option without double negative\n\nRelated-Change-Id: I811642fccd916bd9ef71846a8108d50a462740f0\nChange-Id: Ifab08cfe72f12d80e2196ad9b9b7876ace5825b4\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39983167ea0291ff5c768c49d7b53f0161523ed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39983167ea0291ff5c768c49d7b53f0161523ed1"}]},"branch":"refs/heads/master"},"b5e6964a223eb7b5bd3225c6da66ddced11bcd96":{"kind":"REWORK","_number":2,"created":"2025-09-03 17:01:08.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/91/958291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/958291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/958291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/958291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/958291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/958291/2"}}},"commit":{"parents":[{"commit":"c4cc83c5e7083cad04f8fb9f620b03b5b73e3087","subject":"s3api compat tests: stop asserting DisplayName in Owner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4cc83c5e7083cad04f8fb9f620b03b5b73e3087"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-08-22 15:35:23.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-03 16:57:19.000000000","tz":-300},"subject":"s3api: fix test_service with pre-existing buckets","message":"s3api: fix test_service with pre-existing buckets\n\nThe s3api cross-compat tests in test_service weren\u0027t sophisticated\nenough to account for real s3 session credentials that could see actual\naws s3 buckets (or a vsaio you actually use) - however valid assertions\non the authorization logic doesn\u0027t actually require such a strictly\nclean slate.\n\nDrive-by: prefer test config option without double negative, and update\nansible that\u0027s based on the sample config.\n\nRelated-Change-Id: I811642fccd916bd9ef71846a8108d50a462740f0\nChange-Id: Ifab08cfe72f12d80e2196ad9b9b7876ace5825b4\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5e6964a223eb7b5bd3225c6da66ddced11bcd96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5e6964a223eb7b5bd3225c6da66ddced11bcd96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
