)]}'
{"id":"openstack%2Fswift~958292","triplet_id":"openstack%2Fswift~master~Id175329d717bb4959def8b43765fec668aa26696","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Id175329d717bb4959def8b43765fec668aa26696","subject":"docs: Fix python3 package names","status":"MERGED","created":"2025-08-22 16:02:14.000000000","updated":"2025-08-22 18:43:16.000000000","submitted":"2025-08-22 18:41:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958292","meta_rev_id":"a12d16b97db8104d3b81d307d8f9ea6fe43ea1c1","_number":958292,"virtual_id_number":958292,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-22 18:41:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-22 18:26:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-22 18:26:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-22 16:18:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bbcb8eaac80a6ee480dedda0a791b523e103d39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-22 16:02:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a06662e6e5e1abdcbdef1f13ccb4bae9a3dbb29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 16:18:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cb1709ce38b4f77b43254e469add428\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8bdb33b792e43d6a39b00b3331d6eea : SUCCESS in 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c45921f78b8bc3bc6cf0968cab4e163b1e0f0501","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-22 18:26:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7f605318316ce9fa1a2f175d7cf4ff40f2ec64e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 18:26:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e77b0bf33c39a0ce343499aba9be3e3d3205df4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 18:41:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61c496d9ba8b4df1aff3a741ba779bb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/227e5567a6aa4f67a1c77d809bffb0b6 : SUCCESS in 10m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0beaec92899e44b796b5616a11a010df : SUCCESS in 5m 33s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/08bf4dcbf6a24351bae252c7cbd9123c : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f13281b2b0df85cfa1d196e466bc491843b4628e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 18:41:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a12d16b97db8104d3b81d307d8f9ea6fe43ea1c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 18:43:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f54edfbf4d654c998d2c618bbddd92e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38e6153cb4254703aa0fbfa694f50a8a : SUCCESS in 1m 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/20309afbf4c24adc862f2e61cd929937 : FAILURE in 29s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa3a4eeb277e5a65ed2b53b3ff110b0014af57e4","revisions":{"aa3a4eeb277e5a65ed2b53b3ff110b0014af57e4":{"kind":"REWORK","_number":1,"created":"2025-08-22 16:02:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/958292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/958292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/958292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/958292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/958292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/958292/1"}}},"commit":{"parents":[{"commit":"457af40c9b4efcd3cdbb35cb8a16bda79bf80569","subject":"Merge \"s3api compat tests: stop asserting DisplayName in Owner\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/457af40c9b4efcd3cdbb35cb8a16bda79bf80569"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-22 16:02:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-22 16:02:01.000000000","tz":-420},"subject":"docs: Fix python3 package names","message":"docs: Fix python3 package names\n\nCloses-Bug: #2120783\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Id175329d717bb4959def8b43765fec668aa26696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa3a4eeb277e5a65ed2b53b3ff110b0014af57e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa3a4eeb277e5a65ed2b53b3ff110b0014af57e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
