)]}'
{"id":"openstack%2Fswift~958803","triplet_id":"openstack%2Fswift~master~I6fb806d299fc30f6ceaeba78cf3a810298e94f26","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-08-28 22:04:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I6fb806d299fc30f6ceaeba78cf3a810298e94f26","subject":"Switch py39 jobs to use py3-constraints.txt","status":"MERGED","created":"2025-08-28 18:51:41.000000000","updated":"2025-08-29 01:16:21.000000000","submitted":"2025-08-29 01:13:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958803","meta_rev_id":"1df9455faa803f4e1b9d36f12996dcc59ee6847b","_number":958803,"virtual_id_number":958803,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-29 01:13:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-28 22:04:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-28 22:04:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-28 19:58:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6723958d27b94ea1aa0f9fc138461cd4b1859a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 18:51:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b8dc5f892a1a9f9f76ebf2a3b97b6bcfc7c5665","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 18:54:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96e3c292582d3f532ae77bde3181de2b1230b5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 19:58:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd0bef5e7f59474f8650855aa0be60e3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/411ba5a2f0494d198730156069984854 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad0afcd799634327b4fa3bb46cccf010 : SUCCESS in 13m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/304808b13dda4ca996f7674450a684e7 : SUCCESS in 6m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/1c364785af3340deb54e013c283d6765 : SUCCESS in 27m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ae0990154e074f69951eb9f8ec6b9dfd : SUCCESS in 30m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ea9c6aff30d241a79fdb1bd19facd603 : SUCCESS in 33m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26cfec72372445ef83a1d90a7bc17fbd : SUCCESS in 59m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cba36b370f8f4dc2a046943614ceb406 : SUCCESS in 10m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/119448f896ec45349b06712beaf9e4fb : SUCCESS in 10m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee335d55f91f4e7197d811c367c577bd : SUCCESS in 19m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c8ff8f94cbbd4cba8a77c8b467ea669c : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1aa50a58f6414c239fb8ed0fe415e223 : SUCCESS in 25m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/76d8be36dba14208962739589a79922a : SUCCESS in 9m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d52a5db1ff414b729829c13511452f5a : SUCCESS in 11m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7d38924c081842808bf760821acd7f9a : SUCCESS in 11m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1f65117003274696b7421c04699ccf8d : SUCCESS in 13m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9c74848f1ccb46709a6cb5ec5653d71f : SUCCESS in 9m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e05fbd8c84cd4f01913474c225ba5952 : SUCCESS in 13m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a92e411198534248a51235536664f83a : SUCCESS in 8m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f5c03cf347444c318f20f9e7debd31b3 : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/732cc97c462a4ec284cfa65eae3694c3 : SUCCESS in 8m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a91acccc42b46809b13081913123b5f : SUCCESS in 58m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6909c27891ea45faac53524c7679eba3 : SUCCESS in 49m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f04b2c97f4746b0a174d4487d6caa04 : SUCCESS in 22m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f5293de744543aaa94a7f42cfecbab9 : SUCCESS in 24m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59bad3503a9d40368c5e6ca38dd032b3 : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ea1ac80be4d3457292a19c0acbbf6e3f : SUCCESS in 10m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ac698c8d5d34154b260a65696207ff2 : SUCCESS in 32m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0711216ea41a4c2bb013d97b2f80305a : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a90eae8752355aba16c93c638c625af0a9113224","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 20:12:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9498cfa060d4b65a9689f033a7eb93c7e523064","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 20:13:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5015921bdf28020b01893b93acfcfadc7782e8a0","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 20:13:15.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cec4a5ab295f404e8f64b3cdaea18e56\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5eb27e75f996406cbb65ec768d7ad084 : SUCCESS in 1h 07m 50s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/e2233cfae5a644fe9cb7fd3ae29664b0 : SUCCESS in 36m 18s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/a77ee8a4a46e4caf86b0609b64dac15b : SUCCESS in 47m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f193ef506cf24b0eaa6ddb29d6a12df6 : SUCCESS in 17m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a03608677a584adaa212135cae69e1de : SUCCESS in 21m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4b8d29320ea14d0aa0f67447f8162193 : SUCCESS in 27m 32s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/96a3c0af2ebe45e1a6e300e6c384efcd : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/1dc12d5c1582408cb2bcd1c3ceaaaa52 : SUCCESS in 18m 30s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe6e7512b5dc400d903a88f1281363cf : SUCCESS in 19m 31s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/adf93226290f40f29ad453ebd968f347 : FAILURE in 27m 30s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/15bf40530221419595f73716f16d5a54 : SUCCESS in 21m 21s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/9f8402cb9ba84153bb1d27aa8c754c8c : SUCCESS in 18m 11s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/f322e89ae8b14821a895cc4aab650309 : SUCCESS in 18m 22s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/3e1af26711f242cda3ee1d8eb5cee256 : SUCCESS in 19m 07s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/9b03cfd2c5d24c5ba36058776a4df1bf : SUCCESS in 17m 48s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/0866f2990f3a4fa995cd20943520a17f : SUCCESS in 17m 54s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/14871a3a5ec74957b2f4fff7dfde7732 : SUCCESS in 8m 55s\n- swift-multinode-rolling-upgrade-epoxy https://zuul.opendev.org/t/openstack/build/1a3418dff22c4ea68c15b864c9fd9554 : SUCCESS in 27m 45s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/23716deb1db344f094acd9e90128989b : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a7f7137103a8b0f09698e5d9658d69f76e817094","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 21:25:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81980eae64584efd938d2c707d81636f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d2a456530e454c15a058e6ec1ac1e630 : SUCCESS in 1h 14m 24s","accounts_in_message":[],"_revision_number":1},{"id":"bf529e6dd96b72f66d98836fec1d06625c43a4c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 21:29:06.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"19cb8eed0c8483719e68e68a6ebf03d43399d8b5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 22:04:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0eac621d774e699f24f1424ac0a6529262b00b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 22:44:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/203f041c13ae49f2a79138383b1c7396\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a0755eac885941bab66ad084b85e39c5 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d66eb7df0da424893e62ae7c8057506 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e17bc3db37704f62b35fe8b7778d6c18 : SUCCESS in 7m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/ec04607c4eef457fbaf0de82777e1481 : SUCCESS in 1h 00m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b64174025012472c8b5858dc7ca3a5df : SUCCESS in 59m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/84f66472148b4a2d9c5e1410bf2a67b9 : SUCCESS in 35m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e6dfb6c27a148a2bbf24632976512a4 : SUCCESS in 1h 04m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0a906c88bf8407a88375a40e1417ebc : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d5aef245521497d866c841f2adcddd8 : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0bfe65cdd7a481ab4256f945288b411 : SUCCESS in 19m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/456a0b2140424821979cb62de8fa900a : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1f34ee58e03244fbb325a5c9608638ee : SUCCESS in 11m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c134b80332eb49f2a2999b34b0223946 : SUCCESS in 23m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/624c4fc72c824216af0879e27a95123d : SUCCESS in 30m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4c24acfede0243be9ac92a98b802a79e : SUCCESS in 28m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d0e603f1932f4dcf86042f97f2cc84ff : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bcc95dab274d4f5aa6468a947cb3474f : SUCCESS in 15m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4f8a55001b2149909681322277683294 : SUCCESS in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/105b54b9360d41d09066f1a2d58da455 : SUCCESS in 7m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b5b70df5864f431184867d45232d04ef : SUCCESS in 14m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/201cebceff5a40d19e101b5974044141 : SUCCESS in 5m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e765b593cd234ac8b663b5b9b37078f1 : SUCCESS in 36m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b109d90f5e94ba4ba5c435d7e7c8d16 : SUCCESS in 51m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1380eb53688643dd9184b0eb1a69085f : SUCCESS in 52m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96e3cfda5b1e4a5587b1aecf8cb081de : SUCCESS in 28m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b97909d84c4072bb3a2d41060dba2e : SUCCESS in 2m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/117b46132c7643a4a979ccf8ef8f3ca5 : SUCCESS in 19m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96c8fa0302f24d8b9374ae82dd2d48ac : SUCCESS in 35m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a1a265ce59f4f1190e3f8233cd23737 : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":2},{"id":"c4f7335e4b05b6c7982ecdfba6eeab50ef2d90c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 22:45:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"829d09c7de7ed0a6144edb3c84f81e93a9a91580","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 00:55:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13cf9fa5d93a462b86c49f50c7cb741d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cbfa2d115dd14e578f9ba4998bf77342 : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d9d6be3454e5095b704b429309c0aeef9bddb45b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 01:13:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebbf55884e61404e97fc9f09445ead4f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cb0bac42b10341538625c7cb45de546a : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da1b08ecea184485811b4f0913b8f618 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c137cddf04e24567aadaa2711fc90d95 : SUCCESS in 6m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/b431fcef40c64643afe3f17c00cb8ad2 : SUCCESS in 1h 01m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a5249018cf0f47339a634708abc0885e : SUCCESS in 1h 03m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d90ca6370777446d9cf131cfb0e9e762 : SUCCESS in 49m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0692230e49db4e1ab433b8a0df69a6d1 : SUCCESS in 1h 04m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c4bf2c0470c84fbf9f37a00c793e7b4b : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fe6927b4a765429c8d16453df6f30bb3 : SUCCESS in 11m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/753d2212155b49098e34fb9dfa2682d8 : SUCCESS in 18m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8b4b0dcd280f44adac96438c7167dc48 : SUCCESS in 19m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/61cada3f4844440c8a46a21114124294 : SUCCESS in 11m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5bc8a96225ed4c608280bdbd7916f03b : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6a6782a9d5924543858887fe272a14ed : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9bbb791d607d4369b9b39c166754a643 : SUCCESS in 9m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a15eba825dee4699936bc1a61d54e27d : SUCCESS in 6m 16s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c735978b13094e0fa734d97d5c98e002 : SUCCESS in 5m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/156290c34cab4dc6a0dcecb6c4ea5a80 : SUCCESS in 51m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95ad8eb66f604889aec2c075e1d57620 : SUCCESS in 51m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1bce2f82adcc409d84f650fbd5187026 : SUCCESS in 50m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06e5f0f7327d4fd6811f530c2ac63f8e : SUCCESS in 24m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e1f9a6bd363428fba8826400e9562b3 : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10f2bdb075594fc9924083534f14624e : SUCCESS in 8m 54s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7577311067a47b6812e8d6c03dc6795 : SUCCESS in 44m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8d6bc3cba444fc58e011dc9421b6c8c : SUCCESS in 43m 32s","accounts_in_message":[],"_revision_number":2},{"id":"512926bec02346fffe0db58dc6d1ecd6e9f6f63c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 01:13:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1df9455faa803f4e1b9d36f12996dcc59ee6847b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 01:16:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5cf2696fb1848a3beedcb2736b5f097\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0aba069dff4c4b6d8e588ff8846104e7 : SUCCESS in 1m 15s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e52eb41b7454ec18470e14bc6e68a6c : SUCCESS in 2m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd47bb902c114918856220075d36df43 : SUCCESS in 2m 17s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9e3e299868e24d74ae2e6537e67cbe83 : SUCCESS in 52s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a18fb08b48b9a728a52fb06600791f7f9e5d48b6","revisions":{"3c4ec9d1b3cb9a473e7d622d1fe9144e2a09cdc6":{"kind":"REWORK","_number":1,"created":"2025-08-28 18:51:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/958803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/958803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/958803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/958803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/958803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/958803/1"}}},"commit":{"parents":[{"commit":"e10c2bafcb8fc80929bce3e064fee5f86a49e718","subject":"Merge \"proxy-logging: create field for access_user_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e10c2bafcb8fc80929bce3e064fee5f86a49e718"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 17:58:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 17:58:52.000000000","tz":-420},"subject":"Switch py39 jobs to use py3-constraints.txt","message":"Switch py39 jobs to use py3-constraints.txt\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I6fb806d299fc30f6ceaeba78cf3a810298e94f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c4ec9d1b3cb9a473e7d622d1fe9144e2a09cdc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c4ec9d1b3cb9a473e7d622d1fe9144e2a09cdc6"}]},"branch":"refs/heads/master"},"a18fb08b48b9a728a52fb06600791f7f9e5d48b6":{"kind":"NO_CHANGE","_number":2,"created":"2025-08-28 21:29:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/958803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/958803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/958803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/958803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/958803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/958803/2"}}},"commit":{"parents":[{"commit":"e10c2bafcb8fc80929bce3e064fee5f86a49e718","subject":"Merge \"proxy-logging: create field for access_user_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e10c2bafcb8fc80929bce3e064fee5f86a49e718"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 17:58:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 21:09:56.000000000","tz":-420},"subject":"Switch py39 jobs to use py3-constraints.txt","message":"Switch py39 jobs to use py3-constraints.txt\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I6fb806d299fc30f6ceaeba78cf3a810298e94f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a18fb08b48b9a728a52fb06600791f7f9e5d48b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a18fb08b48b9a728a52fb06600791f7f9e5d48b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
