)]}'
{"id":"openstack%2Fswift~958804","triplet_id":"openstack%2Fswift~master~I481efbd2627e517edf49f3025b3399a86d1b4f3e","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-08-28 22:05:33.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I481efbd2627e517edf49f3025b3399a86d1b4f3e","subject":"Update py3-constraints.txt","status":"MERGED","created":"2025-08-28 18:51:41.000000000","updated":"2025-08-29 01:18:39.000000000","submitted":"2025-08-29 01:15:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958804","meta_rev_id":"875178d9b26816aabc8613df32552d47ef1a8866","_number":958804,"virtual_id_number":958804,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-29 01:15:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-28 22:48:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-28 22:48:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-28 21:15:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5efbb175a4b66ecfc62fcc5ce5a8897e48ef21fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 18:51:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a261071a8eb0d414326fbbbac266fc1e89bdd1f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 18:54:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d291c205d1ea244b57391b603441daae0e5d9243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 19:10:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b89f52eb2647f67d31acc46d36002b222653b3e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 19:59:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee337bc89e69cd5ae4b636897c5d507d76b40cfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 20:07:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"66c27a12f1f5f6363c5046c12ef46024d09a2657","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 20:20:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8b1d14303384a38eb3fcade5608604d334c150b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 21:15:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a61ad4ae6c0f468490f5ad468628c436\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff559febdec0456f9b2e1f3fcae9cf9e : SUCCESS in 11m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/315008f3369f46b188a33c42e247ed7e : FAILURE in 20m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38e5705134fb40cbb7027ca13e69018b : SUCCESS in 56m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9c722cdb5e8440ec981056f0959a180b : SUCCESS in 53m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82afe627d88442c88103e1dcc80c3e2b : SUCCESS in 1h 06m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/617350964479447096a6ec965480f6e1 : SUCCESS in 20m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/773ddac84b1a412fbe460949d3830d2d : SUCCESS in 12m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c578d664d2f34c11af2a41bbaff1518d : SUCCESS in 22m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9f0b5c66940f4340a25aa93c65770ced : SUCCESS in 8m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b387271d716a4d1b92ca7a5622e514d8 : SUCCESS in 27m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e33ecfb4f86041fe9113592bcfc82208 : SUCCESS in 19m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/134cbf54de2b403fb77c0a09a0e634ad : SUCCESS in 10m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1cb3476c33534f2384c925995344d53f : SUCCESS in 38m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3dd4ec40859d4978b303162099d7cf0e : SUCCESS in 20m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8df13e5c00c64cedb54e2b7d9c7a8288 : SUCCESS in 16m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d75a409cd8d443879afcdc48a45cd0b0 : SUCCESS in 9m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/da3466d81c0d47a99f508c9fa67d13f1 : SUCCESS in 4m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/98a7e8f56fd04c739ac31a58c414ded1 : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d2f7edff0b748458d16518c9b068f5d : SUCCESS in 9m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8919a65f07bf4a588a2a59b0f5ff41b7 : SUCCESS in 51m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8919567ffb5940b19a73e63c510ca00b : SUCCESS in 22m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/18080866d222471da7087a60d4ffc605 : SUCCESS in 52m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/075d6f2666424798b367509d170fb7cb : SUCCESS in 27m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21380a65b29d43618f0ded595ba0526d : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/97e4c2a271aa4c728d0fc0c8dafe7cc7 : SUCCESS in 18m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae7e1995c518484090d1f26275581664 : SUCCESS in 35m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6496c0a43f684ab0b375ba1ff1a4c914 : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":4},{"id":"4bd734cba59ed5d988da0715375612e4085abc2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 21:29:06.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"782a99d31d405508b6540dd660ad7c7239de411f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 22:05:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aed878ff659e1aa7eee87aefd173495475195009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 22:36:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c392236fdf7648dba52f2f591f6c2876\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea69420087d144ac907177691ff16928 : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/f998d6b84b8145479278495a06542e06 : SUCCESS in 1h 00m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b8fe872dd3e249c98eed0556c0527ee6 : SUCCESS in 36m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9033ce8e14d14e2884b729e85d6a2e8c : SUCCESS in 49m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4c67f94582d41b5be1260f69c43d096 : SUCCESS in 59m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d147bbc4d77b45e3926a4d77e4a7ff4d : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ccbedbe0b944bb1a6b0a032d2628aa0 : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/532b2d368ecf4b18aa203fa70923b9ba : SUCCESS in 20m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/683d70f57c7241a0bb683a799da91f6c : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/45bc1da31b4a4ceca64a828756d66ecc : SUCCESS in 26m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f7b3533796d4a3ba54fb0821add8b2c : SUCCESS in 26m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/16fb598862c846c3aa37fca211253fa5 : SUCCESS in 22m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/badaa961ce174e7bba2fb9b58cec155c : SUCCESS in 28m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/45d0853cfc2b4875afc17960e19e4883 : SUCCESS in 8m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c5da8470e6d2498997634efce68681c6 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/56e02e8cc6aa4470afae2f229dd1fdd0 : SUCCESS in 15m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/82c071e5dec24ba2818bca99b441876f : SUCCESS in 6m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/81d91c8b78ec49a897660c84040cbb22 : SUCCESS in 8m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b4e694c91c74457dbf5c6a3f40a98188 : SUCCESS in 4m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/de2d16a3bbaf4c80bbe327f2950c8903 : SUCCESS in 1h 00m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9ad9cad5883e4c7abd34da0eaaa88c38 : SUCCESS in 51m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a763c17dbf1a4ecf8457897f155ee8ca : SUCCESS in 49m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9cde8a1b5f04af4a1c55207a1bae85d : SUCCESS in 27m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cfbb3a798e2443eb4bd4b69dc50f0fa : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf2de9bcea9d4014b038e13c61fcf976 : SUCCESS in 8m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71d09d0af7864967bfaa9ce3d40554d8 : SUCCESS in 38m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14a5e9cfd956447e9719088887295598 : SUCCESS in 55m 22s","accounts_in_message":[],"_revision_number":5},{"id":"01bd7598e21988056011588ce1dd4d46608dab98","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 22:48:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"69d920bae8132b09792e83c23b16983a19c35f1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 22:49:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1626ebd69eccaae83f4686e1474e718778ab3178","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 23:20:39.000000000","message":"Patch Set 5:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/577f482fdbad4ee384695f729cbae1a3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec886e4476dd4e6a876cec6059c2168f : SUCCESS in 1h 05m 30s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/89f0bb17d72c43febbd5453f7c475502 : FAILURE in 46m 24s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/3be5f898a4324bad86b456efd05efa13 : SUCCESS in 37m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/41cf003ad55f44f2b7366671d64a5e84 : SUCCESS in 19m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/42e2d56c61c04540b37c7f2b6566e36f : SUCCESS in 19m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/679153a474574d5193e10d19180d4f27 : SUCCESS in 8m 50s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c685111ff0524b7db7d13fc0ce2ec70c : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/33ca7eb5cd5f4202b75a56d73c00bbe5 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6971dcb40d944c798777b8090bd3c2c : SUCCESS in 19m 23s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/32cc59c53edb4833a4df406826802439 : SUCCESS in 27m 20s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/38082a3afa634f60ad983b7e2ed28c40 : SUCCESS in 18m 47s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/d27bafe5976f4f9d99ca5b568cb14077 : SUCCESS in 19m 14s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/a5616f4761b24b9b9d2d3e4cc22017eb : SUCCESS in 18m 33s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/9c7fae1621334f2eb94baedfb8acc7be : SUCCESS in 8m 34s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/ca49fc0afabb445ca8a30b685007f3ef : SUCCESS in 19m 04s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/9c164a26411440899a3ff543a217a9d7 : SUCCESS in 18m 27s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/35e2264dbed04e6d89b08c0355fade99 : SUCCESS in 10m 41s\n- swift-multinode-rolling-upgrade-epoxy https://zuul.opendev.org/t/openstack/build/d34f29bd52e3487fbff53c033d230e81 : SUCCESS in 16m 55s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/396cf5b4ecd44f82b6bc85e53771e255 : SUCCESS in 18m 54s","accounts_in_message":[],"_revision_number":5},{"id":"4e66ebc1206388873fefa8ff75cef16436c95007","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 00:32:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4304a6b129ff47fd96797f88d213ec34\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cce8be6250394d8988313b6eb85fd954 : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":5},{"id":"a99aa9987874271fdcae6c9790382b3e6b64238b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 01:15:50.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17120ff6edcf48ad9efde5512f2a08a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00b1fc6601a04460b090daf0475c4c86 : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/ff393e100e5d4a58bcb3a285df689da0 : SUCCESS in 57m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/016b587c2d6d459b9779826ecd2dbdd5 : SUCCESS in 1h 01m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d77b44b83d8246698e2b7c6e5ad4afe4 : SUCCESS in 56m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98f5b07aed7f4a9ea6bc87c7ac619f3b : SUCCESS in 59m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6917c8244597448a89b4adc24b2c3a52 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0a5b08e7011b42378f402a2a7b7e433c : SUCCESS in 24m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d834459743143a3bc46a934f11b38f0 : SUCCESS in 19m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/db08ffc6021448f3a0362051eadd4bf6 : SUCCESS in 12m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/60de51282e184c00b1bd90dc1f188097 : SUCCESS in 11m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/377a4594b7e7477692e65e01c6db9e8e : SUCCESS in 8m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/52d6f3c6f73447a3a95317bddca5ce0a : SUCCESS in 14m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b63113cfe84a4fbeb54937851764be6c : SUCCESS in 8m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e016bdb83d5348fe9a9c9a66190f5c6e : SUCCESS in 7m 14s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8c6dfeed45d34cc8b4d16d4841bae23b : SUCCESS in 6m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7421fc401c6e4fb1a6d1dc27193b077f : SUCCESS in 37m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e7a23186406472a8a6545be19e584af : SUCCESS in 35m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7573419d1c5b48c48d715c664429e5b5 : SUCCESS in 31m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f13a282edcc4440b06194123b6bdc7c : SUCCESS in 24m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9ec186909a04963b16c5612c09b6203 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/252fe730238e4a2a8d64dd2182689f25 : SUCCESS in 19m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2822127576674a9a9996f68f57d51459 : SUCCESS in 48m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4be4416953004c5fa372eee70264a949 : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":5},{"id":"b58817beec906b4924afcc51fe14cfd341b7e4b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 01:15:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"875178d9b26816aabc8613df32552d47ef1a8866","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 01:18:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8cc4a2da7cf4d1fbaef4e1c56c7a604\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b46e681964a5464d8d8e2d33cee3fc6a : SUCCESS in 2m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/76ac09bcd3234aa592ea51944d9bb927 : SUCCESS in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1ed7b71bb56436e4aafc8210117b1d3c404e0bd8","revisions":{"42f34432c1f0af0bc759ce8f7ce3c92e2a239290":{"kind":"REWORK","_number":1,"created":"2025-08-28 18:51:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/04/958804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/958804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/958804/1"}}},"commit":{"parents":[{"commit":"3c4ec9d1b3cb9a473e7d622d1fe9144e2a09cdc6","subject":"Switch py39 jobs to use py3-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c4ec9d1b3cb9a473e7d622d1fe9144e2a09cdc6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 18:50:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 18:50:59.000000000","tz":-420},"subject":"Update py3-constraints.txt","message":"Update py3-constraints.txt\n\nWe\u0027re starting to see projects drop py39 support; seems as good a time\nas any for an update.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I481efbd2627e517edf49f3025b3399a86d1b4f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42f34432c1f0af0bc759ce8f7ce3c92e2a239290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42f34432c1f0af0bc759ce8f7ce3c92e2a239290"}]},"branch":"refs/heads/master"},"4a380459144f6f5f73e6e9f23079db2119065e11":{"kind":"REWORK","_number":2,"created":"2025-08-28 19:10:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/04/958804/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/958804/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/958804/2"}}},"commit":{"parents":[{"commit":"3c4ec9d1b3cb9a473e7d622d1fe9144e2a09cdc6","subject":"Switch py39 jobs to use py3-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c4ec9d1b3cb9a473e7d622d1fe9144e2a09cdc6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 18:50:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 19:10:08.000000000","tz":-420},"subject":"Update py3-constraints.txt","message":"Update py3-constraints.txt\n\nWe\u0027re starting to see projects drop py39 support; seems as good a time\nas any for an update.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I481efbd2627e517edf49f3025b3399a86d1b4f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a380459144f6f5f73e6e9f23079db2119065e11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a380459144f6f5f73e6e9f23079db2119065e11"}]},"branch":"refs/heads/master"},"136d00a6095542364cefef97804c9a18476f717d":{"kind":"REWORK","_number":3,"created":"2025-08-28 19:59:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/04/958804/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/958804/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/958804/3"}}},"commit":{"parents":[{"commit":"3c4ec9d1b3cb9a473e7d622d1fe9144e2a09cdc6","subject":"Switch py39 jobs to use py3-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c4ec9d1b3cb9a473e7d622d1fe9144e2a09cdc6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 18:50:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 19:59:51.000000000","tz":-420},"subject":"Update py3-constraints.txt","message":"Update py3-constraints.txt\n\nWe\u0027re starting to see projects drop py39 support; seems as good a time\nas any for an update.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I481efbd2627e517edf49f3025b3399a86d1b4f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/136d00a6095542364cefef97804c9a18476f717d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/136d00a6095542364cefef97804c9a18476f717d"}]},"branch":"refs/heads/master"},"2c9fd11112aa8dd5e2c64fdcd97a0a65a8d64d2f":{"kind":"REWORK","_number":4,"created":"2025-08-28 20:07:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/04/958804/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/958804/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/958804/4"}}},"commit":{"parents":[{"commit":"3c4ec9d1b3cb9a473e7d622d1fe9144e2a09cdc6","subject":"Switch py39 jobs to use py3-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c4ec9d1b3cb9a473e7d622d1fe9144e2a09cdc6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 18:50:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 20:07:02.000000000","tz":-420},"subject":"Update py3-constraints.txt","message":"Update py3-constraints.txt\n\nWe\u0027re starting to see projects drop py39 support; seems as good a time\nas any for an update.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I481efbd2627e517edf49f3025b3399a86d1b4f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c9fd11112aa8dd5e2c64fdcd97a0a65a8d64d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c9fd11112aa8dd5e2c64fdcd97a0a65a8d64d2f"}]},"branch":"refs/heads/master"},"1ed7b71bb56436e4aafc8210117b1d3c404e0bd8":{"kind":"NO_CHANGE","_number":5,"created":"2025-08-28 21:29:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/04/958804/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/958804/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/958804/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/958804/5"}}},"commit":{"parents":[{"commit":"a18fb08b48b9a728a52fb06600791f7f9e5d48b6","subject":"Switch py39 jobs to use py3-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a18fb08b48b9a728a52fb06600791f7f9e5d48b6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 18:50:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 21:09:56.000000000","tz":-420},"subject":"Update py3-constraints.txt","message":"Update py3-constraints.txt\n\nWe\u0027re starting to see projects drop py39 support; seems as good a time\nas any for an update.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I481efbd2627e517edf49f3025b3399a86d1b4f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ed7b71bb56436e4aafc8210117b1d3c404e0bd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ed7b71bb56436e4aafc8210117b1d3c404e0bd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
