)]}'
{"id":"openstack%2Fswift~958811","triplet_id":"openstack%2Fswift~master~I26be3752b8c67b7a6a0a9c75571a44f3827cbb90","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-08-29 11:39:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I26be3752b8c67b7a6a0a9c75571a44f3827cbb90","subject":"CI: Remove a bunch of unnecessary bindep profiles","status":"MERGED","created":"2025-08-28 20:38:53.000000000","updated":"2025-08-29 11:43:49.000000000","submitted":"2025-08-29 11:39:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958811","meta_rev_id":"c85ad44360760b8f9bef556b7f5d01ff19bec424","_number":958811,"virtual_id_number":958811,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-29 11:39:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-28 22:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-28 22:51:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-28 22:59:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-29 08:00:08.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"f703d2b759f3c26fab044a664d139deff55158a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 20:38:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27d5aac123e75be26faddc32bdd2f4d82e24cc78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 21:29:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"273ac0ba92e199e63dc42ba2031e526d08ee5939","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 22:51:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"13401b03df1ddbad8ea1471512d6b87c703a80ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 22:59:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6409f849f64401fb046028370fe9dbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/069f69bf28ca4bd58b4a16a57abf1721 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/9be5a454b51c40f09191ba5e2718a056 : SUCCESS in 53m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fc57a9ab06614949a680229692d4cdfd : SUCCESS in 59m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9470bb64175143c3a29a4f2653726ffb : SUCCESS in 39m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29b0255345014342ae97bb60b9199d55 : SUCCESS in 52m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb7ac3bca97c462398aaca342b8a671d : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e73d37369ee44fd28ec5f0660dbfc883 : SUCCESS in 11m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70e85e9b388d4a068df579f64c5743d5 : SUCCESS in 11m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/da3632e86d7c4e4c8db174e16c11d803 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/84ad2739889d4a3a8d0632aec670ff06 : SUCCESS in 24m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/371a7d06e5f543c1a7f35db045de274c : SUCCESS in 11m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/85b07d46e2bd4f3fb9f74f96edc0344e : SUCCESS in 28m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/eee1ddac2965408fbffb97c059e5ca88 : SUCCESS in 22m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/52f39f79a8fb4d71a69f1a4d790016e5 : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2db0b7652df840c6bc0fd4cefe42fe52 : SUCCESS in 8m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/59bfcd0fa0cf40728462e9ab43eb4bec : SUCCESS in 16m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6ae14fed54c5456996fb75df61d6c3a2 : SUCCESS in 9m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1af24e65f96a474c80768629c22df700 : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7ca3f0381352415c8246903fc563bb11 : SUCCESS in 9m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/600185acefee4e939af4300e9318e213 : SUCCESS in 1h 23m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85c7b67b664e494086dcece04fc87294 : SUCCESS in 48m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a755f08a429443ceb53b2d4f559c3885 : SUCCESS in 40m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/282e348d1478473597af7549fb518816 : SUCCESS in 38m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ea4275afce439ebac979a779ade15f : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6cc6d7c7d2ef4d16be42a9f25288f4fd : SUCCESS in 25m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8bcfb3913554cd6853d100ae3e7ca8d : SUCCESS in 1h 06m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2181f761c7cb4390a4eecd72fec09834 : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":2},{"id":"0e30cb3cf6e4b0c69df79dd02ef21f581141e9cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 23:01:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0717557b224f191293254f2e3676ea06960063e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 01:25:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c4260aadf5246df9de1f216d9d1e738\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0cc68a191974ed093ff413f3423753a : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/4bc8ee2f93454729888b9eb91d26756e : SUCCESS in 40m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/39b3ffabb52b43c98628b7ce841732a1 : SUCCESS in 1h 04m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e3ae8f1fa0ec41568ad470b187563aac : SUCCESS in 47m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/683c387aee5546fca538199255dcc9b1 : SUCCESS in 37m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c16bb03d07684d998bde3f90a242c542 : SUCCESS in 9m 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c835d84715084608a32eeaab54a642ae : SUCCESS in 10m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c8a16a044bb34579b2c7927835804867 : SUCCESS in 20m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bf992ef1183d486a9f97733b23467b42 : POST_FAILURE in 51m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cf7284d39bec49cbaf19e7af043c26b6 : SUCCESS in 25m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d9ed8b2379af4dfe85e424322399a6dc : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/096a0566ece8451abb342cf1a3be39d7 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8fb4afe1649b4644af575bfba1b1a347 : SUCCESS in 13m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f05874dfbb364cff96a66b87c773eb79 : SUCCESS in 3m 53s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c4587d72c8de4bd384d4bd28aede2e14 : SUCCESS in 5m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7865260869e942bb9218662f0fe5ee8e : SUCCESS in 58m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e5714bf4cd74460acee8e6c20a754ec : SUCCESS in 50m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3105cc6ee2804b52aa3015c4fafeaa7d : SUCCESS in 58m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9367b51cdddf49f88bbbab34cde35a09 : SUCCESS in 26m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf81b785c58e464db7d94d6b8aafa1cc : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cec4fca0cc824563816d94b692342594 : SUCCESS in 19m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c81a02b854784a588ecd2d95878a3a86 : SUCCESS in 38m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d39f690dd704d78ab1c2274cdb88774 : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":2},{"id":"1cf3bf3de1d2207b02f9f8c94801ff8dc410a84c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 01:27:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82c0fadf513c48b6a1d19c1c0d458994\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/784f0f7abd44409aaef0354b8f4c3f3f : SUCCESS in 1h 16m 43s","accounts_in_message":[],"_revision_number":2},{"id":"a7983b96597b6e65289a6c558cbe1bff37786bff","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-08-29 08:00:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f995f483d3e87af638342f09d63abc1aab37e0fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:23:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a20035742d8048afa79d0396a976f972\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d356235b91d948a1a960b5b790cbca41 : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/d0fd6125d78f4bd5bb04e647d24d8c18 : SUCCESS in 49m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4292b3b1db114c33971fb16478876062 : SUCCESS in 59m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f20a0c459f4847b6942bdaff9772cdb8 : SUCCESS in 34m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9cc02145fa584a75ab7325d0a0feaa0a : SUCCESS in 1h 00m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6a1696801064f34b46b22f6893bdad4 : SUCCESS in 21m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b8caab3f74841b48830a0c5ad082649 : SUCCESS in 21m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb8f515b24914f2a9f38177b9254d8eb : SUCCESS in 18m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7eec58bc81414fac8292fadd02e2cea9 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e8d10dee45bd4216beb66514d92cbc46 : SUCCESS in 28m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ddfba7596a3c461ca34b78a22f1af0f1 : SUCCESS in 19m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b1ffaae104314cf99c4d3d5ea489dc80 : SUCCESS in 19m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/18e88ce7be5d457c84bfdc39ca0fe7e2 : SUCCESS in 14m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/76a5a8df8b464c1e984780326bc3cf72 : SUCCESS in 13m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fb5183ef8a3749408f091613dfff9a5f : SUCCESS in 10m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2a7b24d03fa24fc58c2570e6ca5068d1 : SUCCESS in 18m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b31b6daf2fe0408fb764d9746a14c03c : SUCCESS in 7m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e26d32d4ca8145f3ac1e79304d649afd : SUCCESS in 13m 40s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/90eddbbc26bd41689160939a1419a127 : SUCCESS in 10m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/387e463162a549b2ad581464e8aa2384 : SUCCESS in 56m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85d5e03c5c0846b7836c2a19d1a34e5f : SUCCESS in 47m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/01c82e78f66f4ef9ba1a34f40c7264b9 : SUCCESS in 52m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1842841c20ac474e9ad88ce32706878a : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ece536624a4c509665826ccfdfae1a : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bcec95ef0ef24e87b9883e838b7fdc27 : SUCCESS in 20m 03s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46cf8ecbb0e74c5793e8401ec5ce639a : SUCCESS in 44m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4b565ee438c4384af02972d318fd96b : SUCCESS in 1h 14m 32s","accounts_in_message":[],"_revision_number":2},{"id":"5310d4fa66ab3141584af192fa9b728f7bf0b10e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:26:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"94f713aec8dfbb6c00d5b47a8dedab7e38d1553f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 11:14:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d7b913a3b3c4416ad8a92fd91354e6f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1d6df64de16e4b02a9f15695177ed967 : SUCCESS in 1h 15m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ffa985872599059ad091976f2f69b6233862866a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 11:39:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27e0dac0f14740708feef9b174699a94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e99945a66884c338769544fd8a1c73e : SUCCESS in 10m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/e88c13e1557243c5b8189b7f716b1e7a : SUCCESS in 55m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c9384a5f95ce49328fa97c0b86e05fc5 : SUCCESS in 1h 04m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f28e77484f94f1aba1fd456258d2591 : SUCCESS in 1h 00m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21f1e4ae7cdb42c59476debe2c7bcfd8 : SUCCESS in 32m 04s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/413eb2f981034996a5140c1c8f166dd7 : SUCCESS in 4m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2a5fe33948ad4c059092cbcaadf7c025 : SUCCESS in 42m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4f36fb53d5554db18466b7c30bebacff : SUCCESS in 8m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4c7527716428421aa327dd710d512d2f : SUCCESS in 22m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d8b12e642e4e4202b10723013352154d : SUCCESS in 27m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/91904a8f235a49ccad66f9b697056479 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5b6518ccd08c41ad9858ac9cfd0feccf : SUCCESS in 9m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b8663b55669046ab9b35d99eecd5464f : SUCCESS in 15m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9aea9dc965c4e3a87933d90e298aa64 : SUCCESS in 7m 15s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/086f33a5bede4997abbae9cbc3971dc6 : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f2b35f8c9994b8db1eb013710b43e6b : SUCCESS in 34m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b21b9e8b32834674aef50ebe4f0e8614 : SUCCESS in 51m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1a75c83c7f0f463cb93a6e6df4e163e8 : SUCCESS in 49m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63b922f1c13d496e934a6b417ef33fa0 : SUCCESS in 28m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cfee9ae712545c9ac0782078c790f59 : SUCCESS in 2m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aaff360381e742bf923e59b300a8bd20 : SUCCESS in 18m 33s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c389f19fa4546b78b1ea3282b149a04 : SUCCESS in 58m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/315deaa6bbab498e83d940525805df47 : SUCCESS in 59m 19s","accounts_in_message":[],"_revision_number":2},{"id":"d0bec983d3a076ae2533a691e1d1893a9cbaa1fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 11:39:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c85ad44360760b8f9bef556b7f5d01ff19bec424","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 11:43:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9cab248e04b4e1fb7c59b44231e6984\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4558e757bf154d199488b82378fcadc5 : SUCCESS in 2m 05s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/352f68708e91421799c3f42cda19a529 : SUCCESS in 56s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"21325988dfe7d63d1f3ef78b9af8a2297f787e1f","revisions":{"9e8fc130a74494b130821f140a920f497d04322c":{"kind":"REWORK","_number":1,"created":"2025-08-28 20:38:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/958811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/958811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/958811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/958811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/958811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/958811/1"}}},"commit":{"parents":[{"commit":"f86f6446539930e3686ba1c82d28931dc36f4253","subject":"Clean up some py36 infra","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f86f6446539930e3686ba1c82d28931dc36f4253"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 20:36:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 20:36:45.000000000","tz":-420},"subject":"CI: Remove a bunch of unnecessary bindep profiles","message":"CI: Remove a bunch of unnecessary bindep profiles\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I26be3752b8c67b7a6a0a9c75571a44f3827cbb90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e8fc130a74494b130821f140a920f497d04322c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e8fc130a74494b130821f140a920f497d04322c"}]},"branch":"refs/heads/master"},"21325988dfe7d63d1f3ef78b9af8a2297f787e1f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-28 21:29:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/958811/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/958811/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/958811/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/958811/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/958811/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/958811/2"}}},"commit":{"parents":[{"commit":"3db8e2d05fd17858e4c2b54ad2813a26e6112d5b","subject":"Clean up some py36 infra","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3db8e2d05fd17858e4c2b54ad2813a26e6112d5b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 20:36:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 21:10:35.000000000","tz":-420},"subject":"CI: Remove a bunch of unnecessary bindep profiles","message":"CI: Remove a bunch of unnecessary bindep profiles\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I26be3752b8c67b7a6a0a9c75571a44f3827cbb90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
