)]}'
{"id":"openstack%2Fswift~958812","triplet_id":"openstack%2Fswift~master~Iad899f0a1e7dd665d819b457a6545e391b4a9da5","project":"openstack/swift","branch":"master","attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-09-17 16:51:30.000000000","reason":"Vote got outdated and was removed: Code-Review+2"}},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-07-13 15:29:11.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Iad899f0a1e7dd665d819b457a6545e391b4a9da5","subject":"CI: Move (most) CentOS 9 Stream jobs to 10","status":"NEW","created":"2025-08-28 20:38:53.000000000","updated":"2026-09-17 18:11:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":14,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f4b6554eeaf2e7cd027e4d57372eea34e65c8354","_number":958812,"virtual_id_number":958812,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-17 18:11:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-28 20:40:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-29 08:00:44.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2026-07-10 15:35:16.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"fbbe5ee833139e9b056faf48d3320ff563e3ddac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 20:38:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8acf1d850b8b7ad423c1aefa9ccb4b2a9d943d4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 20:40:02.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  The nodeset \"centos-10-stream\" was not found.\n\nThe problem appears in the the \"swift-tox-func-py312-centos-10-stream\" job stanza:\n\n  job:\n      name: swift-tox-func-py312-centos-10-stream\n      parent: swift-tox-func-py312\n      nodeset: centos-10-stream\n\n  in \"openstack/swift/.zuul.yaml@master\", line 152","accounts_in_message":[],"_revision_number":1},{"id":"f0bfcd35f362f9860f3caacfa6c3c4192305a339","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 20:40:14.000000000","message":"Patch Set 1:\n\n(5 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  The nodeset \"centos-10-stream\" was not found.\n\nThe problem appears in the the \"swift-tox-func-py312-centos-10-stream\" job stanza:\n\n  job:\n      name: swift-tox-func-py312-centos-10-stream\n      parent: swift-tox-func-py312\n      nodeset: centos-10-stream\n\n  in \"openstack/swift/.zuul.yaml@master\", line 152","accounts_in_message":[],"_revision_number":1},{"id":"78173dfd45ff0a0145e0abb31d6337478e9b6dfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 21:29:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c68d9cc98b59988953c406f1ecd5757b2dc4ea6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 21:29:54.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  The label \"centos-10-stream-arm64\" was not found.\n\nThe problem appears in the the \"swift-probetests-centos-10-stream-arm64\" job stanza:\n\n  job:\n      name: swift-probetests-centos-10-stream-arm64\n      parent: swift-probetests-centos-10-stream\n      nodeset:\n        nodes:\n          - name: swift-centos-10-stream-arm64\n            label: centos-10-stream-arm64\n      description: |\n        Setup a SAIO dev environment and run Swift\u0027s probe tests\n  ...\n\n  in \"openstack/swift/.zuul.yaml@master\", line 330","accounts_in_message":[],"_revision_number":2},{"id":"97dcc45d289c0d8ab02b7bba54522942766abd37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 21:30:47.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  The label \"centos-10-stream-arm64\" was not found.\n\nThe problem appears in the the \"swift-probetests-centos-10-stream-arm64\" job stanza:\n\n  job:\n      name: swift-probetests-centos-10-stream-arm64\n      parent: swift-probetests-centos-10-stream\n      nodeset:\n        nodes:\n          - name: swift-centos-10-stream-arm64\n            label: centos-10-stream-arm64\n      description: |\n        Setup a SAIO dev environment and run Swift\u0027s probe tests\n  ...\n\n  in \"openstack/swift/.zuul.yaml@master\", line 330","accounts_in_message":[],"_revision_number":2},{"id":"8ad5ee787b188696032588bcfff1639d209fada0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-28 21:33:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6d7a85c190873f175cea85147b7b6522a27d39dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 22:38:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fca872c2c6d4e8a8ab0b83fc0e9c270\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009fd11ef1de4aa79d87759ba6617272 : SUCCESS in 10m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/91d04119c920404c96d80156e5755ab8 : SUCCESS in 58m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/42a51cd84d7b42fc8606741f37a60365 : SUCCESS in 34m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/01816339dd4e4c93bbb0f581713ce65f : SUCCESS in 32m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e764bfcd6054b71b2bf029ac30aedae : SUCCESS in 1h 01m 08s\n- swift-tox-func-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/5171308e906e4dd584647cb0cdbbb86d : RETRY_LIMIT in 1m 39s (non-voting)\n- swift-tox-func-encryption-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/d8bfa64daf044164b1d6cc604cfdc05a : RETRY_LIMIT in 2m 00s (non-voting)\n- swift-tox-func-ec-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/e6bd19e7486146759e08b626c3e30edc : RETRY_LIMIT in 2m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/adf6318bc6e7491ab80608f2eaf1f275 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9b59467408934397a991a2cb36e9c34d : SUCCESS in 26m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/86d67de8f40a4e72944e165225e25de8 : SUCCESS in 19m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e468b2447f7b4d6dbcc5a94a2a11fcca : SUCCESS in 28m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f62b27ca9270434780ab814bf7b6d575 : SUCCESS in 31m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e46667e061534a8fb2334bbc0ba89863 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1bc8668e506b4bbb9556ff8694920d19 : SUCCESS in 14m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c34cb0911fba465990b24a404d6f563a : SUCCESS in 12m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3a9e2557ff7440309f685788f7da41c2 : RETRY_LIMIT in 2m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a644dcf56d0c43a4b7441c9cdace9caf : RETRY_LIMIT in 2m 41s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6c331ee5cf1e4b61b0a0ad3b9e744325 : RETRY_LIMIT in 2m 58s\n- swift-probetests-centos-10-stream https://zuul.opendev.org/t/openstack/build/53ca654b93bc4fdf81dd4854858eb958 : RETRY_LIMIT in 2m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c37d200d959d45bb8023ef177b22a5b4 : SUCCESS in 52m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f309270d62648cea1556cade7e2b634 : SUCCESS in 42m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c85ffca8f7cb41519dc2e7a5f3b6889f : SUCCESS in 27m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/076dd7aedead47a28c658a9a4e021240 : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ddaddae7d40e469e943eca24e7bde6cc : SUCCESS in 9m 14s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8a1544fe92c4ebcae4967da7f8bcee5 : SUCCESS in 57m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b36f42a4c8a24c438ad6280b9ef236ac : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":3},{"id":"f27d2a2a2dad2901899ee23233415851548d596a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 01:42:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dda70ea3c5b3482dbaecdd43988f0791\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cccb85493fb14086821e9d7b9cbc07d9 : SUCCESS in 1h 19m 55s","accounts_in_message":[],"_revision_number":3},{"id":"21967f937710d297f0c15dbc804e81fe98edca77","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-08-29 08:00:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"87cca0fe2469d37c29ec7f3df1d7daf909f52558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:08:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5a53d3bc82b409a9f4b65aed30c767c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71da8b879df0482a82152b2e17128028 : SUCCESS in 5m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/16e978ee7a33436babc508259fcccad5 : SUCCESS in 29m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7082d89c98494fa2bb8e961e5eac6c17 : SUCCESS in 33m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c27b350345f47c9b008a2bde87fd9ac : SUCCESS in 52m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/231fefd8ab674305b4ddc997134f1c09 : SUCCESS in 1h 03m 33s\n- swift-tox-func-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/b14c51a3becb40229fb14738add0fa6f : RETRY_LIMIT in 2m 51s (non-voting)\n- swift-tox-func-encryption-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/96e1be4614f84fbca70fb520bbb67555 : RETRY_LIMIT in 1m 27s (non-voting)\n- swift-tox-func-ec-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/3fb1b6669bfb4acaa2418160207c6b79 : RETRY_LIMIT in 1m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c72757c10eaa40a58c168151d4c43b0d : SUCCESS in 7m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9c27ba435db64c4ebb6a29f251d3c92b : SUCCESS in 24m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/705fc01b35334c7f9fb95ac29391a163 : SUCCESS in 22m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2e0b4ca3abe5407abbad77c2ae1da1d8 : SUCCESS in 28m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/63f9e9dbf58d464a88e08399b8cc25af : SUCCESS in 22m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/30221add417a483fa0a8673e79f54570 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fe78f726f3344cff97fce0fb57fb9557 : SUCCESS in 13m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3d9a643550b34d109c731464b0f1ec70 : SUCCESS in 9m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/735800d13eef4472ae88d3de966c52dc : RETRY_LIMIT in 2m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fec8e498a0ba4cc0a9cce002d03c834b : RETRY_LIMIT in 2m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0cd3abe35d2348a28b5590db999cfb51 : RETRY_LIMIT in 2m 38s\n- swift-probetests-centos-10-stream https://zuul.opendev.org/t/openstack/build/52563d0e17124dc0b66766831dca5313 : RETRY_LIMIT in 2m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/894510b43c0742d398df83798c214de7 : SUCCESS in 51m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea0c11b4131544e88ee43bdfed025cae : SUCCESS in 41m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1352df1ad77147ba8c7b38e9e6fc8c6a : SUCCESS in 26m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8767876b8a784d2ab5886ee78c1be89e : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/32e34e3aec514c85a2efe445c680bf76 : SUCCESS in 11m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47e64bae34eb4724bd81e80dd216519d : SUCCESS in 48m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49a8a8c79e0d43d8b80f264f7a7b3b03 : SUCCESS in 52m 36s","accounts_in_message":[],"_revision_number":3},{"id":"99bd85da7494b38bcdd84718d4ae48ce34d691fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 11:46:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68f0f0b2c8404a068ff9c8e04a2f2d2a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f1e4b33eded240f8bf8a8066d4a8f345 : SUCCESS in 1h 12m 03s","accounts_in_message":[],"_revision_number":3},{"id":"5138f39f79e46ad516266b39364fa5c621013060","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-11 21:22:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1c5009eac931b9837b9e49d065100bd48cdf0484","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 21:53:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b7fda5dbf5e4cf2a32c38c7dc04cb59\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32014e783a304168a1b5c602eed8083f : SUCCESS in 28m 27s","accounts_in_message":[],"_revision_number":3},{"id":"70817ddd02a75bcee8ee357a7ac3fe0691ffca96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 22:53:55.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d557de6131c14bd6a95f08c045ecbb76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd0e940af07546ff8a373af19fb56f52 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/82d074a1fa3d4953baffc804378c3ec3 : SUCCESS in 34m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aac2723f814e4e7b812c7a6a1475dad5 : SUCCESS in 1h 02m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8a39ad1f0f254d98a8b62e3dcf1f2f31 : SUCCESS in 58m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9aaad9113d334b35b72aed55ae53152e : SUCCESS in 1h 30m 27s\n- swift-tox-func-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/f9e390dfad4541898eed478917856370 : SUCCESS in 12m 56s (non-voting)\n- swift-tox-func-encryption-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/a14619e414d841409dc7166b2b91e445 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-func-ec-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/8259a3eb9eb046c6aefe9d4afd264952 : SUCCESS in 14m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/943f9302d0b246f1b1b3843137a770e0 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8d53473da0004450a3eace8eaffa45c8 : SUCCESS in 24m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/33a4aea8e8c14006a34986fea2aaafdc : SUCCESS in 22m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9488a9cf61de428fb136842126881601 : SUCCESS in 28m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4d1ca55efc8140dda7475272679ac843 : SUCCESS in 14m 36s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2038f2d96d8540acaf6db91d97b70054 : SUCCESS in 20m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5e40971d268c42ab9b382baf5445ac8a : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6ba99e85e5f24219b9d854608ffa0eb5 : SUCCESS in 14m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1366b5c5913d4b33af1632f8e625129c : SUCCESS in 16m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7334b489bf344107960c6ef77780f4b7 : RETRY_LIMIT in 6m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ba48dde536f64f098c11946445fad7b7 : FAILURE in 7m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/300345a6512a412d9502ee107cb6446e : SUCCESS in 11m 19s\n- swift-probetests-centos-10-stream https://zuul.opendev.org/t/openstack/build/b9d416969d2a4c608414570827468078 : SUCCESS in 46m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a25d3e7256f47358c6972e0f413ebf8 : SUCCESS in 35m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b980b5cd075c4f73bac60eb037e2a4aa : SUCCESS in 50m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adc50088c3e94c62aaaf4b820fd51d01 : SUCCESS in 27m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe2917a0da724af3b51af672b57300b1 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21a7861a0d06460eabcca608346f5f94 : SUCCESS in 17m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c18aa5e6a3d4f3bb3693981b11c18f7 : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":3},{"id":"8c5ac4feb95f9603d6c96a9b1ef75ad7b79531e4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-12 16:46:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8d3be07a78c4bbdefee676f5792dc594023aca25","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-13 03:45:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"084d66d8e0068541f76041e9d800a41584e297fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-09 18:18:11.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c3983c1d681dadb172222eefed5276b6c0d172b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 20:54:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05da8f62ffd44eae8cde0202b9150417\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21adb7771c304f128a866e72d1bb764f : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56c85b56d806454a9d031a0d6fb406b0 : SUCCESS in 4m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/2867dfa7054540bf83d328ffbe16e579 : SUCCESS in 1h 01m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ef80b22179854b91ae88a2e35b2d9911 : SUCCESS in 1h 04m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c54728b04ac643eeae5a20a4c3ff648f : SUCCESS in 56m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c78af7d49dc4009899eaa8d978e5e72 : SUCCESS in 1h 03m 29s\n- swift-tox-func-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/89ec81ef6ee647f89fa50256e47a4d05 : SUCCESS in 22m 30s (non-voting)\n- swift-tox-func-encryption-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/681e10ae3eac4acba3d89e9f69dc227a : SUCCESS in 30m 27s (non-voting)\n- swift-tox-func-ec-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/cbcc9cf1f13f46819f31b1038c7d08ee : SUCCESS in 23m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2a0f137d03f434597c300f72457c03d : SUCCESS in 7m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/098b894c7daa443bb9c5732948c8bdbd : SUCCESS in 26m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/59d7d222ac324faebd394db2800d1246 : SUCCESS in 22m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b238cbf9adbe4b9fb413030942682574 : SUCCESS in 32m 28s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/47d3f66d38e646dba91dd51a4b644a4c : SUCCESS in 17m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/53ba44fe38dc45daa8b4008d59e95593 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4720bf30c5f84a109443e3c664981dcd : SUCCESS in 16m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1dea23b266c6426eaac5f372cb34332a : SUCCESS in 16m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6680ebcbb5ad42acac4a6b8071fab80a : RETRY_LIMIT in 6m 22s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/932c22a5a4d24e11aef6955432be1f43 : FAILURE in 10m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/086b66dac1b340e4a1c81f8095a53fee : SUCCESS in 15m 00s\n- swift-probetests-centos-10-stream https://zuul.opendev.org/t/openstack/build/17fdec463ae942e7a03bfdeb5c05b7d8 : TIMED_OUT in 2h 01m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dce900eb6bde4e9a8c603510a64e42d3 : SUCCESS in 31m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c27e0b78304c45e39ddf396833a50aa7 : SUCCESS in 41m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5812060a89b8456e920dab1133f1ce34 : SUCCESS in 30m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1586d51d9b65447f834f92bb1c9c2106 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4c892fc53a8145cf9632f09f0e5545a6 : SUCCESS in 21m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1460f6cc469c41d98fa87524db58d29d : SUCCESS in 1h 02m 07s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"9bf108278aedc11fdc3635a500de69cdb317aeef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 21:43:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/732f4aa382044f77baec5624dc25c713\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ebf0915b1b0c4b5f839f735a25351947 : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":4},{"id":"3a6b843d8507c22ea3f21739e90d723b92fd3e10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-09 21:47:02.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"fb76d54e4423179895b493acff048d8ffcbf269c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 23:02:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63482b462ab44e369c29f2c29a091c14\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/71ea8189d8144fffb27f3e362c1489d0 : SUCCESS in 1h 11m 42s","accounts_in_message":[],"_revision_number":5},{"id":"9c6487b41b54d63c2286a3dce12353db430ae87d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 23:30:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/710da0ec23634fbaa9dc54884278435b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d4e3a9ea8c43a48115d10344ea4223 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/203b69ed7909466c89577eb888a47e65 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/3129f134a44b4b9c993e23fcb3ca8ada : SUCCESS in 33m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fdc1295b789c421d91e9b710d20a51ff : SUCCESS in 45m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/76c5aa6519884178a30e92094140b3c5 : SUCCESS in 57m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d557215102484f9585ab09690c5aedaf : SUCCESS in 52m 13s\n- swift-tox-func-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/1aa3aa647b1e49068f8f3c2900b1a51c : SUCCESS in 21m 09s (non-voting)\n- swift-tox-func-encryption-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/126f085d8a2f49458f16312384243464 : SUCCESS in 33m 05s (non-voting)\n- swift-tox-func-ec-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/eb60b513745d474d8cb365cba8a8388e : SUCCESS in 26m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e0c5dea44054b5caf2e919f3ca2b305 : SUCCESS in 8m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0f6f63ff46894f1aa6f370665d3b99b6 : SUCCESS in 24m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bd89c066d7e44bcbbe05259d4d8c05c7 : SUCCESS in 23m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ddeec677fdd34bb7b9c566c6c9226765 : SUCCESS in 29m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2fc8b8621c06449f922fb14f29400a3e : SUCCESS in 16m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2894a4d0ebec4713bc9d35d74ccc06e1 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d14c35a509c642e8ac9c371036def6d9 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/142129b088204e79b670e1301d40fe6a : SUCCESS in 17m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0fd5ee411d9a4955b891357ed4bfc501 : RETRY_LIMIT in 3m 46s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e116b87962b044189633adff130096ef : FAILURE in 13m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fb2cec7277224517950f07f892772d49 : SUCCESS in 10m 13s\n- swift-probetests-centos-10-stream https://zuul.opendev.org/t/openstack/build/ee42a521dd884d668f3634d2974be741 : SUCCESS in 1h 25m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1620affd14ba44a9927bd1022f22315b : SUCCESS in 1h 00m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b37765068b834b65ae4a1c829e0f7f20 : SUCCESS in 57m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20b2ec410c2642daa32ba07ca3acd2aa : SUCCESS in 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2a9f6a5d21242fcb4cf8195b3470dd2 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a69169674e314c9ca1a0af7d7c18233c : SUCCESS in 18m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3302067fb7e04f879fe53dcea36f83ff : SUCCESS in 56m 40s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"dc9fd77fce5c13160a635c91616a8d2379446c2b","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-10 15:35:16.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c127d8534a8ba7512c4fc0dbcb3b3032486d4b67","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-10 16:26:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e1bcea27ed23d67f3ce0961e28d7f13242c857f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-10 16:28:31.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"e431924ed1df0d31b20d3460fc6f6f3b1aa817d1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-10 16:28:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bbcbae5e71d3877cb8d53e208b254d43d1743c01","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 17:32:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75d550d7f1db490b81ea0c4566e514fa\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/374decd8a88e48d08ca87ca48190df03 : SUCCESS in 53m 19s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/7a0b54c770564a6280a6ccfb90ba7e47 : SUCCESS in 40m 56s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/04d627cedb164964a3409c3187c483cf : SUCCESS in 40m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/619011e5d61c476fa545a1391860eba3 : SUCCESS in 7m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c4379f97b7914ed9a605502c5dcc6566 : SUCCESS in 8m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/005511a16de6476e943c66c45142b943 : SUCCESS in 24m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/038b929c496143fab4aaa7d62a41bf0b : SUCCESS in 19m 32s\n- swift-tox-func-py312-centos-10-stream https://zuul.opendev.org/t/openstack/build/86c9c97bd5df476ebd07e383bbf6f2d1 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py312-centos-10-stream https://zuul.opendev.org/t/openstack/build/18ec7799d6054c16af623ca12e3a484a : SUCCESS in 17m 03s\n- swift-tox-func-ec-py312-centos-10-stream https://zuul.opendev.org/t/openstack/build/225b22d3d99448c4a0ca57bee9845fe2 : SUCCESS in 17m 59s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/7588b147403c4a8cb42ae1799d518f98 : SUCCESS in 10m 49s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/a77f7a9b01ee40eab43c95f9df3e8ab4 : SUCCESS in 8m 58s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/304d0c752b574913be2aa4dcd83bdc12 : SUCCESS in 21m 25s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/4f5bef7d153d4488b1732994966d121d : SUCCESS in 28m 28s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/e1a598b122864420afa48d160aa87c02 : SUCCESS in 18m 53s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/f005e511e9784b8894632e55e0ea6abf : SUCCESS in 9m 15s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/d53c499eb2334976a273c54f4ad79308 : SUCCESS in 18m 37s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/6ec46922ad5f4f3484fe09407535f539 : SUCCESS in 19m 06s\n- swift-multinode-rolling-upgrade-epoxy https://zuul.opendev.org/t/openstack/build/26ec4e621bf345d4bc0a1c82bf03143e : SUCCESS in 19m 04s\n- swift-multinode-rolling-upgrade-flamingo https://zuul.opendev.org/t/openstack/build/e3a519d77c1e45249acaee72e0f04e46 : SUCCESS in 10m 01s\n- swift-multinode-rolling-upgrade-gazpacho https://zuul.opendev.org/t/openstack/build/a1948f0813b84343b3acf5636a3ae63c : SUCCESS in 9m 56s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/456145fcf96b4684a0adcb8a1c60470c : SUCCESS in 21m 06s","accounts_in_message":[],"_revision_number":6},{"id":"32a7feedbeacb72199f7369abce42202adfd8842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 17:51:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed214346df6546749739eb40f2f1b650\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ed9f0ab40804f64b74061eed9442932 : SUCCESS in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8930a7d952384ef9a216c2660af5cd83 : SUCCESS in 7m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/6ceb06d1cfac41b7ad4789924a2ad9df : SUCCESS in 31m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/658e9231c5be4ddba53cf55c4902fb60 : SUCCESS in 1h 13m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1515d1d73caf447fa84134ab56c96adf : SUCCESS in 57m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d71dcd1277144bfb36c5cecc99e7645 : SUCCESS in 1h 03m 08s\n- swift-tox-func-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/47af37ea967f4d3697c14b543defc76e : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/f6715285e5f94d189e7ef7a2cc4347da : SUCCESS in 32m 52s (non-voting)\n- swift-tox-func-ec-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/ea6ce805dec04713b58d317260ddf22f : SUCCESS in 33m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77c46ba0754e4aab841bb051a0e2c31d : SUCCESS in 4m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7dcb66ee8495448183c50100d3a00bbb : SUCCESS in 11m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ceef514a33cd4caeaa82c9f7a33c7b7a : SUCCESS in 22m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f43dd746338b495eb111d102e2cdfb81 : SUCCESS in 13m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/131726b7b5b64cf5bc37bc8a95b5ac33 : SUCCESS in 21m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0e86440811b34f2dbeed5bc2fb0acf0b : SUCCESS in 9m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ef25cf8fdae645d4934ac19a3a0fa9e9 : SUCCESS in 12m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cc304f8494a5440a9b8f2e3de85f6795 : SUCCESS in 9m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/40969914c7c74a5f9fea25c5c3f55a8f : RETRY_LIMIT in 5m 41s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eedd3168a0cb45db9b848bb1695095db : FAILURE in 12m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/225bdbed5e35428db59adb15d5e5bbb8 : SUCCESS in 9m 28s\n- swift-probetests-centos-10-stream https://zuul.opendev.org/t/openstack/build/3dec6e4ffd3b490a9295a4b4cbe2a4e3 : SUCCESS in 1h 20m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/beefb98daca9446993a2de09e66ec4bf : SUCCESS in 25m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7e5c27308bd24eaea6bc9547a3d95ec3 : SUCCESS in 56m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0744246e5a1e499e97a77d5e728d7931 : SUCCESS in 29m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/333bdb14a28c4955827839b3f822d3eb : SUCCESS in 2m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/57fea7b4636e4dddb3ddbce61dccd1e3 : SUCCESS in 11m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba0ffe1d13ee474e9507fa88d4e03519 : SUCCESS in 1h 00m 21s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"99f8c06c0667d0097eee85b974fa32c4de947db9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 19:40:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bb317c4e04f4c7ab57e343567327795\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2b5efa4cac7a46ad810bbf40709d9af1 : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":6},{"id":"2e3e5e8d3777c4ed027b64a49bb773226e89b14a","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-13 06:55:49.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"71992d54dca7ae77cdc3d601a6fc260912823b3e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-13 15:29:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d302b5fd63d85619dfb81a3feea4ad3e19972ecb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-13 17:05:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d4f9bd89a6788e4b0ad0c62b0c4d3d0361d45c96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 18:00:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cdec7bcea96458993f5555a5342d70a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/738cc64e28354fc0bd814ec2acf83ebd : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":6},{"id":"e5c73888e02b5e81746a1ed186ef977d632ae152","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 18:32:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d75b0e81b42a43ef9add17411027633b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d8854137f74da4919dee03665e6ae1 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d245d30df42643d4a8d935026ff227b9 : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/de096f8b7d314e86ba9e87462099c7d6 : SUCCESS in 1h 01m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1ea105d517d543719b840b72edf76b70 : SUCCESS in 48m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4144364003cd4946ae52da611b9dc9e2 : SUCCESS in 34m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc7867626af5464a956965a66d624240 : SUCCESS in 33m 19s\n- swift-tox-func-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/f71cc880207440b1b02ce5d7ef69a5fd : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-encryption-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/e0a515a7f55d4f3499b2c49bea982720 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-ec-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/7502861a807b4d7db4cacbc2e7fef6fd : SUCCESS in 22m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/765c54abd37e487e807a3816395b36b1 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ca9d79fef9f44b318db3a558fe2d07e9 : SUCCESS in 10m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d812ece67a2b42b99838a4800712dcf8 : SUCCESS in 16m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0a3c37b01009440f9b9123dd01eaa9d8 : SUCCESS in 31m 48s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/08f4778c09344353a90e6ec47fed5ecc : SUCCESS in 11m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8d824421c404432fb6319dd28a2bfe06 : SUCCESS in 9m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8e4eafa939c94a6f90ee2d0ba5600661 : SUCCESS in 8m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1b2b5af20c4548f8b6d23c07b469c54a : SUCCESS in 9m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05eb90372c054ebbb79291f2f7a8253d : RETRY_LIMIT in 4m 04s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ec9e5fdce2674df4b2c20ea07cce2402 : FAILURE in 1h 14m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3886c242193d4749b6b34b7c88f0a788 : SUCCESS in 10m 45s\n- swift-probetests-centos-10-stream https://zuul.opendev.org/t/openstack/build/46c042c9aeeb4872b7653a6ca23989fd : SUCCESS in 1h 19m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8084ee9a08964f3ca69794372ec38e25 : SUCCESS in 36m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/66ad639252d84978b4c851f4d748ca0c : SUCCESS in 22m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/810fc309e52c4a34b5b8cb0a0bc0e5f2 : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b60e5cc35624f9f8625e5225cbec7cb : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76b4c0d9c70b45efbf3832c80839759f : SUCCESS in 11m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a7732f5938e41eaa43386165003cfde : SUCCESS in 1h 00m 35s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"e6a8160835c33c051ef1f3c5b71f8c6939047b58","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-13 21:10:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1d71797797019671f58f3090eb01c88370c5b49d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-17 16:51:30.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"85b04969a5c94265ac1702b8b15c51aae59de029","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-17 16:53:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ee8e787b0fcb11c61f802eb66f0f3ad7d1851c66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 17:25:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01af3435dfff47559543e9e250d2cab7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b5f22f8c321a4544b10fdbde6c091106 : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":7},{"id":"fdfa2d5c6e99579c2dbb6bab8041aa534085ebfe","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 17:42:22.000000000","message":"Patch Set 7:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e2dd540dde8498a830954cf7b00ae56\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cda13933ddf941ffaa307cf453205e04 : SUCCESS in 40m 28s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/48c6e99fe4354851900064424dbc5060 : SUCCESS in 34m 05s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/03c01deff167421fa553cb1bfa1420ab : SUCCESS in 23m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/26bb01f839b94c7c8f8c12acaf47041b : SUCCESS in 16m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/280d3f44e6374df49d0132ae17fef12c : SUCCESS in 15m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ae817d4bd71245a9889354fcad80668d : SUCCESS in 9m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3ffe19877b4744f6a2681677996af2ad : SUCCESS in 36m 41s\n- swift-tox-func-py312-centos-10-stream https://zuul.opendev.org/t/openstack/build/9f699b3f6fd544ef92cf86ebdb422b5b : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py312-centos-10-stream https://zuul.opendev.org/t/openstack/build/2c0a692edf8046ff8222d0ef1d7373a9 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py312-centos-10-stream https://zuul.opendev.org/t/openstack/build/685ebedbf95c4ad8b54b0ce1d34547fd : SUCCESS in 8m 47s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/680b9e71f36d44edb9449bfe83e2a785 : SUCCESS in 9m 00s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/fceffa15c2d44a3da2a56839ad2e73f4 : SUCCESS in 20m 28s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/3428eeb8b9184120bd01cab6b66964bf : SUCCESS in 18m 55s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/5294dbf4fdf54f2e89442bd1528f3c96 : SUCCESS in 16m 33s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/e3851e79db4844ac92545b7d0c64b470 : SUCCESS in 26m 58s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/5dbdab9fa8464bfa9bc83f66dbfc7f11 : SUCCESS in 16m 19s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/680c0733cc9a4915bac144530a1fcc58 : SUCCESS in 19m 07s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/1e9de2d353954788b596c6d01d4a7e9d : FAILURE in 21m 38s\n- swift-multinode-rolling-upgrade-epoxy https://zuul.opendev.org/t/openstack/build/60fdd54c114946c3b7471e9ddd611da7 : SUCCESS in 17m 42s\n- swift-multinode-rolling-upgrade-flamingo https://zuul.opendev.org/t/openstack/build/1a461770e4ed4007a4e1a456203314a8 : SUCCESS in 21m 40s\n- swift-multinode-rolling-upgrade-gazpacho https://zuul.opendev.org/t/openstack/build/58e2fbdfe52144cbb9717a876587f12a : SUCCESS in 26m 34s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/aa58f8561bce4a45b88f332589b1133c : SUCCESS in 18m 10s","accounts_in_message":[],"_revision_number":7},{"id":"f4b6554eeaf2e7cd027e4d57372eea34e65c8354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 18:11:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/867ffd594bad47f1b24bbfe7a050352b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f9e62f691134a0fa943fc8be8e51652 : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa59c10f264b4cc6a7774d76474a8cbe : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/ca38ab5b534247b1afba90224ca65016 : SUCCESS in 1h 10m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b43ea9880c8404db38654d1af2f16bd : SUCCESS in 1h 00m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e0c1f8b773d646878834f1ea8489ac51 : SUCCESS in 51m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bbac72d697946bb87b43468c3bdcab5 : SUCCESS in 58m 13s\n- swift-tox-func-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/45a15afb2c1543b7b38dd6c74a5ddc93 : SUCCESS in 14m 54s (non-voting)\n- swift-tox-func-encryption-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/3bc4da9912c04808a81408bb0822cba1 : SUCCESS in 16m 25s (non-voting)\n- swift-tox-func-ec-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/aa1bf6a88d134d2cab11d9ffff473828 : SUCCESS in 16m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d8818df1f9348468ffcb1b0d989f403 : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dc254864c81a48cf8bc2536e34eba87c : SUCCESS in 25m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/085d1d9fde1f415c898ce82dc526ad58 : SUCCESS in 9m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/42ef78979e514f6ab6bfae943e33c3de : SUCCESS in 37m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/707b94cb479f4ff68c5ba8e04970a31f : SUCCESS in 21m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ed7a2fcadea246bb817d879503c7387c : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7f032dedc1e54fdabb24b1908490e7b3 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5d483998169d407580120bac68d1928b : SUCCESS in 17m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/95778ee6e62f40329bb4ca852ffd11d3 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/46c2b07953bf442eb10fcc38b1413644 : SUCCESS in 10m 39s\n- swift-probetests-centos-10-stream https://zuul.opendev.org/t/openstack/build/ebaa62ca45ca45cc9d24f1c09a0b9288 : SUCCESS in 35m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f841929c29b74276825a2abf9ddd6a8e : SUCCESS in 49m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fca1d1ab7a874450bbf24872b7099918 : SUCCESS in 59m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a619e2c12564b1f873e12532aeb946b : SUCCESS in 29m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dd755d279b24d53a20d7114fa8c78d1 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3196e9b5d2bb4ad4a54a5416429cbb25 : SUCCESS in 19m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b75aea401a374cc4b98c34351c5665b6 : SUCCESS in 52m 02s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"830ef25bb040e371157e1bd4eaae851462bd7e41","revisions":{"93960a2b1b4c91aaee4528abbabdaf23264e8c74":{"kind":"REWORK","_number":1,"created":"2025-08-28 20:38:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/958812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/958812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/958812/1"}}},"commit":{"parents":[{"commit":"9e8fc130a74494b130821f140a920f497d04322c","subject":"CI: Remove a bunch of unnecessary bindep profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e8fc130a74494b130821f140a920f497d04322c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 20:38:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 20:38:10.000000000","tz":-420},"subject":"CI: Move (most) CentOS 9 Stream jobs to 10","message":"CI: Move (most) CentOS 9 Stream jobs to 10\n\nLeave the rolling upgrade jobs, at least for now, as the expectation\nshould be that you\u0027re coming from an older system.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Iad899f0a1e7dd665d819b457a6545e391b4a9da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93960a2b1b4c91aaee4528abbabdaf23264e8c74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93960a2b1b4c91aaee4528abbabdaf23264e8c74"}]},"branch":"refs/heads/master"},"d3617883cb3d60145acd3a66a8039d954ccfa314":{"kind":"REWORK","_number":2,"created":"2025-08-28 21:29:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/958812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/958812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/958812/2"}}},"commit":{"parents":[{"commit":"21325988dfe7d63d1f3ef78b9af8a2297f787e1f","subject":"CI: Remove a bunch of unnecessary bindep profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 20:38:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 21:15:31.000000000","tz":-420},"subject":"CI: Move (most) CentOS 9 Stream jobs to 10","message":"CI: Move (most) CentOS 9 Stream jobs to 10\n\nLeave the rolling upgrade jobs, at least for now, as the expectation\nshould be that you\u0027re coming from an older system.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Iad899f0a1e7dd665d819b457a6545e391b4a9da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3617883cb3d60145acd3a66a8039d954ccfa314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3617883cb3d60145acd3a66a8039d954ccfa314"}]},"branch":"refs/heads/master"},"92fb3ab4167b0527c405bbb2f6cacbac1645d475":{"kind":"REWORK","_number":3,"created":"2025-08-28 21:33:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/958812/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/958812/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/958812/3"}}},"commit":{"parents":[{"commit":"21325988dfe7d63d1f3ef78b9af8a2297f787e1f","subject":"CI: Remove a bunch of unnecessary bindep profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 20:38:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 21:33:19.000000000","tz":-420},"subject":"CI: Move (most) CentOS 9 Stream jobs to 10","message":"CI: Move (most) CentOS 9 Stream jobs to 10\n\nLeave the rolling upgrade jobs, at least for now, as the expectation\nshould be that you\u0027re coming from an older system.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Iad899f0a1e7dd665d819b457a6545e391b4a9da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92fb3ab4167b0527c405bbb2f6cacbac1645d475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92fb3ab4167b0527c405bbb2f6cacbac1645d475"}]},"branch":"refs/heads/master"},"bea9ba99507fcf1caf75675655382364af1639ea":{"kind":"REWORK","_number":4,"created":"2026-07-09 18:18:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/958812/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/958812/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/958812/4"}}},"commit":{"parents":[{"commit":"8be4f02ef61a09b044965134b9b7f95a5eb88bc1","subject":"crypto: Fix listings when pipeline was misconfigured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8be4f02ef61a09b044965134b9b7f95a5eb88bc1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 20:38:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-09 18:17:23.000000000","tz":-420},"subject":"CI: Move (most) CentOS 9 Stream jobs to 10","message":"CI: Move (most) CentOS 9 Stream jobs to 10\n\nLeave the rolling upgrade jobs, at least for now, as the expectation\nshould be that you\u0027re coming from an older system.\n\nAlso leave the CORS job on CentOS 9 until we can figure out how to\nget a virtual frame buffer working for selenium on CentOS 10.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Iad899f0a1e7dd665d819b457a6545e391b4a9da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bea9ba99507fcf1caf75675655382364af1639ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bea9ba99507fcf1caf75675655382364af1639ea"}]},"branch":"refs/heads/master"},"caeb1a3de8e00480b748feb3c13c36881821c2da":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-07-09 21:47:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/958812/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/958812/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/958812/5"}}},"commit":{"parents":[{"commit":"f18d0e76bab7661fc9c2847ef65b9e1a10257e4e","subject":"Quiet eventlet deprecation warnings.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f18d0e76bab7661fc9c2847ef65b9e1a10257e4e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 20:38:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-09 21:46:55.000000000","tz":-420},"subject":"CI: Move (most) CentOS 9 Stream jobs to 10","message":"CI: Move (most) CentOS 9 Stream jobs to 10\n\nLeave the rolling upgrade jobs, at least for now, as the expectation\nshould be that you\u0027re coming from an older system.\n\nAlso leave the CORS job on CentOS 9 until we can figure out how to\nget a virtual frame buffer working for selenium on CentOS 10.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Iad899f0a1e7dd665d819b457a6545e391b4a9da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/caeb1a3de8e00480b748feb3c13c36881821c2da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/caeb1a3de8e00480b748feb3c13c36881821c2da"}]},"branch":"refs/heads/master"},"db611a59dee69d268c037ad3eca7cdc67f5ae982":{"kind":"REWORK","_number":6,"created":"2026-07-10 16:28:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/958812/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/958812/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/958812/6"}}},"commit":{"parents":[{"commit":"f18d0e76bab7661fc9c2847ef65b9e1a10257e4e","subject":"Quiet eventlet deprecation warnings.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f18d0e76bab7661fc9c2847ef65b9e1a10257e4e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 20:38:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-10 16:28:25.000000000","tz":-420},"subject":"CI: Move (most) CentOS 9 Stream jobs to 10","message":"CI: Move (most) CentOS 9 Stream jobs to 10\n\nLeave the rolling upgrade jobs, at least for now, as the expectation\nshould be that you\u0027re coming from an older system.\n\nAlso leave the CORS job on CentOS 9 until we can figure out how to\nget a virtual frame buffer working for selenium on CentOS 10.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Iad899f0a1e7dd665d819b457a6545e391b4a9da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db611a59dee69d268c037ad3eca7cdc67f5ae982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db611a59dee69d268c037ad3eca7cdc67f5ae982"}]},"branch":"refs/heads/master"},"830ef25bb040e371157e1bd4eaae851462bd7e41":{"kind":"REWORK","_number":7,"created":"2026-09-17 16:51:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/958812/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/958812/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/958812/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/958812/7"}}},"commit":{"parents":[{"commit":"bd664103928dde998f7b79ec73c572153c8b1b52","subject":"replicator: Continue handoff cleanup after quarantine failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd664103928dde998f7b79ec73c572153c8b1b52"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-28 20:38:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-17 16:50:12.000000000","tz":-420},"subject":"CI: Move (most) CentOS 9 Stream jobs to 10","message":"CI: Move (most) CentOS 9 Stream jobs to 10\n\nLeave the rolling upgrade jobs, at least for now, as the expectation\nshould be that you\u0027re coming from an older system.\n\nAlso leave the CORS job on CentOS 9 until we can figure out how to\nget a virtual frame buffer working for selenium on CentOS 10.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Iad899f0a1e7dd665d819b457a6545e391b4a9da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/830ef25bb040e371157e1bd4eaae851462bd7e41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/830ef25bb040e371157e1bd4eaae851462bd7e41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
