)]}'
{"id":"openstack%2Fswift~958859","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I81c7785a66eab002a56d7eac6d869a04837b8616","project":"openstack/swift","branch":"feature/threaded","topic":"create-threaded","hashtags":[],"change_id":"I81c7785a66eab002a56d7eac6d869a04837b8616","subject":"Update .gitreview for feature/threaded","status":"MERGED","created":"2025-08-29 11:56:06.000000000","updated":"2025-08-30 02:30:59.000000000","submitted":"2025-08-29 23:33:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958859-create-threaded","meta_rev_id":"c76a784352901da83a1cba78e023b5540c2a7b6f","_number":958859,"virtual_id_number":958859,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-29 23:33:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-29 21:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-29 21:21:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-29 12:34:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-29 21:21:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"34fc3faf6d845a31391b19a020c45a710319227f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-08-29 11:56:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"353df2d919482db59cf14e746227260838bc4e48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 12:34:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78e474e3dc6440cf89b5494e1125e330\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c2799bc1a84850935a195fe8a39db6 : SUCCESS in 10m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/31b8eac084a2462a8b6ac28587f3c8f2 : SUCCESS in 7m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/67bd41f455d6450fa8489539b5aaf53b : SUCCESS in 12m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/82bec20e9284462392f70da4c5124e4f : FAILURE in 3m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e5fedea331004d2d8c09429d9e5d054c : SUCCESS in 29m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4fc8a3cf472b4bc8942e00db49212803 : SUCCESS in 21m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6351583be83f43a89633ad904e0d59f7 : SUCCESS in 14m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4ea98bebad1455191240db4ea7c7936 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"14bef6bde62582e08743244140ebd5fef74970e1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-29 21:21:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"376c0c6d592eab31e6bdea4c38fbcdaccebc03ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 21:57:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ac8f00e990345209f03be2065dbffd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de159d57af1847ee8dcd64a973f3705a : SUCCESS in 8m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d4d1f04b07745eb8194706c2de42804 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ca7400ee29814d73acf3dfe53ab09b57 : SUCCESS in 28m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ea6392616e9f49489894849a7a7776fe : SUCCESS in 22m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b92896e74c17426db59c09c349d4db93 : SUCCESS in 20m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b1d7112c9f974b1bb8c1bac09513c55c : SUCCESS in 27m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/392637e61c7d433288015b09d30558a5 : SUCCESS in 28m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6798a8bf41f4e6fb44b89608f1d79f7 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"81aa51690589d51c884f68dbc6a7c8dbdd939a44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 21:58:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f65e20a20f53df4ee8d1b55f941eb38cf1a2e7d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 23:33:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c70eb33ded74380b797097e229e433c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cea7397047c4a648588a27b4574dda4 : SUCCESS in 5m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c92fd010e2fa4e19ac097c202924b5d9 : SUCCESS in 4m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/af340b8c1eb942cabb6e0e7265a19a37 : SUCCESS in 23m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/59d0bb4a7edd471cb3b2c98e00b1f090 : SUCCESS in 16m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/83211d8d921149cb87f69352ddf61a07 : SUCCESS in 11m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ec8624f10c3f41e788541aa90f0663ad : SUCCESS in 13m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8646365272fe41b8aa908227c990047f : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/34077b082b1144349ca0c7b18ffd8c9e : SUCCESS in 13m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/94ed9f12caad41a6a0c56ff8cff5d70b : SUCCESS in 15m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9c2256069b64a2c9e6c5470e47790a4 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d83433827cc7441c80b32ba7662021c5 : SUCCESS in 8m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e76cdecbdb444b47b1ef7b8b271508b2 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/121c9f2b773f44989979d634c99daef5 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d1cf0404c1b6f3041f2c242631ce646601f54f27","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 23:33:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"16e5a3b40a23c5873a1266e2d550288c42c7e5e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 23:36:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a87755f62f54354a23e4a995542554e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73214337a3d64677a84838e50c587536 : SUCCESS in 2m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fb82304d787b4a7ca100c08215835c95 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c76a784352901da83a1cba78e023b5540c2a7b6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 02:30:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17a297ba502e4b12b360d5189922a34c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/52b9c83e469445ee8419a620dadf3144 : SUCCESS in 1h 29m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e0354f67deaff915a485c9ea4c1dffd4fc0bf82","revisions":{"1e0354f67deaff915a485c9ea4c1dffd4fc0bf82":{"kind":"REWORK","_number":1,"created":"2025-08-29 11:56:06.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/59/958859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/958859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/958859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/958859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/958859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/958859/1"}}},"commit":{"parents":[{"commit":"e10c2bafcb8fc80929bce3e064fee5f86a49e718","subject":"Merge \"proxy-logging: create field for access_user_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e10c2bafcb8fc80929bce3e064fee5f86a49e718"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-08-29 11:56:04.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-08-29 11:56:04.000000000","tz":0},"subject":"Update .gitreview for feature/threaded","message":"Update .gitreview for feature/threaded\n\nChange-Id: I81c7785a66eab002a56d7eac6d869a04837b8616\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e0354f67deaff915a485c9ea4c1dffd4fc0bf82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e0354f67deaff915a485c9ea4c1dffd4fc0bf82"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
