)]}'
{"id":"openstack%2Fswift~958860","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I20b2d264405983312ab0a42fd4362785809c892c","project":"openstack/swift","branch":"feature/threaded","topic":"create-threaded","hashtags":[],"change_id":"I20b2d264405983312ab0a42fd4362785809c892c","subject":"Update TOX_CONSTRAINTS_FILE for feature/threaded","status":"ABANDONED","created":"2025-08-29 11:56:08.000000000","updated":"2025-08-29 16:22:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6013509e9cfac7eb14ee0d5d626ab94d954a117f","_number":958860,"virtual_id_number":958860,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-29 13:37:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-29 13:37:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e1c962d0ec7bd468d40e9e85ad1b102b4ac3968","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-08-29 11:56:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56fd1b76595c4ce5a4e0fd49ad702b5920a6158e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 13:37:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56e5797ede504bd880cb55aad005df16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c1122f0652440eaa9ce8734dc9eeb6 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0928d19f5e854da881665205d5360a5e : SUCCESS in 6m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da72f6fe2abb4704af2b59be2eedb3e6 : FAILURE in 6m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/311c2c8f2930424a943c320f96a2732b : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b853e922d30b45e8b730c44bdb822912 : FAILURE in 6m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e56925b2bb14eceba539b0343e06a6f : SUCCESS in 4m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0348208b11914db7aad0244396593b79 : SUCCESS in 11m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6dae3545ef8c447ba14c089a4c7a2ff8 : FAILURE in 3m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/16d661f6d79d4050af540b562007c5ba : SUCCESS in 19m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fab2ee42500d46c28ba8f403d6b5b48e : SUCCESS in 14m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/63d6e9cb59624a08bcfd33723adb87a4 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f09f27795b7a4d88aa0dd8cdc0cbf44b : SUCCESS in 14m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6d53d73591fa454cb8c4ad82e2da60dd : SUCCESS in 12m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d464aae3138446a08ffbf9dbdfbd4121 : SUCCESS in 6m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/766eb020c14f481e8c56a4c617261383 : SUCCESS in 14m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e2829bbaac344fffa1960be8162c0965 : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a73ea76fd01a41469798d8fd9873f57b : SUCCESS in 33m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/639bb017d11b4b9897db6a05ce090e23 : SUCCESS in 27m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fd8d38a4d64cf48984bce43e957c43 : SUCCESS in 2m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8db9fa6bdd574263abcb3875c744669a : TIMED_OUT in 1h 31m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6013509e9cfac7eb14ee0d5d626ab94d954a117f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-29 16:22:54.000000000","message":"Abandoned\n\nThere\u0027s no expectation that the requirements repository will get a feature/threaded branch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eca780a3ab58bd6817331dcbcf691ff749ba107b","revisions":{"eca780a3ab58bd6817331dcbcf691ff749ba107b":{"kind":"REWORK","_number":1,"created":"2025-08-29 11:56:08.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/60/958860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/958860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/958860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/958860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/958860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/958860/1"}}},"commit":{"parents":[{"commit":"1e0354f67deaff915a485c9ea4c1dffd4fc0bf82","subject":"Update .gitreview for feature/threaded","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e0354f67deaff915a485c9ea4c1dffd4fc0bf82"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-08-29 11:56:06.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-08-29 11:56:06.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for feature/threaded","message":"Update TOX_CONSTRAINTS_FILE for feature/threaded\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as feature/threaded branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I20b2d264405983312ab0a42fd4362785809c892c\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eca780a3ab58bd6817331dcbcf691ff749ba107b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eca780a3ab58bd6817331dcbcf691ff749ba107b"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
