)]}'
{"id":"openstack%2Fswift~959192","triplet_id":"openstack%2Fswift~feature%2Fthreaded~Ia47a23dfafe193527506771d351ada806eed69b7","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2025-09-25 19:09:19.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-09-25 18:18:55.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ia47a23dfafe193527506771d351ada806eed69b7","subject":"WIP: Add gunicorn as wsgi server","status":"MERGED","created":"2025-09-02 14:31:56.000000000","updated":"2025-09-25 19:10:14.000000000","submitted":"2025-09-25 19:09:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"959192-eventlet-removal","meta_rev_id":"b1cbf6cd03b3f9315de64c4944ebc06ff504c478","_number":959192,"virtual_id_number":959192,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-25 19:09:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-08 19:33:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-25 18:18:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-02 14:55:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-03 00:33:57.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-09-08 19:33:20.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"adfca0248427cedecf9f412a23c007f86cfc3462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-02 14:31:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82be7395d28fa8cdacd6f990008329573d9683b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 14:55:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/618c3f429b854236b3d2e7f2f58a460f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/160f4b1374c54a5fa34f3989814a32e6 : FAILURE in 7m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f02f7c5b5356460789c93939b2ec10c7 : FAILURE in 7m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f1a3969ecfa4cc4b39b3deddb214249 : FAILURE in 5m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91d58f7c61c7417793f6669d0d12fd41 : FAILURE in 6m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f74349777ce2497d866514a687eaa17b : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f82ac94b967840308ee72284dff7b861 : FAILURE in 8m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a0f4d8fad5a14500894d5175027f357a : FAILURE in 4m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3bec64326e004ae9b7e168401cadd451 : FAILURE in 6m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b89166b955af41bca108070fd57a5009 : FAILURE in 8m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/acf08062860f455caa36229501997596 : FAILURE in 4m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9d7c7904730544f492421a23a5a335fd : FAILURE in 5m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/05ca323f955b4cabbe3c9481e67cbac5 : FAILURE in 4m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62e2068afb124836b97da7ca17d99b2e : FAILURE in 4m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d63cff1b6ba4445a630358b695b9618 : RETRY_LIMIT in 5m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/71bef1b1159f42c4a3233575a769f0f4 : RETRY_LIMIT in 4m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a286d7b067ec4b0abfc90ddb401355b1 : FAILURE in 4m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33bc40e7aa524c0494db80ebc95eceaa : FAILURE in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c7e64b4b7a346589b9cb4f1ad53c1ad : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a52874c3bd1b43bca4f07c7a0a6db7e4 : SUCCESS in 11m 14s","accounts_in_message":[],"_revision_number":1},{"id":"203a8277d4c3820d7880d771f73a86e57532980d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 16:23:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3441ca33e88c49a5bfccc4e3a90a4bda\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/411775fd55dc4c2fabc99f2b696a025f : FAILURE in 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3a3298067710fc1612ed81bbc5df18aa93a0dcbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-02 23:34:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a837ef26732f357295cf285b7b10efc5a943c741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 00:21:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ac1d5a10471441db13c2846fb764196\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4240128ba994858b77573064d4f23d7 : SUCCESS in 7m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad66bed83a0e405d96e2f04b08dff4e9 : SUCCESS in 16m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/142c400c4b8147958a0b4366715d644d : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da394a5c8a3d4890a3ce0932d3acd367 : SUCCESS in 19m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b99163546e734cc19801a4e2e7f63353 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dbe9f1eea59248d2b84a680ec8baa251 : FAILURE in 19m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e3d100fe069a4102bebdfafc8fa842cd : FAILURE in 15m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f1d0c0c86b1c4c74811ccb2894c29a4d : FAILURE in 18m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e409b158ebdf4f529eeee0adc900c3ef : FAILURE in 11m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/73144fe68a974955a1ec1fbad1a42ea4 : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/60f5ace91abe420fac02d74e80063698 : SUCCESS in 14m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2389b9319724457cbb83c20e550727a7 : SUCCESS in 12m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6cdf81c484ad48b68ae17810ecb5242f : SUCCESS in 5m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/03558ae5d7dc4b508daed5a9632d450b : FAILURE in 7m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f9a8dbca241c47d8b28c3afac00d6f65 : FAILURE in 25m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e999773794c34bdf887354e92bdb9d87 : FAILURE in 41m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cf9145f5e2a408fb620696116f6836d : FAILURE in 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c73bf0d123a4980b5577077ed8fe2ba : SUCCESS in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/54648e22256a483a9a29ff511ab02498 : SUCCESS in 19m 06s","accounts_in_message":[],"_revision_number":2},{"id":"b09248ec11e076341a05c438273087bb393f0aa4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 00:30:31.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4be6b406b344a408d70a7b85a0094a3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0acf6c93a6cf4830955dbb6f4f17c2f1 : FAILURE in 49m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1457b87439322f706849ea01ccf991989a440987","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-03 00:33:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0643d91de26c158e60ce687590aa98e33c8a3e86","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-03 05:29:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4022263308cdcae0c62a411be3a5abb3e0a7b42e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-03 10:10:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"45b5088f61a365f65bba439843802bf751191539","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-03 10:13:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cc6021890688c7d2426608f9b7ceab227c99621d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 11:11:10.000000000","message":"Patch Set 3: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62ea26a0cbe24fd89a968601ae9fb6cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea149bd3b624ef896da5c941a8d4639 : SUCCESS in 8m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e91bc0c01b3b4d4db7969388d2142d90 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/550ff226df144ec392069e5758dfa549 : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48eb8eb3191b4939be2e5d1b66d4ab8f : SUCCESS in 17m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c6a77cd0e62542c0929bea1ca6aba1be : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/63f7d58d20ec469c91ac735ededb6ea9 : SUCCESS in 12m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e73c05d4375842a2ab96ed1b7026eaf0 : SUCCESS in 11m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/361253f7afe545b4953e07c8086de44f : SUCCESS in 13m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6f0f901e8e6648b9a5c547ffc2d5a15e : SUCCESS in 24m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ddc4f83b1586441d8261b609724971ab : SUCCESS in 11m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a17fe9cc09e24c0b898ee04f011cff5a : SUCCESS in 12m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d00574ccefde4a59af3e64ebac5ae8de : SUCCESS in 12m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/552b313b81104bbbbd7a75a836437251 : SUCCESS in 5m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e18fb430da9940af9e2e001dd1ce0152 : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dbeeb7b718994fa0a6fb922f1650a2bf : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a211b6f3bdf94e92a05685a7f7e23e7b : SUCCESS in 58m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce321ce7daf44541998e8976010c948a : SUCCESS in 40m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d7b284880041279b5823632eaf38ff : FAILURE in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/245322c0b537433e8c7d3c9b78766f19 : SUCCESS in 19m 16s","accounts_in_message":[],"_revision_number":3},{"id":"4118e717a1fabb44659c531e12d383b40354c81e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 12:50:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6096275630fb4dd3a810300df6f2d93a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e93255c7cb4849b680b9a1a896ea1aa7 : SUCCESS in 1h 07m 15s","accounts_in_message":[],"_revision_number":3},{"id":"2aee58492a839e3679fe9093329f437ee3af9c41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-05 16:39:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"49c938b32e0d5dd1b60243e63d30a478a36cfa29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 17:28:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36e405668ca34546b00176363947f637\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d4197ccbf24d33a0ac2c6fd687d815 : SUCCESS in 6m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45157b9c366749abbc3edf2df7390d81 : SUCCESS in 14m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c00d8402c8ca42f88825c95f1ffe66dc : SUCCESS in 10m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f59324715ec4ef88e8d5380396c6bcd : SUCCESS in 21m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d0e26b48626f42908a12e925fd3d51b7 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bd929e0261c94e5bba53f6ba98a6aa06 : SUCCESS in 25m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e83eeb0c34e460bb92400cd634fcf59 : SUCCESS in 21m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/93843dee0d6046f8a83cae159f421887 : SUCCESS in 20m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6d7ab37f153843cda8fc00dec4d0c065 : SUCCESS in 28m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/25751e0fcfd2462288626f2ed73c94e3 : SUCCESS in 7m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/530f212d20114fba945eeb8bbb9479a1 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e8b3e92d579c4751bca998f472d7c552 : SUCCESS in 15m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/365a071921a74af9a6f14c1f41efd461 : SUCCESS in 4m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d7f36def4274ff08fb1c0ada62d0624 : SUCCESS in 8m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a81c2a1fdbc04cadb1cf5bc8c77345e0 : SUCCESS in 10m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/488f2076d8f741b58f02d7d3b233cf83 : SUCCESS in 48m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/198660e6816b4d408c69098039666e8d : SUCCESS in 22m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7effa01d70e1451b91b1cbcf1bf94939 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/efed80b4315e4eeda45bc6049fe7bd35 : SUCCESS in 18m 48s","accounts_in_message":[],"_revision_number":4},{"id":"454c1f2dc730c22a7f25e1e1bcf957dfe030aa38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 18:49:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c0cfc4ea0534bd29909d465051d7488\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/db04422d13e14ff391bc46a03c0b23e2 : SUCCESS in 1h 14m 50s","accounts_in_message":[],"_revision_number":4},{"id":"9d5a92e9e7fa2e223b0ad99f366312d6da3eb64f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-08 19:33:20.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3b94de41fa7d0fa509fde5a6f727b2c63f989d92","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-09 14:15:15.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0347758fb4bae7afc5c073b8d31d3840041a1bdb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-25 18:18:55.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d1ad8a907131166baf10502a176c4b23cc5a56ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 18:19:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cfb7aaf08f89c1760c0607fad18fa74c71510ecf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 19:09:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e027d7b44fda4568a70558668bd3e309\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3fd3d3ca6e4d01a9bf2d7eee568360 : SUCCESS in 5m 45s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/940d9e89a5034cfe9dd85d3da85d64be : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/805f009b76dc49f999b51d876c6be695 : SUCCESS in 16m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c473b8843f72469f8719d6c0ccb33500 : SUCCESS in 14m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3e82158c87d5487d9fd3d35269f376e8 : SUCCESS in 11m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/76e0b7eba36d459b92b59024c3df0188 : SUCCESS in 11m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/29aff23794b04706be71f7c82b203c51 : SUCCESS in 8m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f241cc522d264d8db661af9b9b1ed12d : SUCCESS in 9m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/168d659d309b4d2782ea3bd8d4182c52 : SUCCESS in 9m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45533763762840dd8581336a9758420c : SUCCESS in 6m 45s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/377881a2edd443878ff17c030d15b9a7 : SUCCESS in 5m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a39ed9bc4efb4ae7ba28cf361bfddc94 : SUCCESS in 45m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/234e12e712304cc1980a0de8c6e8f28a : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ef2a82694774a6e8509f6278e1dd858 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6063146ff41c49e2a7abc084c08ef251 : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":4},{"id":"689b391eb627d516900222bf2ec971a814beccdc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 19:09:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b1cbf6cd03b3f9315de64c4944ebc06ff504c478","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 19:10:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9cfc0e737554c769bfcf69d255eca1b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6dda3806c845149ae9cb3824cbe8e9 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/59169043f92b4a7399c7a15733f2cb14 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ed267339b0eaf6cb6cd60f661e714582b3c1e174","revisions":{"1de6326e87e7603f1c38ab6d6b607fb024ca1dd7":{"kind":"REWORK","_number":1,"created":"2025-09-02 14:31:56.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/92/959192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/959192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/959192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/959192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/959192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/959192/1"}}},"commit":{"parents":[{"commit":"e10c2bafcb8fc80929bce3e064fee5f86a49e718","subject":"Merge \"proxy-logging: create field for access_user_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e10c2bafcb8fc80929bce3e064fee5f86a49e718"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"subject":"WIP: Use gunicorn as wsgi server","message":"WIP: Use gunicorn as wsgi server\n\nSimple replacement using gunicorn as a wsgi server instead of eventlets\nbuiltin server.\n\nHeaders are not capitalized when using gunicorn, thus it was safe to\nremove the corresponding eventlet patching.\n\nThe SwiftHttpProtocol/SwiftHttpProxiedProtocol classes are unused in the\nPOC and likely need to be refactored to avoid regressions.\n\nChange-Id: Ia47a23dfafe193527506771d351ada806eed69b7\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1de6326e87e7603f1c38ab6d6b607fb024ca1dd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1de6326e87e7603f1c38ab6d6b607fb024ca1dd7"}]},"branch":"refs/heads/feature/threaded"},"266c2341ff57db4c80ed4f1059db990c98da3869":{"kind":"REWORK","_number":2,"created":"2025-09-02 23:34:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/959192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/959192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/959192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/959192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/959192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/959192/2"}}},"commit":{"parents":[{"commit":"8f5464fd3744146f6dd6edf5bba6d54d622cc0c3","subject":"Merge \"Update .gitreview for feature/threaded\" into feature/threaded","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f5464fd3744146f6dd6edf5bba6d54d622cc0c3"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-09-02 23:34:48.000000000","tz":-420},"subject":"WIP: Use gunicorn as wsgi server","message":"WIP: Use gunicorn as wsgi server\n\nSimple replacement using gunicorn as a wsgi server instead of eventlets\nbuiltin server.\n\nHeaders are not capitalized when using gunicorn, thus it was safe to\nremove the corresponding eventlet patching.\n\nThe SwiftHttpProtocol/SwiftHttpProxiedProtocol classes are unused in the\nPOC and likely need to be refactored to avoid regressions.\n\nChange-Id: Ia47a23dfafe193527506771d351ada806eed69b7\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/266c2341ff57db4c80ed4f1059db990c98da3869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/266c2341ff57db4c80ed4f1059db990c98da3869"}]},"branch":"refs/heads/feature/threaded"},"987b90fe6ad7e211d70189bca76fbe3eff9e24f3":{"kind":"REWORK","_number":3,"created":"2025-09-03 10:10:21.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/92/959192/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/959192/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/959192/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/959192/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/959192/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/959192/3"}}},"commit":{"parents":[{"commit":"8f5464fd3744146f6dd6edf5bba6d54d622cc0c3","subject":"Merge \"Update .gitreview for feature/threaded\" into feature/threaded","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f5464fd3744146f6dd6edf5bba6d54d622cc0c3"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-03 10:08:51.000000000","tz":120},"subject":"WIP: Add gunicorn as wsgi server","message":"WIP: Add gunicorn as wsgi server\n\nSimple replacement using gunicorn as a wsgi server instead of eventlets\nbuiltin server. This commit only adds a basic gunicorn wsgi server which\ncan be used if the environment setting \"DISABLE_EVENTLET\u003dtrue\" is set.\n\nEventlet dependencies are not removed with this commit, and in fact both\ngunicorn and eventlet are running together (though not performant).\n\nChange-Id: Ia47a23dfafe193527506771d351ada806eed69b7\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/987b90fe6ad7e211d70189bca76fbe3eff9e24f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/987b90fe6ad7e211d70189bca76fbe3eff9e24f3"}]},"branch":"refs/heads/feature/threaded"},"ed267339b0eaf6cb6cd60f661e714582b3c1e174":{"kind":"REWORK","_number":4,"created":"2025-09-05 16:39:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/959192/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/959192/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/959192/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/959192/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/959192/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/959192/4"}}},"commit":{"parents":[{"commit":"8f5464fd3744146f6dd6edf5bba6d54d622cc0c3","subject":"Merge \"Update .gitreview for feature/threaded\" into feature/threaded","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f5464fd3744146f6dd6edf5bba6d54d622cc0c3"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-09-05 16:36:26.000000000","tz":-420},"subject":"WIP: Add gunicorn as wsgi server","message":"WIP: Add gunicorn as wsgi server\n\nSimple replacement using gunicorn as a wsgi server instead of eventlets\nbuiltin server. This commit only adds a basic gunicorn wsgi server which\ncan be used if the environment setting \"DISABLE_EVENTLET\u003dtrue\" is set.\n\nEventlet dependencies are not removed with this commit, and in fact both\ngunicorn and eventlet are running together (though not performant).\n\nChange-Id: Ia47a23dfafe193527506771d351ada806eed69b7\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed267339b0eaf6cb6cd60f661e714582b3c1e174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed267339b0eaf6cb6cd60f661e714582b3c1e174"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
