)]}'
{"id":"openstack%2Fswift~959193","triplet_id":"openstack%2Fswift~feature%2Fthreaded~Icc621e2e92ef86b418194241b4cc5396f30ac49e","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-03-06 11:14:39.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-09-10 08:15:54.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"Icc621e2e92ef86b418194241b4cc5396f30ac49e","subject":"Replace eventlet imports with native Python modules","status":"ABANDONED","created":"2025-09-02 14:31:56.000000000","updated":"2026-03-06 11:14:39.000000000","total_comment_count":10,"unresolved_comment_count":9,"has_review_started":true,"meta_rev_id":"5819020bb1e992d3591ad673183479d89189bba3","_number":959193,"virtual_id_number":959193,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-27 12:26:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-02 15:05:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-25 19:19:02.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"da621ca7127cc17a5ee281f3785f429281ce3365","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-02 14:31:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a9a1cc2d91ef879532517c6af5b2c6e296261ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 15:05:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee0c6ee5d81d4bba978f0b970ca50c01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd2cfe8a98349868f8a5973a3dc6623 : FAILURE in 6m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9268405f52b545dbbe0c404acb192572 : FAILURE in 6m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0029a1691a3e4cad9111ce1ed037601a : FAILURE in 4m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbff386a438e40f38bac806230c3e56f : FAILURE in 5m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f498d948eac048239823844cdd8be15e : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c12b5752e2464d75b9c3b4f84a01d1e9 : FAILURE in 8m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/164c97f4fea84e1a83021446f4672841 : FAILURE in 4m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/acf1ac08e91949e084769cec0a0a9b75 : FAILURE in 10m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/29c2b88e564f4ad4a7530f8a272e7b50 : FAILURE in 6m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c02dd167309041309eca3cdca63a9f57 : FAILURE in 4m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6bcc8de4f02045749a9cc7a4bb73389e : FAILURE in 3m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b3cedbbb667847b591ceda2a04011848 : FAILURE in 5m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/96c0749e097a4f24a0fcb3b58cc6f287 : FAILURE in 6m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fac59b1529b0401ab692656bb3d340dc : RETRY_LIMIT in 2m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9c0a51fa75704600801a12370a178f32 : RETRY_LIMIT in 2m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/16dc2123e0524369a3aebe8b6edd95b5 : FAILURE in 5m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aaed1ea96c5c42ae933a264d40e94d43 : FAILURE in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f53270e47f4d929d52746bf93db986 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ead4b7fd9824693977a2993f991cc18 : SUCCESS in 18m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b740291fc42d47b87215675bc513ab2c73e60c2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 17:05:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0125072a73b4842b659bd41244d836b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/43ea14bcda19488980d411e4342c5bbf : FAILURE in 17m 02s","accounts_in_message":[],"_revision_number":1},{"id":"faefab0b42f15d934c60c7ddf8a9adec0e10e860","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-03 10:10:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bf733df9dc46b28bf591d435e5ccb830e61ccc3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 12:17:16.000000000","message":"Patch Set 2: Verified-1\n\n(14 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a95a8ecc35b4269bad95c58f977ceec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09ba7959cacf4f62832ae1b422faf377 : SUCCESS in 10m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0d8219773db43b58bd345a2de95523f : FAILURE in 6m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0083095816ba4089a0f3797521798beb : FAILURE in 7m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4d0a5baa9a34da8b6bf426a2a2c1815 : FAILURE in 6m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/80fdf97b3f564ff698b28d3b4ce5c60e : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/49085af60bce4ab4af55dc1aacb57493 : TIMED_OUT in 1h 30m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0aae2c1f75a343d1af49e5b5e5395028 : TIMED_OUT in 1h 31m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/44e0c973056d4722a6004a006edaffb7 : TIMED_OUT in 1h 30m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d3ca3b9b469143c8bc3ba155fc99cf10 : TIMED_OUT in 1h 30m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ba1c792f0ed64bbf91c5b686526bae1f : FAILURE in 4m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/10c5cddf33314f3492e77c0d147314dc : FAILURE in 4m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d2452e52f9c142b0aef359fe8ab5e42c : FAILURE in 4m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f57d436336e74060a1728e02b376e020 : FAILURE in 6m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49fa46e1257f42dcad1666377cc69dc7 : FAILURE in 6m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f655cd285eeb4382a7838a1ac5ba3cfa : FAILURE in 17m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4dce7ec0782d47cdaa94a7021be548b3 : TIMED_OUT in 2h 00m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fff60dc7e3145cbaa86df05344e7080 : TIMED_OUT in 40m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf3e58a2aab46e6ba4436a8125ee59a : FAILURE in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d8d0ac75593140adad6fc944636a2b0e : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":2},{"id":"ab0a897b0dea0974899b72cacf9f89649b68484a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 15:09:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbc8ab2cffe248579e854aebf0b164f0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/63746429f28348949d2f5da2b2189abd : TIMED_OUT in 3h 00m 55s","accounts_in_message":[],"_revision_number":2},{"id":"d4489bdeb8314a34ee166737034e7d05e5020101","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-05 16:39:07.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"51446d3f8e996efe514504dabae2e5414daa06e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 18:48:13.000000000","message":"Patch Set 3: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eac25c716b44402da5076c6e5c24418d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/111221a1395f4a11b5852484145a4f6e : SUCCESS in 11m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb947144266844fc94860a4c4816bbf2 : FAILURE in 7m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf9961781d1f4f649a0574937d9eeb68 : FAILURE in 6m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/729dc33241cf4376afd9167917dccc42 : FAILURE in 7m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/387d7466d1df4711b9e76078ea05aa50 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3f54188ece3a42c09658ff8f0a6062d9 : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/098c156fa94e41ec98a77c12408742df : TIMED_OUT in 1h 30m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a087c5189d474767965d755682468ac5 : TIMED_OUT in 1h 30m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bbf97494974942238f4da3299f950fef : TIMED_OUT in 1h 30m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5290e068fcea4e57acfa395a74fdf2ff : FAILURE in 3m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6c1bc3373673470da5745700fb399a76 : FAILURE in 3m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fc4ead7b69334a06afa68e4aa878e01d : FAILURE in 4m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6b2d7ba2c024e27869f223bdcdc17fa : FAILURE in 6m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1bdf571e9a9348cb871b2f910cce01aa : FAILURE in 6m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d6aa5108171e4a018b5978a27021fefd : FAILURE in 18m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d3929c029e043e7b7283115a32746a8 : TIMED_OUT in 2h 00m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8870c431f70644c6b762aa00d2bd7e62 : TIMED_OUT in 40m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28ba544657e24c5aa005de74fc1b135a : FAILURE in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7b510aa577cd4581b03969c0d2cb1d0a : SUCCESS in 16m 12s","accounts_in_message":[],"_revision_number":3},{"id":"c517d95cc33edfcbced3a1c5ef14541b50cb859b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 19:59:48.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2bcfbd918304f1f97d4d17feb364b3a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d7b56ceb40394d0580ed5169ec3c305c : TIMED_OUT in 3h 01m 02s","accounts_in_message":[],"_revision_number":3},{"id":"7990a9345263d2df5be26615425365439f14166e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-10 06:10:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7fecb6bbd881ae38a45795d39d50f49bdecd4981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 08:15:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fc7f8ea542f4b17866f2af90288d1e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/261c88494fd3493da906bf40441a9b92 : SUCCESS in 9m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d33c72bb60b48109d811a9df574a7eb : FAILURE in 5m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9541d87bd2449a0af00f68d156b4eff : FAILURE in 5m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b61fbd096a7a4dd6ae143eafb084e043 : FAILURE in 5m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9281282b39524f2ab1778c34d045f639 : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/87eb3a44e24b4db6a0bb1619844bb86d : TIMED_OUT in 1h 30m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a2b9aaa2cbd34d10a79aca9f69ad233b : TIMED_OUT in 1h 30m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2e08fad5f6d641ca8d4fa912ed416f65 : TIMED_OUT in 1h 30m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/581d584b7aef4c37b60fc85168e7c4b1 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d67e31a2e0ad41f4b855d3597b293f5a : FAILURE in 2m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6e0bafff93b248868c7fd6870338912f : FAILURE in 3m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e9b3eb5738424dc0af81ec73a670f820 : FAILURE in 2m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57dbaa97ab1a4ce58b77cc3c607e7839 : FAILURE in 6m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/953bbfc7bbc5444eac39ef181250f786 : FAILURE in 7m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/670201c75fdf4703acb5c8e6f76c0436 : FAILURE in 16m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f39802d0b68d45f49c6ed3203f15eced : TIMED_OUT in 2h 00m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb2ac63950224ca68a1dbb675a4ea4fd : TIMED_OUT in 40m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7be7f0dc551b46b2868ef02a4fb2e131 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dbfafe2e4f1642f1b5ed9987996781cc : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":4},{"id":"80599f37d914de021e271a60ed34706eb5864136","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 09:13:24.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/600e1e012b9d4c0f8dfb755cb23eae1c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4fd2b54c4cf24e9fa5c95cdcd91104e5 : TIMED_OUT in 3h 01m 00s","accounts_in_message":[],"_revision_number":4},{"id":"91729885fc3217a56d3e912ee6d2f25d02851d38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-11 10:39:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"722ab1a962fa693b5265fedb1fb3f835a2549565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 12:16:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7078c795e764976993efd201efef40f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0414a3e7634c9b9dd6ea1b8a9b16b5 : SUCCESS in 10m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e55194aa2c654c0399afdd285f9f3595 : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92ad9d1f773f49c79d8936eabcec43b7 : SUCCESS in 12m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78ee7b70813245f2a0ba42972755ca70 : SUCCESS in 20m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/35fa11590b80484c9fb4ea52ad2266dd : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/45da3f13ec26438394a0675ff54e6474 : TIMED_OUT in 1h 30m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/401cf5a3848041e890079d9749d8d4b7 : TIMED_OUT in 1h 30m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bca05cef44c146318a4ba5c15d984c3b : TIMED_OUT in 1h 30m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e9f2ab57170742c28b7b5d93ff166a2d : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9eb4bc346ce14a58a616c81d547903ea : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e875fe92a1174c088a1d3e50cca4ecfc : SUCCESS in 14m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/676947171f554a9198d80dff3d76cb37 : SUCCESS in 9m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89b0a26ffc27429188810d09e15123f0 : SUCCESS in 8m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87e645295c934fa48077bd9188e5d384 : SUCCESS in 14m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b297498c2dcf467b9877e34977f85abd : SUCCESS in 11m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/584a78e1d51b4b1287d39b1862f45636 : SUCCESS in 34m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4af3a72f0279415dad7a17995c686234 : TIMED_OUT in 40m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5eb270aaff4ab9b1b5985a95c39d2d : SUCCESS in 3m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e8982083c4ab4e888f1cfe2cb3b28f7b : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":5},{"id":"83538cf130b7d184acec68b8a5332a2cb95261ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 13:42:42.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe0f9b494d974f809d7d15f707e61740\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cdff832689e74bf99e867d88de819376 : TIMED_OUT in 3h 00m 56s","accounts_in_message":[],"_revision_number":5},{"id":"1809642b239f8e6d300162419014852b32a7ab99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-11 14:32:50.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"005258fa48999fb7c4bccea15f6cc57f56479c1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 15:34:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c00b900d51944e5682d8cc1072012d4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b480d5b092459ba8b94df157bc0ff2 : SUCCESS in 10m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c1f55e42dc74480af38f3167934b016 : SUCCESS in 10m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df90ee5b1da347f3aded08996a4e2490 : SUCCESS in 10m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ab9ea2469c743ad8f1bae7a34eeec49 : SUCCESS in 21m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b42b11afb4284e4082dfa5653dd2da4c : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7c05b4921ee24514adbed24391fac64d : SUCCESS in 21m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e81f6a1a643f4340b5ad0b77875976e0 : SUCCESS in 21m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3a13ff8f5b4441ecaeaee9cd70aa058f : SUCCESS in 14m 27s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fea593d7aabc47dfbee72cb1a12b0927 : SUCCESS in 11m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e7fc968b5ac148b8bd8a180046566e63 : SUCCESS in 23m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d6415db0a35246038265b6f7af7bdce6 : SUCCESS in 15m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d03f7ea364b9476cb9e1904f776bfdf9 : SUCCESS in 14m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/59bd42c3df754c65a2f24f86dca51099 : SUCCESS in 5m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e1f30ac514d54b2baf259a48986115a8 : SUCCESS in 13m 41s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5cdca8f8d3e14cfc8408dc1e65a78932 : SUCCESS in 8m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c09b9e5849642369b57976e1fb14300 : SUCCESS in 57m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1aa28a4ebb764c70b9526b9368a70e27 : SUCCESS in 22m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d263ea31826a4024bcf6fff9ff4628ec : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1c48fceee4ea4066ae7d356b6b6f4bfa : SUCCESS in 17m 16s","accounts_in_message":[],"_revision_number":6},{"id":"f2ace726f0e14a98989ef990257810bb3a515ad0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 16:27:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b53e2e360a9d48e88ea6daea87d1d495\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c5a5c0ffc4a0458e95f59ffc62f3b564 : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":6},{"id":"7f349c71b139c735a13fab7d9169e339bd793b02","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-12 06:55:55.000000000","message":"Topic eventlet-removal-gerrit-functests removed","accounts_in_message":[],"_revision_number":6},{"id":"6cc59083c1fdb8bec806991e7c0b431a7ec3e37c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-12 06:56:02.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":6},{"id":"43b80b0145f341a50d0e7c0a86e1878e69cd3b1c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-25 19:19:02.000000000","message":"Patch Set 6:\n\n(10 comments)","accounts_in_message":[],"_revision_number":6},{"id":"824a4c05e7aceb876ed014940d8e8c8615af3052","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-10-27 11:17:19.000000000","message":"Uploaded patch set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"84db41f283379a04d7a83f0823cc39415e2fb815","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 11:50:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/482f19a566ad440ead04453797a3748b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/91fe1d4e38d34488a70bade89b835166 : SUCCESS in 31m 26s","accounts_in_message":[],"_revision_number":7},{"id":"8cd7d144ad8fcacd7b6a721b6372af1061e9f0c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 12:26:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc28247608b942139dd4a039d536e795\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87c0ffeedc1c44f3b626146e0f672867 : SUCCESS in 10m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ed259d4b57340cbb93179dd358f599d : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce6ab9a9eb594fd7af413d31eaac48a3 : SUCCESS in 10m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca6ed8e974f248a582a8c437e0fb30f1 : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a29e4004c9f744febeedae020cfaf0d8 : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e607b1f0eab64f8d97226155d3560ebb : SUCCESS in 23m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9bb4ab829af24ed38bb8fd42586ef42d : SUCCESS in 23m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/406e9b4c8b344b3fa29545c5c9e8d97b : SUCCESS in 31m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8cf79bf184c849adb7519f94f3ef4721 : SUCCESS in 32m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0a55046b21d64367a8174fb113199ca5 : SUCCESS in 12m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/03a282b336da42848f7771e43770c01e : SUCCESS in 16m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a52de406067c4cd2b385af0559db586c : SUCCESS in 14m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8729fc5802804984b7b9d97df966f120 : SUCCESS in 7m 16s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/95df06c4ae894300b94194b1c42e3664 : SUCCESS in 15m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d68d120103404ae89383b90f3cf34d35 : SUCCESS in 8m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/27f69fee26c348dabe61bda6c39df91d : SUCCESS in 1h 02m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/311d163aa84f49acb387fa486dacb6a7 : SUCCESS in 27m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85729db447f473ea6b907aa7863d736 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f5a6f2d60ab544cea4e191e29754a173 : SUCCESS in 17m 21s","accounts_in_message":[],"_revision_number":7},{"id":"5819020bb1e992d3591ad673183479d89189bba3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-06 11:14:39.000000000","message":"Abandoned\n\nProposing new patches shortly.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ebc819807432fd311087e5a5193f3c042c93d212","revisions":{"880b2611ca0485b59665dc8a86403d29fc1b70ea":{"kind":"REWORK","_number":1,"created":"2025-09-02 14:31:56.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/93/959193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/959193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/959193/1"}}},"commit":{"parents":[{"commit":"1de6326e87e7603f1c38ab6d6b607fb024ca1dd7","subject":"WIP: Use gunicorn as wsgi server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1de6326e87e7603f1c38ab6d6b607fb024ca1dd7"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"subject":"WIP: Remove classes not used by the POC requiring eventlet","message":"WIP: Remove classes not used by the POC requiring eventlet\n\nThis commit deletes a couple of classes not used by the POC\nto remove the eventlet imports. It needs to be reverted and code\nrefactored to be usable without eventlet.\n\nChange-Id: Icc621e2e92ef86b418194241b4cc5396f30ac49e\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/880b2611ca0485b59665dc8a86403d29fc1b70ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/880b2611ca0485b59665dc8a86403d29fc1b70ea"}]},"branch":"refs/heads/feature/threaded"},"d18b37caf7f4f2d27fc4c8de886b4cf8cd308387":{"kind":"REWORK","_number":2,"created":"2025-09-03 10:10:21.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/93/959193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/959193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/959193/2"}}},"commit":{"parents":[{"commit":"c55186c460c41e9f048865bb16ca498ebe703007","subject":"WIP: Add dummy Timeout exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c55186c460c41e9f048865bb16ca498ebe703007"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-03 10:10:13.000000000","tz":120},"subject":"WIP: Replace eventlet imports with native Python modules","message":"WIP: Replace eventlet imports with native Python modules\n\nThis allows running the Swift account-server without using eventlet if\nthe environment variable DISABLE_EVENTLET\u003dtrue is set. In this case a\ncouple of eventlet imports are disabled. Native Python modules are used\nwhere required to run the account-server, eg. time.sleep, socket, ssl,\nos, http.client, threading.\n\nChange-Id: Icc621e2e92ef86b418194241b4cc5396f30ac49e\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d18b37caf7f4f2d27fc4c8de886b4cf8cd308387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d18b37caf7f4f2d27fc4c8de886b4cf8cd308387"}]},"branch":"refs/heads/feature/threaded"},"2f461f490f1a164011b74190a901774b85ecff9b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-05 16:39:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/959193/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/959193/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/959193/3"}}},"commit":{"parents":[{"commit":"99c7a8b2c5ef104f0b17c736a9915e918fb10589","subject":"WIP: Add dummy Timeout exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c7a8b2c5ef104f0b17c736a9915e918fb10589"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-09-05 16:38:59.000000000","tz":-420},"subject":"WIP: Replace eventlet imports with native Python modules","message":"WIP: Replace eventlet imports with native Python modules\n\nThis allows running the Swift account-server without using eventlet if\nthe environment variable DISABLE_EVENTLET\u003dtrue is set. In this case a\ncouple of eventlet imports are disabled. Native Python modules are used\nwhere required to run the account-server, eg. time.sleep, socket, ssl,\nos, http.client, threading.\n\nChange-Id: Icc621e2e92ef86b418194241b4cc5396f30ac49e\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f461f490f1a164011b74190a901774b85ecff9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f461f490f1a164011b74190a901774b85ecff9b"}]},"branch":"refs/heads/feature/threaded"},"9a2002451c96707c6a3a2033ffc7b6b3bbb846bc":{"kind":"REWORK","_number":4,"created":"2025-09-10 06:10:04.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/93/959193/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/959193/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/959193/4"}}},"commit":{"parents":[{"commit":"99c7a8b2c5ef104f0b17c736a9915e918fb10589","subject":"WIP: Add dummy Timeout exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c7a8b2c5ef104f0b17c736a9915e918fb10589"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-10 06:09:36.000000000","tz":120},"subject":"WIP: Replace eventlet imports with native Python modules","message":"WIP: Replace eventlet imports with native Python modules\n\nThis allows running the Swift account-server without using eventlet if\nthe environment variable DISABLE_EVENTLET\u003dtrue is set. In this case a\ncouple of eventlet imports are disabled. Native Python modules are used\nwhere required to run the account-server, eg. time.sleep, socket, ssl,\nos, http.client, threading.\n\nChange-Id: Icc621e2e92ef86b418194241b4cc5396f30ac49e\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a2002451c96707c6a3a2033ffc7b6b3bbb846bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a2002451c96707c6a3a2033ffc7b6b3bbb846bc"}]},"branch":"refs/heads/feature/threaded"},"670e6cb6f36c4d00be6ecc12421ff5df11b806e1":{"kind":"REWORK","_number":5,"created":"2025-09-11 10:39:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/93/959193/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/959193/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/959193/5"}}},"commit":{"parents":[{"commit":"99c7a8b2c5ef104f0b17c736a9915e918fb10589","subject":"WIP: Add dummy Timeout exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c7a8b2c5ef104f0b17c736a9915e918fb10589"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-11 07:05:13.000000000","tz":120},"subject":"WIP: Replace eventlet imports with native Python modules","message":"WIP: Replace eventlet imports with native Python modules\n\nThis allows running the Swift account-server without using eventlet if\nthe environment variable DISABLE_EVENTLET\u003dtrue is set. In this case a\ncouple of eventlet imports are disabled. Native Python modules are used\nwhere required to run the account-server, eg. time.sleep, socket, ssl,\nos, http.client, threading.\n\nChange-Id: Icc621e2e92ef86b418194241b4cc5396f30ac49e\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/670e6cb6f36c4d00be6ecc12421ff5df11b806e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/670e6cb6f36c4d00be6ecc12421ff5df11b806e1"}]},"branch":"refs/heads/feature/threaded"},"211a6477c1da58d78edfd8d3e06a01e6c475d224":{"kind":"REWORK","_number":6,"created":"2025-09-11 14:32:50.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/93/959193/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/959193/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/959193/6"}}},"commit":{"parents":[{"commit":"99c7a8b2c5ef104f0b17c736a9915e918fb10589","subject":"WIP: Add dummy Timeout exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c7a8b2c5ef104f0b17c736a9915e918fb10589"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-11 14:31:21.000000000","tz":120},"subject":"WIP: Replace eventlet imports with native Python modules","message":"WIP: Replace eventlet imports with native Python modules\n\nThis allows running the Swift account-server without using eventlet if\nthe environment variable DISABLE_EVENTLET\u003dtrue is set. In this case a\ncouple of eventlet imports are disabled. Native Python modules are used\nwhere required to run the account-server, eg. time.sleep, socket, ssl,\nos, http.client, threading.\n\nChange-Id: Icc621e2e92ef86b418194241b4cc5396f30ac49e\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/211a6477c1da58d78edfd8d3e06a01e6c475d224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/211a6477c1da58d78edfd8d3e06a01e6c475d224"}]},"branch":"refs/heads/feature/threaded"},"ebc819807432fd311087e5a5193f3c042c93d212":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-10-27 11:17:19.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/93/959193/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/959193/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/959193/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/959193/7"}}},"commit":{"parents":[{"commit":"99c7a8b2c5ef104f0b17c736a9915e918fb10589","subject":"WIP: Add dummy Timeout exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c7a8b2c5ef104f0b17c736a9915e918fb10589"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-27 11:12:44.000000000","tz":60},"subject":"Replace eventlet imports with native Python modules","message":"Replace eventlet imports with native Python modules\n\nThis allows running the Swift account-server without using eventlet if\nthe environment variable DISABLE_EVENTLET\u003dtrue is set. In this case a\ncouple of eventlet imports are disabled. Native Python modules are used\nwhere required to run the account-server, eg.  time.sleep, socket, ssl,\nos, http.client, threading.\n\nFor example, using a SAIO deployment you can test this by stopping the\ndefault account-server first and then running one without eventlet:\n\n    swift-init stop account-server\n    DISABLE_EVENTLET\u003dtrue swift-account-server /etc/account-server.conf\n\nswift-init does not work if running without eventlet (yet). You can also\ndeploy an additional virtualenv and removing eventlet there to enforce\nthis, for example to ensure there is no eventlet module still in use:\n\n    virtualenv  ~/.venv3withouteventlet/\n    source  ~/.venv3withouteventlet/bin/activate\n    pip install -r requirements.txt -r test-requirements.txt\n    pip uninstall -y eventlet\n    python setup.py install\n    DISABLE_EVENTLET\u003dtrue swift-account-server /etc/account-server.conf\n\nChange-Id: Icc621e2e92ef86b418194241b4cc5396f30ac49e\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebc819807432fd311087e5a5193f3c042c93d212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebc819807432fd311087e5a5193f3c042c93d212"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
