)]}'
{"id":"openstack%2Fswift~959196","triplet_id":"openstack%2Fswift~feature%2Fthreaded~Ia32f48d062698c3c7513150a3d518e2ee18f470e","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2025-09-03 10:10:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia32f48d062698c3c7513150a3d518e2ee18f470e","subject":"Remove eventlet_debug setting","status":"ABANDONED","created":"2025-09-02 14:31:56.000000000","updated":"2025-09-03 10:10:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ef2fd49ad712fd1b5a0fc8a5bffa244f7942dcbd","_number":959196,"virtual_id_number":959196,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-02 15:01:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-02 15:01:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f471e9d955a82563055de1da318c065001292744","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-02 14:31:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2ccf2fa267e5f90e1909c4b324d7ec8fdd8e65c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 15:01:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a24e00b85a8f423598ca3d55f78c180e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd9fbd1855b541459ab4ea5a9a1fdb74 : FAILURE in 5m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/996b06b5dc554393963344c17744b586 : FAILURE in 4m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c02f9cc47254e9782254559a5989b33 : FAILURE in 6m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ad83581220f4a2495dc4203d7104217 : FAILURE in 6m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e75f103d00854184856a47a50f84b6a3 : SUCCESS in 7m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/51816f7c0bfa46eeac53a6600e66527c : FAILURE in 9m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8966a5afc861431293fb4ab5fc29ea57 : FAILURE in 4m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ceb4e076d5b045eca87374315694e1ea : FAILURE in 7m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/05d6b7c831a340b3b4e4564bc74dbd3c : FAILURE in 8m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c9c99b19dfd84b2782bb2f81c43a11b4 : FAILURE in 4m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6f81381297e54867a608202186eac40e : FAILURE in 2m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f0ec59c2e7f94cd7be4e12069aa639dc : FAILURE in 2m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/23481d63e9ac445d87ac2c409b2a539f : RETRY_LIMIT in 4m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/793c239db48a434ca9742fad064df92d : RETRY_LIMIT in 3m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0dd711948aa2491f8243392c551aeb68 : RETRY_LIMIT in 4m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/35f63a2bf4cd40e9bd1d7cee86b1ceee : RETRY_LIMIT in 2m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a26ac1056ee4032bfdff0dac328f893 : FAILURE in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/331ba9d923354a9a935e4995f7707097 : FAILURE in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12123b4f36c04c3a8b530f5ef756d1db : SUCCESS in 18m 32s\n\nWarning:\n  Comments left for invalid file swift/common/exceptions.py","accounts_in_message":[],"_revision_number":1},{"id":"8c608dfa2e1191dee47aac0598d979b836706ca1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 17:06:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50cfe186c1a14d65a95361c6100432a5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d22dffbbe1a042aea14eec8635e3b952 : FAILURE in 15m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ef2fd49ad712fd1b5a0fc8a5bffa244f7942dcbd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-03 10:10:58.000000000","message":"Abandoned\n\nSquashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20ba2c6911a87f28a254a1973c6b2b9fa70dabc1","revisions":{"20ba2c6911a87f28a254a1973c6b2b9fa70dabc1":{"kind":"REWORK","_number":1,"created":"2025-09-02 14:31:56.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/96/959196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/959196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/959196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/959196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/959196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/959196/1"}}},"commit":{"parents":[{"commit":"ec62b2ff31ca72f13aef638c255e95f0c100b4ae","subject":"WIP: Use native Python modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec62b2ff31ca72f13aef638c255e95f0c100b4ae"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"subject":"Remove eventlet_debug setting","message":"Remove eventlet_debug setting\n\nChange-Id: Ia32f48d062698c3c7513150a3d518e2ee18f470e\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20ba2c6911a87f28a254a1973c6b2b9fa70dabc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20ba2c6911a87f28a254a1973c6b2b9fa70dabc1"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
