)]}'
{"id":"openstack%2Fswift~959197","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I868324ceae760b9434570e38cdcf3e58bbbc4023","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2025-09-03 10:11:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I868324ceae760b9434570e38cdcf3e58bbbc4023","subject":"Remove utils.get_hub()","status":"ABANDONED","created":"2025-09-02 14:31:56.000000000","updated":"2025-09-03 10:11:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"31e8cb8f44d695c811ef2ab6db740e9175b34e66","_number":959197,"virtual_id_number":959197,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-02 15:06:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-02 15:06:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c0c0ab0bf9c96b51c33a5b5f4b676943ca9a3c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-02 14:31:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8c18763f746e494a84721252d4f5b522b79f18c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 15:06:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdde5eff0d0f41daa0ecdccabad2ca07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ed627c6d184f3480b6a564f8d9eccb : FAILURE in 7m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ed568726e154ad6babc7eddc5a39780 : FAILURE in 4m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d240b774f9324d9bb7ab3e5294b02bd2 : FAILURE in 8m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec23ddc44db440ae85669dab02682f9d : FAILURE in 7m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18bc221a39b44caf9b9e9fc983902cc9 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f2519547a0044762bbf5f9ffe80ac806 : FAILURE in 7m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/31476af94c184a4389232e10e0bcd7a2 : FAILURE in 3m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b19bb7da5a244254ac503f4638ac31e6 : FAILURE in 6m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/229cb40a7a034ae79dff92130b53cf41 : FAILURE in 9m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/116a7b207f93491e842391e53ad4a4b5 : FAILURE in 3m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/40d3654e33c448878c91cff51fc32a5b : FAILURE in 4m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ace0b70c8290415ea3ea078b96f2eff6 : FAILURE in 4m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9d444fec2dc43b496421e9998ba6d3f : RETRY_LIMIT in 4m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/498c57e32746462b830c3e811b0aec26 : RETRY_LIMIT in 4m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d4924f3f807c4c61920c5d5559dff937 : RETRY_LIMIT in 2m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/263baa6963594c05aa429555f08bedab : RETRY_LIMIT in 4m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7be8ef021b144c56a33bc2221af93a6c : FAILURE in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e175552e2442d49b5445a4cf7b07de : FAILURE in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/14a3cac0257b4f848f4ebe70fa1c2215 : SUCCESS in 25m 05s\n\nWarning:\n  Comments left for invalid file swift/common/exceptions.py","accounts_in_message":[],"_revision_number":1},{"id":"44a537fb7a7566da38962e417ee70f68767e579c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 16:39:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f703105fad5a448183a8bee620c546ba\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6875b70ac90945f4bc2223e6595fc3a9 : FAILURE in 17m 03s","accounts_in_message":[],"_revision_number":1},{"id":"31e8cb8f44d695c811ef2ab6db740e9175b34e66","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-03 10:11:12.000000000","message":"Abandoned\n\nSquashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb06a6d101bdd531eeb8cb7d4fe62e2f8edf300a","revisions":{"fb06a6d101bdd531eeb8cb7d4fe62e2f8edf300a":{"kind":"REWORK","_number":1,"created":"2025-09-02 14:31:56.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/97/959197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/959197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/959197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/959197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/959197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/959197/1"}}},"commit":{"parents":[{"commit":"20ba2c6911a87f28a254a1973c6b2b9fa70dabc1","subject":"Remove eventlet_debug setting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20ba2c6911a87f28a254a1973c6b2b9fa70dabc1"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"subject":"Remove utils.get_hub()","message":"Remove utils.get_hub()\n\nThis also removes any calls of eventlet.hubs.use_hub(utils.get_hub())\n\nChange-Id: I868324ceae760b9434570e38cdcf3e58bbbc4023\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb06a6d101bdd531eeb8cb7d4fe62e2f8edf300a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb06a6d101bdd531eeb8cb7d4fe62e2f8edf300a"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
