)]}'
{"id":"openstack%2Fswift~959199","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I4588295998203a3ee8dc4127966e229f4be0fc75","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2025-09-03 10:10:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4588295998203a3ee8dc4127966e229f4be0fc75","subject":"WIP: Disable eventlet monkeypatching","status":"ABANDONED","created":"2025-09-02 14:31:56.000000000","updated":"2025-09-03 10:10:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d066a6714669831b47fa9ed2d6d5f38b7dfa45a0","_number":959199,"virtual_id_number":959199,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-02 14:57:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-02 14:57:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec80bfb050e46aea12aba90c8ab6abbfe55be8cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-02 14:31:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70f28cce50ddec19c6ff2cede56caf9480b516ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 14:57:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f46b08fbe29470ea05dd143c6c20fcf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e46dcf5c2b47437fb3fa6476b4ffe17f : FAILURE in 5m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05185e9ccdb4495a8a06dff78f2352ea : FAILURE in 4m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f542c8fdd0494d29b9f9d8bc6517ba5c : FAILURE in 7m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3e2c5f21fb041c69c7b4ce3dd14d231 : FAILURE in 6m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e99de7d6d7d4905a2cd9e6dcc9b38ec : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ebdfd232117343aebc64172ad2336f12 : FAILURE in 10m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/02020f0b4eeb46bcb37fa54de957f6e9 : FAILURE in 5m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2f05616a71e3456cb164e587f5393d87 : FAILURE in 8m 30s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/89e81b681fff411e8d041f65d363fd11 : FAILURE in 6m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d067c96918b84a669e3e221fcd448ff9 : FAILURE in 4m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3d945aa1549c451a87db1706e25cfdc8 : FAILURE in 3m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/be3c4e4ab13e465cb53bbc98eb77174a : FAILURE in 4m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fc3b4f2e7b1e4a648c2a3901ac3c6549 : RETRY_LIMIT in 2m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f32159b72beb4e2cba1252117baa2cdc : RETRY_LIMIT in 4m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/23b058c67e5441b59ac67a05a59fc08f : RETRY_LIMIT in 5m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f45a3ea44b84063b1f8c6b46031ab99 : RETRY_LIMIT in 2m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acd483d04319492e81c8a84bfd80e2b3 : FAILURE in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0578f75d3e3649c1ad78022219f3b45b : FAILURE in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59bbb6390e24402bad8d153c0c4f766e : SUCCESS in 19m 04s\n\nWarning:\n  Comments left for invalid file swift/common/exceptions.py","accounts_in_message":[],"_revision_number":1},{"id":"6a90e77c93c59c95ae59342dd385b885cdeebe39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 17:12:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84bc7c2ada6441e2b598a78731ae45f7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/220a77e81a1b456f934197444d9de654 : FAILURE in 15m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d066a6714669831b47fa9ed2d6d5f38b7dfa45a0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-03 10:10:54.000000000","message":"Abandoned\n\nSquashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db1039a731e30b4d5ec3df149703adac52260c6c","revisions":{"db1039a731e30b4d5ec3df149703adac52260c6c":{"kind":"REWORK","_number":1,"created":"2025-09-02 14:31:56.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/99/959199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/959199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/959199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/959199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/959199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/959199/1"}}},"commit":{"parents":[{"commit":"5b0c302cf3824ae71d6a7374aa69f3e7485c4277","subject":"WIP: Remove eventlet.debug.hub_prevent_multiple_readers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b0c302cf3824ae71d6a7374aa69f3e7485c4277"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"subject":"WIP: Disable eventlet monkeypatching","message":"WIP: Disable eventlet monkeypatching\n\nChange-Id: I4588295998203a3ee8dc4127966e229f4be0fc75\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db1039a731e30b4d5ec3df149703adac52260c6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db1039a731e30b4d5ec3df149703adac52260c6c"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
