)]}'
{"id":"openstack%2Fswift~959414","triplet_id":"openstack%2Fswift~unmaintained%2Fzed~I1245d9a2cb2e5cd59dd5131c425a3c99503c92fa","project":"openstack/swift","branch":"unmaintained/zed","attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-09-10 10:53:45.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1245d9a2cb2e5cd59dd5131c425a3c99503c92fa","subject":"DNM: CI health test","status":"NEW","created":"2025-09-04 07:10:52.000000000","updated":"2025-09-10 12:34:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4ea3890157e44f6b06524ce5df30b8956a9a603f","_number":959414,"virtual_id_number":959414,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-05 16:59:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-04 07:56:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-04 08:55:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ff483eff2347aad0fae38a6359b0e82d8333c51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-04 07:10:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"175752aeb76d5026f5cceaa56da97d1c681a73db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 07:56:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b126c4e236d49118b3d8608d1d6ce16\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5ca1af8bc944329b59e268b67767bc8 : SUCCESS in 40m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d8668c5f6b5442b8d35584a94743993 : SUCCESS in 41m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/876f100250ab4d6888f1e8d6964f6e3a : SUCCESS in 38m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/934374921ccf4487bfbff742a898008b : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b8651c11ed730c9f8b9bf6703ab544067ef681ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 08:55:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9fbcac5aee44c7389d31f1dbb42a117\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2bc616bf854f3bab882b5c71627f00 : SUCCESS in 6m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9350e54841ca42388815d1ff1e6e17c7 : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/015e990faa774f339fc236eec66b9fc5 : SUCCESS in 9m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fb56dd31da34ab9bca4a11b4eea7238 : SUCCESS in 18m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2afcf97e3f584062acff74702f860401 : SUCCESS in 11m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4f75ce0690ef47be9e17836a8219c360 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9102a687cc5c425784d6a61e5edc1cfb : RETRY_LIMIT in 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c3556079ea94fcf8d92a1f137f15e90 : RETRY_LIMIT in 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/22485b38bff14dc1bee864bfbf961848 : SUCCESS in 9m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/92eb4a7be0cb49e0b7adef07685b13ba : RETRY_LIMIT in 9s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c275fcaedb447c3be03b7ca77ae0071 : RETRY_LIMIT in 8s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/32248e1b2fd144e8bfa03750750b742d : RETRY_LIMIT in 9s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b572676574784b2cbfadd6f95adc44fb : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1fcbf50b48004fca9fcc7de57d2924ae : SUCCESS in 22m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cca1bf7281a9433b9bebb5e8b84ae204 : SUCCESS in 15m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/414721b05aee4ca7a4ac66a483fe1080 : SUCCESS in 8m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/741e2b6fd1684ea7a31de4aa8b44df5c : FAILURE in 15m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ce38b46adaaf40e181b3ef0d9e832985 : SUCCESS in 6m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8127cef8bcbe49cb82b2bb64f7e567a4 : SUCCESS in 1h 08m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e13e073aeb434d7fae3a2cd5a4f1a7ae : SUCCESS in 56m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ffc25e97b7bd4cafa9d23b8da193886a : SUCCESS in 45m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d751cc8ca51a44a188a01cf6a7f7eef9 : SUCCESS in 23m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbbcab4dd1964d1088d890a41e749f68 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a8552f7bf8ed4129a8fa19299450d1d1 : TIMED_OUT in 1h 31m 13s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/68ff38c494c74f9aa6dddfb40a4fa5ca : SUCCESS in 44m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bf5553ea4b24030a919589809865986 : SUCCESS in 39m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/12a4456d9031410fb1b0f1825ad8e01f : SUCCESS in 56m 08s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/119a94844ef04b21aa6ad0fbb4f1ac0a : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":1},{"id":"744ba6b054bf982ec40b6734d04a961ab35f6454","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-05 15:23:25.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1370caf074a916ae292477fca65f6c9a7491845b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 16:59:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a920cc387724e659200c073f749d856\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dab77e4cdd54c10beab147fce9673b6 : SUCCESS in 8m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92d89ed650a24d6fbb2bee098d3e8ecf : SUCCESS in 22m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47982e8082f749d6aa716531006bf19c : SUCCESS in 24m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/502adaac01b5450194943b44462bf571 : SUCCESS in 24m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18886ac599aa449d85d9b418ddc0cc3e : SUCCESS in 11m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/76b6c41eca0a462a96f73e50268d24ed : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8c62bf702fa84b438e2645c6d56b704b : RETRY_LIMIT in 2m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c20bff61dc05490b99c2494629a09c28 : RETRY_LIMIT in 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/75bf33f39e794405820364013851c707 : SUCCESS in 17m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d6852e02d38444d1b4319af0556c5434 : RETRY_LIMIT in 2m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0e7dab85f3464a5d88b029f34abc7d46 : RETRY_LIMIT in 2m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/849aa13289e946da91427a1ffb163c33 : RETRY_LIMIT in 1m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cf6a7ec77b124ba19bd9105c83ef02db : SUCCESS in 19m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/846fb447e00f47db932b895bad658c29 : SUCCESS in 14m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dd67ceb09d364cbdbd16718cb9605e9b : SUCCESS in 21m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f0572d3ff4e741c59082d9cecd7934c9 : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ad6a557cef9459fab17a6bf45923985 : FAILURE in 16m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/43eb71c09511460a96e9a26949cb7899 : SUCCESS in 3m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/af0b9055412e47728ba666ae0da5e301 : SUCCESS in 57m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/24e6ab0082cb4e9b8ee450b585c1eb01 : SUCCESS in 47m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a42d9496b9544e6b543c86ef1f4b57b : SUCCESS in 55m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22eb8d0364e1489a94d671bdee1f7aa8 : SUCCESS in 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f704ba9c54b344ccbaab38f3254fc450 : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/28acc51adf9a4893a73bd92e32f50043 : TIMED_OUT in 1h 30m 47s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0dc0c34862f94e43abefedeeafd16572 : SUCCESS in 44m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/608c353c4f084d03b09053705bea73f1 : SUCCESS in 44m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/c93558b96d4449439979e6ef52718986 : SUCCESS in 54m 17s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd12163dff114404b1aba4dfddd0656d : SUCCESS in 1h 09m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f7bbc90395bf55930ca24550d419b8edb3be40df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 18:06:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9b209b2e1da4062b2e77bd3f3f2256d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fc672f47aae46478ea58e76aecb176f : SUCCESS in 52m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a7884da99ce4b85ba6ea26bc5e4491b : SUCCESS in 48m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c6d4fb71e5242c19c9a5ba9c44f9fce : SUCCESS in 51m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fcf3369e241b4b90a0cdb6587d807437 : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":2},{"id":"77b58c0f50cfbd6cb47ccea744c55f08a78afb9d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-10 09:14:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e1e9a8217a7e64d37a4ba0539ee2071abfac6b22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:53:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7948f55c4da24fe299b95d036f7fd546\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e73d2da30ac4e07b9b2199ca0ade648 : SUCCESS in 20m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ad6c32d2c2a466cb053b6625d0c98ca : SUCCESS in 22m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ada6812cfcd45838523f00713e18df8 : SUCCESS in 23m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8f79daaf46b47c7812cd7c089bfab80 : SUCCESS in 30m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ad895abe60de460d9cb95325e3701092 : SUCCESS in 13m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f598620e261140c9b87366842b0a4406 : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a66d3fb33be54c00bfb176c10e881905 : RETRY_LIMIT in 1m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c360252168b4a6587f4612b22f5a8b7 : RETRY_LIMIT in 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3a4d62c926b34e54b36591d35cde58a7 : SUCCESS in 8m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/af2c47f86efa4feaa0bbf49d173f4a63 : RETRY_LIMIT in 3m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c3a1e10464084cf1af8af4393b0203bb : RETRY_LIMIT in 1m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cffd62113f0441c4bce022ddbb5f8122 : RETRY_LIMIT in 1m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ad631343d5d64df1a94d3eeb8103b494 : SUCCESS in 8m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dc59684efc074d9faaf5f148f843fb73 : SUCCESS in 11m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5753287a92b54594ad9e1f5f3168d554 : SUCCESS in 20m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/935dde727fcf414fb89e3d9e4ee18824 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6cfea813415d4756b69059295d57401e : FAILURE in 14m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d9f006264d82459a8fdfd877b05354a3 : SUCCESS in 6m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/51b9339fc819489b9e8f139ee0df9a1a : SUCCESS in 52m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55a4b7ebc923453b95809800ea1493c8 : SUCCESS in 26m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/33270060f1af4bc7884c5cfbbf83b57b : SUCCESS in 20m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0aa28bc0479d449a9657e16fe6d16b83 : SUCCESS in 22m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f1a44262324422af0b87a9799a4340 : SUCCESS in 3m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/15563ca7f2d140c1bebc294e3862840c : TIMED_OUT in 1h 30m 51s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7651f07d16c9427594559bd1da2d3b5b : SUCCESS in 54m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6d4fc0ea3b94c2d955357dae6076554 : SUCCESS in 23m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/872ff705949f4cdaa9618ecf10b6152b : SUCCESS in 1h 04m 36s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24fd43f220e5457697d5e6b3dfa92601 : SUCCESS in 48m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4ea3890157e44f6b06524ce5df30b8956a9a603f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 12:34:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3985d6fb67ce45428e18b7ae52c002af\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/65d765b973ad4a43a7a543f33ffa0b85 : SUCCESS in 50m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6cbfc620a6d643bd8791d1bfdda34f21 : SUCCESS in 42m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d421f18392be4bc3a94fa83b750deea1 : SUCCESS in 52m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f981a6f5e5e144a9b41db2a79e9d0d37 : SUCCESS in 53m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e140213384c038f74594a84bb19cf6c8d78fecb8","revisions":{"ed263bafbd05ff52c05719a0c16962bf4c36930d":{"kind":"REWORK","_number":1,"created":"2025-09-04 07:10:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/14/959414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/959414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/959414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/959414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/959414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/959414/1"}}},"commit":{"parents":[{"commit":"fb24f068730d57149b02d175f0919314380e04ff","subject":"Merge \"[CI] Use defined version of probetest job\" into unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb24f068730d57149b02d175f0919314380e04ff"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-04 07:10:28.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-04 07:10:43.000000000","tz":120},"subject":"DNM: CI health test","message":"DNM: CI health test\n\nChange-Id: I1245d9a2cb2e5cd59dd5131c425a3c99503c92fa\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed263bafbd05ff52c05719a0c16962bf4c36930d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed263bafbd05ff52c05719a0c16962bf4c36930d"}]},"branch":"refs/heads/unmaintained/zed"},"e140213384c038f74594a84bb19cf6c8d78fecb8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-09-05 15:23:25.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/14/959414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/959414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/959414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/959414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/959414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/959414/2"}}},"commit":{"parents":[{"commit":"fb24f068730d57149b02d175f0919314380e04ff","subject":"Merge \"[CI] Use defined version of probetest job\" into unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb24f068730d57149b02d175f0919314380e04ff"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-04 07:10:28.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-05 15:23:25.000000000","tz":0},"subject":"DNM: CI health test","message":"DNM: CI health test\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/958668\nChange-Id: I1245d9a2cb2e5cd59dd5131c425a3c99503c92fa\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e140213384c038f74594a84bb19cf6c8d78fecb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e140213384c038f74594a84bb19cf6c8d78fecb8"}]},"branch":"refs/heads/unmaintained/zed","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
