)]}'
{"id":"openstack%2Fswift~959419","triplet_id":"openstack%2Fswift~unmaintained%2Fzed~I83e3bc609433995875e47c4103ca945154088a65","project":"openstack/swift","branch":"unmaintained/zed","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-09-10 10:21:54.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-09-04 10:09:52.000000000","reason":"Reviewer/Cc was removed"}},"hashtags":[],"change_id":"I83e3bc609433995875e47c4103ca945154088a65","subject":"CI: Fix py27/py36/py37 jobs","status":"MERGED","created":"2025-09-04 08:46:31.000000000","updated":"2025-09-12 05:39:42.000000000","submitted":"2025-09-10 11:37:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959419","meta_rev_id":"0d7ba76f4a70bb8769e94fba99e3798d00b1a043","_number":959419,"virtual_id_number":959419,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-10 11:37:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-09-10 10:21:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-09-10 10:21:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-04 08:46:31.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-09-04 09:37:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-04 09:45:11.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-09-04 09:45:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-04 09:46:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-09-04 10:09:52.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"822c677aea4a129cbe328190195bf3a416a02f2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-04 08:46:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5006b80c5a4dcc0997112469cc39031c99b0ab94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:37:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edef1b9439ce4ccbb02089da9e90f5d8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0e5da2877f8426d97b309be6df92613 : SUCCESS in 39m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60cf664ba9dc4a7f99dfbf09a23e0397 : SUCCESS in 26m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/011858dee0c6471198e357f129c74f9f : SUCCESS in 48m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/44a97e1933dd4a6bb15ef206d76cdc70 : SUCCESS in 48m 01s","accounts_in_message":[],"_revision_number":1},{"id":"98a8835d84f7529261dc62825bd632c810fdc554","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-04 09:45:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d64b227b98096c6cea4c87569da33fadcb66a089","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:45:32.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch unmaintained/zed.  The\nproblem was:\n\n  The key \"description\" appears more than once; duplicate keys are not permitted.\n\n  name: swift-tox-func-py27\n      parent: swift-tox-base\n      nodeset: ubuntu-jammy\n      description: |\n      description: |\n        Run functional tests for swift under cPython version 2.7.\n  \n        Uses tox with the ``func`` environment.\n        It sets TMPDIR to an XFS mount point created via\n  ...\n\n  in \"openstack/swift/.zuul.yaml@unmaintained/zed\", line 128","accounts_in_message":[],"_revision_number":2},{"id":"50606a47900ab39f5bf4bc5b784dac6345125231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:45:52.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch unmaintained/zed.  The\nproblem was:\n\n  The key \"description\" appears more than once; duplicate keys are not permitted.\n\n  name: swift-tox-func-py27\n      parent: swift-tox-base\n      nodeset: ubuntu-jammy\n      description: |\n      description: |\n        Run functional tests for swift under cPython version 2.7.\n  \n        Uses tox with the ``func`` environment.\n        It sets TMPDIR to an XFS mount point created via\n  ...\n\n  in \"openstack/swift/.zuul.yaml@unmaintained/zed\", line 128","accounts_in_message":[],"_revision_number":2},{"id":"861b928c2b5c0e4f5ab52c9cfd8cb7a7b0e2b4f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-04 09:46:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d970fa3e9a33b888bbe7d33b98b0c8c06db10c73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-04 10:09:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e4c974d5b1fa97dc4f65b8320165c479ddad85be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 11:21:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8db0e1e01e0b48dba33045964925d1a2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fcad4406a114234823214c84487f751 : SUCCESS in 47m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab9e59be7a6547ccb5d6b9fe3873dd0b : SUCCESS in 49m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb06599c881141ca83efc27a1b274025 : SUCCESS in 50m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6cf0259c837a4c37a4520271e4535d88 : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":4},{"id":"31f8e5e2b3facbfff675b6626302c2fb132971f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 11:49:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64d2fa680afd4f1183586f4f41f3194f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a50035499ae4be484088e954a159d27 : SUCCESS in 8m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4259e99a09564c809d41ba88287c43cd : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8aaebe1373649728cc0fbb55c31d70b : SUCCESS in 9m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9b72c13cd644720a8c648b43d7577a2 : SUCCESS in 15m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/83cc8203307e4d79a5be6a4cf515baa7 : SUCCESS in 10m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/36ed9ad942a2434aafbe6c8c8963647d : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3792e62ae49b4c70b52130e9537a37bd : SUCCESS in 17m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a2eb14cd956a4871bbb747cd21c0405d : SUCCESS in 9m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0fa116aa04454defa099cefa1d0acb52 : SUCCESS in 12m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2d25272430684002b3460e6124cfbaab : SUCCESS in 12m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7d07f5f3616e4abf89a3d00c3c059b14 : SUCCESS in 12m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d94719d022a4b4396bf4c2fcb56b101 : SUCCESS in 10m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/230c02acceaa485d86d6063fd6f18455 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e8a684d213d04392abfecfa4c8fbccd8 : SUCCESS in 14m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c86ec028ceac49c2965a3382c80860c7 : SUCCESS in 9m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/859fb877393c45889f462dfc337fc445 : SUCCESS in 4m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fca464dc3cbc4c4998200cac0994f5f1 : FAILURE in 19m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b8fa76fc57b433fb21802afeff86efd : SUCCESS in 9m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/da13f0bbfb234920a69c565766041862 : SUCCESS in 31m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cfdcb605f3754b8c9fa3a498943877b4 : SUCCESS in 48m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e7a5f97cfeca4e8db51ce8ed5478453e : SUCCESS in 49m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fd54b96af07418e83ce89cc6a8d3bfe : SUCCESS in 22m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/467bb5a9f70549fe9e091263f4e7f152 : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a724fa0ffec46bab4039fc3155b10e9 : TIMED_OUT in 1h 31m 18s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89ecc0b10df1418d9e7173e77a9df41e : SUCCESS in 22m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29abd166366a44f78ad619421c14d237 : SUCCESS in 40m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/60f105abbfc54c57922d80025403c42f : SUCCESS in 58m 14s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c894f5c782f4f418730390feb6d6a96 : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":4},{"id":"13519ffe030ea927132b7427ab893778d1b9605a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-10 10:21:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2cc96f7620db7be2adb7628a79f52c7cd2c74e10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:22:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"71fe6215e10fae42b9cbe290c9e6f2294fc71b6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 11:37:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afafa4c68cb24f2a95f4e87e58ed2f9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb41b2a5817479ea27deeeef10ec409 : SUCCESS in 10m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ce9118b5f31c4735a2432f73c40a2824 : SUCCESS in 10m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e69c1c02cb934090817b3adbeb784e64 : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0332aa57ac644318855c74b9bf80b73e : SUCCESS in 8m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/303802861956454885344d3fc929eaca : SUCCESS in 12m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/32f1b3d040514fc3b6f7ff7160846d81 : SUCCESS in 15m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f6e0391ecae04ee285675fd3b6cd2650 : SUCCESS in 12m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eb083f3215b849528820aa6478bed0f5 : SUCCESS in 21m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/041ffcabfc8e4784b38d125cfa2c3957 : SUCCESS in 22m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/76a9d28867f743ceb6f85da24f5edda2 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/887515d0369a425b82ed88eee0c48ec2 : SUCCESS in 14m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9f0231f295554776abf1a1a1c3095250 : SUCCESS in 7m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/91cd403b9bb841b69d7887ed1c444241 : SUCCESS in 5m 03s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4b1ec3d3c533465e99ec8478ae49f8b5 : SUCCESS in 6m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f6d57f529fa4f478af47b4828b9813b : SUCCESS in 43m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0855b0b70540455c81292c4f6aa6786d : SUCCESS in 26m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cb3ee9f9135d47dc9861b17a05e0304c : SUCCESS in 26m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/969ffdacefa44f81b3c595bd1a29ed1d : SUCCESS in 20m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fafb12f569c947e2837923949bd2e480 : SUCCESS in 4m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/91fa80d39eaa42798e1a06188477f737 : SUCCESS in 33m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efc9635f2aea4b58b67de5be48d0f8f7 : SUCCESS in 45m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6eaf93a1b56542a4b09b0faa302d3910 : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":4},{"id":"fd746e60bd2213962cba46b6dd21aa4044e181ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 11:37:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"df501991cd0bd3f8c1143ebfb5e2dd7fb87c22ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 11:40:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/885d7abcd87c4075a7f65a0889ec8926\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa432cf362c4a77a653a2b481a3ba2b : SUCCESS in 1m 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/09d3aab0c5ae42b0934d797d12c3bb5b : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/aedd36955cf74b0ea931ca0353b0e2a1 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"846298cd1e87a207c4ac7024f75b48d663f33479","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-12 04:01:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0d7ba76f4a70bb8769e94fba99e3798d00b1a043","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 05:39:42.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c8720bf38534fafa696559c02d66421\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/974db3b821484a7082907bc16ca5f946 : RETRY_LIMIT in 9s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1ccc16254a8646eeb0636f14cbbafbc3 : SUCCESS in 7m 13s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7eb92441a54445f28d478a26cab8fd2f : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffb2d5fe6cc644c4844c3fdb4f1db0ad : SUCCESS in 8m 08s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/68477586dfc045e6aa9e3e1db5eb0eb4 : SUCCESS in 15m 46s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/d659af51edd4456ba58c0ed6ce825980 : TIMED_OUT in 1h 30m 58s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/4db8682c8b41492c82801caaf4629e5c : TIMED_OUT in 1h 31m 00s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fc6a0494266df3a480b49e64da83cb1430191a2d","revisions":{"bb9c9cfbd313b86a9e88687b6b8feef4ac008e29":{"kind":"REWORK","_number":1,"created":"2025-09-04 08:46:31.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/19/959419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/959419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/959419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/959419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/959419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/959419/1"}}},"commit":{"parents":[{"commit":"fb24f068730d57149b02d175f0919314380e04ff","subject":"Merge \"[CI] Use defined version of probetest job\" into unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb24f068730d57149b02d175f0919314380e04ff"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-19 16:43:45.000000000","tz":-420},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-04 08:44:38.000000000","tz":120},"subject":"CI: Fix py27/py36/py37 jobs","message":"CI: Fix py27/py36/py37 jobs\n\nStop using bionic for py27 and py37; we can use pyenv to install it on\njammy.\n\nThe py36 job still needs bionic; pin ansible to version 9 now that 11\nis the default.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I83e3bc609433995875e47c4103ca945154088a65\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nRelated-Change: Ida8be4477116d3b44e3d7b470a183d2298e2d3d5\nRelated-Change: I16af4fc74974979bdf68390b42b2f571a8183a05\n(cherry picked from commit 972baf30df79174565b34e8d0081854816e5fc82)\n(cherry picked from commit dc98a8463506ac520c469adb0ef47d0f7753905a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb9c9cfbd313b86a9e88687b6b8feef4ac008e29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb9c9cfbd313b86a9e88687b6b8feef4ac008e29"}]},"branch":"refs/heads/unmaintained/zed"},"67387bb2ab50ad2b1cad7ed38962998a8d4e8ed8":{"kind":"REWORK","_number":2,"created":"2025-09-04 09:45:11.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/19/959419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/959419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/959419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/959419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/959419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/959419/2"}}},"commit":{"parents":[{"commit":"fb24f068730d57149b02d175f0919314380e04ff","subject":"Merge \"[CI] Use defined version of probetest job\" into unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb24f068730d57149b02d175f0919314380e04ff"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-19 16:43:45.000000000","tz":-420},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-04 09:44:54.000000000","tz":120},"subject":"CI: Fix py27/py36/py37 jobs","message":"CI: Fix py27/py36/py37 jobs\n\nStop using bionic for py27 and py37; we can use pyenv to install it on\njammy.\n\nThe py36 job still needs bionic; pin ansible to version 9 now that 11\nis the default.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I83e3bc609433995875e47c4103ca945154088a65\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nRelated-Change: Ida8be4477116d3b44e3d7b470a183d2298e2d3d5\nRelated-Change: I16af4fc74974979bdf68390b42b2f571a8183a05\n(cherry picked from commit 972baf30df79174565b34e8d0081854816e5fc82)\n(cherry picked from commit dc98a8463506ac520c469adb0ef47d0f7753905a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67387bb2ab50ad2b1cad7ed38962998a8d4e8ed8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67387bb2ab50ad2b1cad7ed38962998a8d4e8ed8"}]},"branch":"refs/heads/unmaintained/zed"},"86a87840d582757b94bd36d8c2e7aa3baa8657e9":{"kind":"REWORK","_number":3,"created":"2025-09-04 09:46:55.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/19/959419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/959419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/959419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/959419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/959419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/959419/3"}}},"commit":{"parents":[{"commit":"fb24f068730d57149b02d175f0919314380e04ff","subject":"Merge \"[CI] Use defined version of probetest job\" into unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb24f068730d57149b02d175f0919314380e04ff"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-19 16:43:45.000000000","tz":-420},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-04 09:46:49.000000000","tz":120},"subject":"CI: Fix py27/py36/py37 jobs","message":"CI: Fix py27/py36/py37 jobs\n\nStop using bionic for py27 and py37; we can use pyenv to install it on\njammy.\n\nThe py36 job still needs bionic; pin ansible to version 9 now that 11\nis the default.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I83e3bc609433995875e47c4103ca945154088a65\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nRelated-Change: Ida8be4477116d3b44e3d7b470a183d2298e2d3d5\nRelated-Change: I16af4fc74974979bdf68390b42b2f571a8183a05\n(cherry picked from commit 972baf30df79174565b34e8d0081854816e5fc82)\n(cherry picked from commit dc98a8463506ac520c469adb0ef47d0f7753905a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86a87840d582757b94bd36d8c2e7aa3baa8657e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86a87840d582757b94bd36d8c2e7aa3baa8657e9"}]},"branch":"refs/heads/unmaintained/zed"},"fc6a0494266df3a480b49e64da83cb1430191a2d":{"kind":"REWORK","_number":4,"created":"2025-09-04 10:09:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/19/959419/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/959419/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/959419/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/959419/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/959419/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/959419/4"}}},"commit":{"parents":[{"commit":"fb24f068730d57149b02d175f0919314380e04ff","subject":"Merge \"[CI] Use defined version of probetest job\" into unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb24f068730d57149b02d175f0919314380e04ff"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-19 16:43:45.000000000","tz":-420},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-04 10:09:08.000000000","tz":120},"subject":"CI: Fix py27/py36/py37 jobs","message":"CI: Fix py27/py36/py37 jobs\n\nStop using bionic for py27 and py37; we can use pyenv to install it on\njammy.\n\nThe py36 job still needs bionic; pin ansible to version 9 now that 11\nis the default.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I83e3bc609433995875e47c4103ca945154088a65\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nRelated-Change: Ida8be4477116d3b44e3d7b470a183d2298e2d3d5\nRelated-Change: I16af4fc74974979bdf68390b42b2f571a8183a05\n(cherry picked from commit 972baf30df79174565b34e8d0081854816e5fc82)\n(cherry picked from commit dc98a8463506ac520c469adb0ef47d0f7753905a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc6a0494266df3a480b49e64da83cb1430191a2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc6a0494266df3a480b49e64da83cb1430191a2d"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
