)]}'
{"id":"openstack%2Fswift~959609","triplet_id":"openstack%2Fswift~master~I19d21704c5808502cc64ffaec24d1765fd0e6fa8","project":"openstack/swift","branch":"master","topic":"comn-db-log","attention_set":{},"removed_from_attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-09-30 00:00:42.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I19d21704c5808502cc64ffaec24d1765fd0e6fa8","subject":"sq: self.broker_logger to container/sharder","status":"ABANDONED","created":"2025-09-04 21:33:58.000000000","updated":"2025-10-02 20:05:14.000000000","total_comment_count":2,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"79eea54421c6c620634bbccbc10b2a79bb16bb72","_number":959609,"virtual_id_number":959609,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-30 01:39:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-09-04 22:41:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 15:51:59.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"9675b9d07c7e2722f06ac6d9f974672ff08dc75f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-04 21:33:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4af95defa4bd145abd76f367efc239ebaeeadd2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-04 21:34:33.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7d290dd9747b66b14141b3b6d46ef711fdd10c26","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-04 21:36:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"845e3745797a7b7df976e71f3cd02fead7793801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 22:41:21.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b718a2fd8d4b406581c8a4b39163ff2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3959e2aef0046dab0d1b0d5a9fdc7ff : SUCCESS in 7m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/e786c13b97534191b8cb2a594b94fdfd : SUCCESS in 45m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c983e87c54094cf5a3fedd085ca514dd : SUCCESS in 58m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd6e94270ea04c91b26bef8fd75a2710 : SUCCESS in 39m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6714bfcb533449d5a434087116c9ef28 : SUCCESS in 1h 04m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ee431c46a8f47a0aeb8babfbd296c99 : SUCCESS in 17m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3ee77ad21b94492b46cd998b409ec2e : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba143db50e374decbdc8606eec92085f : SUCCESS in 19m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4e4ba62f12e74dc5a1fdf69e3bc2a7ba : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fec7d1cb26bd47c1b85569e8a49d9bbb : FAILURE in 13m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/63de536ee4764e018d6478c1f363d420 : FAILURE in 24m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d4bebddd322440ea8435fa28e5e15e8a : FAILURE in 20m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c05ae265a70a457fa306f41fd5ec0e1a : FAILURE in 22m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b86ef1da1bf94747b2bcc851925bf1cf : SUCCESS in 9m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/11d5481db6d04d7381db0623e48a18f2 : SUCCESS in 9m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0bb3a0836f5242ada4ce8649f503aba1 : SUCCESS in 9m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9eeff6fbdce845ef9b60b76042a3ca9f : SUCCESS in 8m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9b5da1f5e0fd440da5144b9777d6fcbe : SUCCESS in 14m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/eed376e7df4d4278a9bfa897e47410eb : SUCCESS in 7m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/769c6ae7a2b944cea0232d80ca60a0ed : FAILURE in 58m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e71a3f9fa4c44d0ab9184e2f32b8ad3 : SUCCESS in 49m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14ddf931ca854483bb264da538a87938 : SUCCESS in 50m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d62778459a9047e3adc4b7d716655ff3 : FAILURE in 27m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f3ba8dcbad14204bcaa995f3462dc00 : FAILURE in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e95ddc86d5c4e7f97e42ea434d46b93 : SUCCESS in 9m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ce52bd61125422a85ad36ff11f5d2f8 : SUCCESS in 48m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2916dfc5230a40ec9adbad01e5a85976 : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":3},{"id":"05239da9a4557d7703443af8a28a3c14ba61f3ec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-04 22:45:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"45140246d8390fafdb5bdac2ae07dd737d91f0b0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-04 22:46:38.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"3b8ba9b3a8f97d9861a09c5d1e51e88dae7f8408","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 23:49:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11b2c3f6d9b8431d86dcec8ca72f712c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/83ff1a1e9aa94c76b4f12362ac46515e : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":5},{"id":"45701a05f9e8f8aeb94c7f3c308e4e3e2a4df660","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-04 23:55:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e0ab111acf837e043184560650045e96106aa389","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-04 23:56:16.000000000","message":"Patch Set 6: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cc657e3c634f6915bf5a8dedbe7d9f960198dd5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 00:47:48.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51ee6dc3414542078d03d3211d5befa8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6385cfd1d773452ab48434835bf08900 : FAILURE in 51m 02s","accounts_in_message":[],"_revision_number":6},{"id":"2c2047be9966cbdb8a430f6521cabc6ded798173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 01:02:30.000000000","message":"Patch Set 6: Verified-1\n\n(99 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eceb810b052d44dc9797e2b6d8b51020\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe323eee0aa4ae69ae5de6d05e1e6ca : SUCCESS in 10m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/1060a92a835e413ba6dfb17fc9f5d5e2 : SUCCESS in 1h 00m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d343168fd66e4f31a1c0336f6ed507bd : SUCCESS in 1h 01m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/de7a2e2bc1144121a19a5d1cf2876590 : SUCCESS in 56m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61f0cfec71684a21800a5a7ce690660a : SUCCESS in 1h 04m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c7c3da4f91d4625a7064f8af5c58bc3 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac117fe390354dbc8ea8187f6e061a79 : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c015f14a62d48e690c81e1aeb45d241 : SUCCESS in 20m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0fcfb1b7816d4d3c88c198cebb638efc : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c58be8f2f55a4c65a619b4221f1529ac : FAILURE in 20m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a55a01af33334995ba66b40b1cdc7090 : FAILURE in 9m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d503d36bf8b84076b5fa61f02f087776 : FAILURE in 24m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3312bbf2305f45c59ff46703cc9a0b60 : FAILURE in 25m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f7d8534ffd864c2181f2a4c13c6d2917 : SUCCESS in 11m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/406446f4766f485fa1e4a439f1ca1be7 : SUCCESS in 15m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ca051bfb099f400bacc73bf359140398 : SUCCESS in 8m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/74f50aca8cf0487e8695115dc310b36e : SUCCESS in 15m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5a06bf24b2f14ed994ae467a1ab2d88c : SUCCESS in 15m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/384e9ddb4ed248d69d44aabede2cc973 : SUCCESS in 11m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/99020aaf71fa454998efafffe3437d3b : FAILURE in 55m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/875e73a414084cc1a75afb5829191d3b : SUCCESS in 51m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77bb1ce4fef24034ae343e17439d5e9f : SUCCESS in 39m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09acb2c11fec4ba691293f4130a20687 : TIMED_OUT in 40m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf389c6d8954614be13452302b73148 : FAILURE in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2851cd0b376d42988d3d7951cdf9e0d9 : SUCCESS in 10m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9965789508f54069a1315c21e649fa20 : SUCCESS in 52m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0e81a7cf91649de828f28180f24a216 : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":6},{"id":"21f5ba61be0f96291cf7c05dc16f0ddd6cbd791b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-16 20:04:10.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"befacc54f26cb9beed395e24eaf08f867fe26497","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-16 21:04:14.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1a5b1680c9c5c4b4c176e36fec77a303f296d19e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-16 21:43:15.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"d59bc41a490a757bfb1ef920700fa7fb50570e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 23:03:31.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab68ce159b284a4a8aa57c8aeaf4d4cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cde8225d8c1e4acfa44c1cc3eb982399 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/e6b3f7d6754940fbbcef598cacaf81d2 : SUCCESS in 1h 00m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b059a25d0bbc49ef99dfad76a9028c89 : SUCCESS in 44m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aca4f2f61c0b4e7f813b185dfa1be940 : SUCCESS in 55m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da7ae055a52446d8b42cc226db207745 : SUCCESS in 58m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8a0151f3cfa45d8b10e169ce8f9426b : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a43db36d36c24f7e95f768fe6e9c9a31 : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e53fff8160344a428f0ef3b920d275c6 : SUCCESS in 16m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f7f3308360d74d7888d0525d9989f9e0 : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/22fb559cb67f4cecbbb66df64e44df6d : FAILURE in 20m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a143f17ae531479ea090dd82719d6631 : FAILURE in 17m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/be639cb70b374999a779a2b62d9a071a : FAILURE in 23m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/baac9bab92bc432b962036938b97f1e1 : FAILURE in 17m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/63276988c90641ec8a3291a672fd3c4c : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/246facd7fb3d44a68ef303bf69fc2689 : SUCCESS in 13m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0d11e95afec64d938cc165c5160cb1b4 : SUCCESS in 12m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9137aedd3e54487fb4d3bbd5af6a818c : SUCCESS in 8m 17s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/90f0ab0f0bad4e9bbda2d0006bee20d1 : SUCCESS in 13m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ca88742ef6f84f0b81052caae6bf1460 : SUCCESS in 7m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0b9269cf10446038dfb490f2b9020c1 : SUCCESS in 48m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2179b2017fbc443da34488e113b4a2ef : SUCCESS in 34m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0789f84bb4c0424597cc26377f9dab28 : SUCCESS in 34m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e5f5da556b24411989aeb2042fbb2da : FAILURE in 26m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac6ef962ca8e41f89bced8193f571159 : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/887dba662e3144c9825c4b68f7e6c22a : SUCCESS in 18m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d1396ea7ee84ae1b3d4201921b682aa : SUCCESS in 51m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/839fe3e3e77a426e8d985d05e818f643 : SUCCESS in 43m 18s","accounts_in_message":[],"_revision_number":9},{"id":"1cbb861240abadd35389816165b782ee773ea58d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 23:06:11.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3c453b7741b42a2b417f4c7a5e7bc63\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/43e5d45b422d4748b699033ce58f8de8 : FAILURE in 1h 20m 24s","accounts_in_message":[],"_revision_number":9},{"id":"c72aa0514ab6fdd6bb361761c0dee0aedb72a4a2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-23 15:51:58.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"0d32b886d1d701dfb7aa7ba19a78de27f325b6df","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-23 16:04:21.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"e03dcc6e06ab9e633c54d5ecf34c529b0c73edf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 17:15:03.000000000","message":"Patch Set 11: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af954eeca5e1422b9f195b4007bc8db9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d2c5584748944bea210c500448f379c : SUCCESS in 6m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/1332180559784a3a9379771fc6db3231 : SUCCESS in 33m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3439bf2b560f466ba71d5acaaafbffe3 : SUCCESS in 1h 05m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/143324031aa04b10a71157a29ec54864 : SUCCESS in 37m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6e8e54be2dd40da8d3a0230e532a633 : SUCCESS in 59m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea25904dd7f04610af4966a02315f2fb : SUCCESS in 25m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0c404e00a5d417483b174f6bad5fc1d : SUCCESS in 14m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0972002ac1cb4a8cb19f48d529335f1a : SUCCESS in 22m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be3a13872aa9411e9871de97a8e0268c : SUCCESS in 9m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b43e215f60bd44a59546e1f76b3fe2e0 : FAILURE in 24m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d82b35b0edd44264adedbe6df719b073 : FAILURE in 27m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d5d26996ad8f44b985ffaee82000c873 : FAILURE in 29m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bd98683e61b44e329dda70aa9708e349 : FAILURE in 20m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/de0ebaea6387492d831c1d3c9ee3b6a0 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1cee19afb54249fc86ea540c40b904cd : SUCCESS in 9m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fa4d39d4bfb949949a9fae28e7f71652 : SUCCESS in 16m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e44190e9748402fb5568a301608bbcf : SUCCESS in 6m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c9d934d204314216b2e79794d6b5032d : SUCCESS in 10m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/59196ed250d44ed5b154b94bbc32378a : SUCCESS in 7m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ce0096c4acc4a6dab6c1f5208733f9f : FAILURE in 57m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/673808955905435c84d874c54ffc3913 : SUCCESS in 56m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78b2ae8862ef41ecba9706ea0397b207 : SUCCESS in 56m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77895a084b894d32acb48ec686641682 : FAILURE in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa6e2b015ead4d5ba988342dbb22b1d1 : FAILURE in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e32dd4624a94acb9707607d4951d93a : SUCCESS in 16m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e412a213b7674558a1a414364a064949 : SUCCESS in 57m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a486b31a03d54384930796e6d9538450 : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":11},{"id":"5ef1e8aa18fdab18f53eb77c20991d02d9013861","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 18:00:45.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9cc3009a5154badaa835ff4e3552730\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba0905d923dd438f8368def0e8f2ec46 : FAILURE in 1h 05m 33s","accounts_in_message":[],"_revision_number":11},{"id":"27198e41c14558eda4213643a5d0e7fbdfc29f0c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-28 22:26:24.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"c9dc6e85249fe45463b2e42ef400fb5879803e71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 23:23:12.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/247a17637c7949b8b3bdc6fa6c758e3b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/142163c29c3f41fe9f886c7431f8e26f : FAILURE in 55m 01s","accounts_in_message":[],"_revision_number":12},{"id":"14b1ca674744374b8e91146abfedc6a3575f4ef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 23:36:07.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c337484b068647bf88f53114c628c39a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd7883e88e944315ab492d74fa0d6e28 : SUCCESS in 4m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/9a79ba25213e4774b5fe36054e3ab70d : SUCCESS in 59m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d28bc80c03484b4d96031787753a0a90 : SUCCESS in 58m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/226efc331bdb455582a2bfc1bc1afc6c : SUCCESS in 28m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/953e26d58bf3477bae465a4e1878ff4e : SUCCESS in 1h 04m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb1f9543be5c4a02bd593881f502071e : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/038ecb6e8b4f4f33998ef4a4a3cad25a : SUCCESS in 23m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c397935771304d73ae6b9d52d33863f1 : SUCCESS in 15m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c8e14fd64ae490ea5ffd7244467d7a7 : SUCCESS in 10m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d8c2bf113a5342dd86f107036e861933 : FAILURE in 24m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/983cd37321a84a8fa51dff000cf91d1a : FAILURE in 9m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1c2ffdc1a61640f7b1711d1af867a3fe : FAILURE in 12m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d9f243eb3ae04b08b65724b16fd04d5a : FAILURE in 20m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/be77cb240a98427ba39705d2a6f87007 : SUCCESS in 7m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/605a05038b81431198d409dd5b6f28d0 : SUCCESS in 12m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9dfa349d830144368b3f15f870170495 : SUCCESS in 9m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e49eedb2329f41aaa7a39c0b55bdcd41 : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e685f80180054a95b7c5a4f866412b3a : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/50d998af409745e39f172729b0be4173 : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec4041bc23084beda91a1d35c3c85a97 : FAILURE in 34m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f0260d1e121648a4946e7fda1f056a24 : SUCCESS in 25m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/049e5814f0ec4ddcbb66b265bd2c22e8 : SUCCESS in 56m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d2e1bbbe6524c17834370f2363cf4b1 : FAILURE in 21m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/580317439f624ad98542ce4f339d1487 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0830e27801f4f41a3b19315b4be5530 : SUCCESS in 15m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0cf8fcf6bcf40aa8b613d22dab42b68 : SUCCESS in 31m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a2b1f40cbbc43aa935cec7f78c281d0 : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":12},{"id":"f10c4c373ea1a67fb1f6be66beb2a1e47e7169fc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-29 15:51:59.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":12},{"id":"e9d842d9a11f08e477494a89ace5366ba80712e3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-09-30 00:00:42.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/db_replicator.py\n* test/unit/common/test_db_replicator.py\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"4cfcb33c1135c9020254c06402262b742d17f90d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 00:11:41.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d183c156acde4e3eafdda29f0a2b730e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/677e616a9734471b90087637c7d6a8c1 : FAILURE in 9m 04s","accounts_in_message":[],"_revision_number":13},{"id":"cd3311209f4b0d90903d8fdab5236e6571e1d677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 01:39:32.000000000","message":"Patch Set 13: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2ef28e2d0e240f3bb36f9817f280d67\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07b16398d714a9697f2fa794755b51b : FAILURE in 7m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/6d22bfa0cbc0448aae0da44690c8c66a : FAILURE in 45m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de47b873fa8e42fdac1b4d5d8be17b8e : FAILURE in 50m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b302406336e64191b80220c6bf87f439 : FAILURE in 24m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17785947e4c64714a00e824fdcb36475 : POST_FAILURE in 27m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76f74a1e9d5c4bccb62ba97ab46e95dc : FAILURE in 7m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17ffec22655e405e9c0c2dce1e78a27c : FAILURE in 6m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56b26f08d5a043579b6180d33d8def07 : FAILURE in 6m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dbdebec4c39e49199ab7d4e450942f6b : SUCCESS in 8m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a13e2b3c2ed54b47b9c047b7882e7945 : FAILURE in 8m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a7e6bd88a9f641069e3c73a3ff7a26d3 : FAILURE in 4m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3ba081a9902c4d6c86e6e38940c055f4 : FAILURE in 8m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0d49ca1fd5ed4766b80bd64fec9d47e0 : FAILURE in 7m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/59c1b7fa08fe40d29fda670d3b4aa3e8 : FAILURE in 4m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dd903f8f7db64169943a7a4f50794a64 : FAILURE in 3m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/93b48609035049b9b4905925123fa4e3 : FAILURE in 2m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/70bd4c6a6b3a4f4db91c1f327073dd66 : FAILURE in 6m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f82efb2613f6458db8770e389ee9f60c : RETRY_LIMIT in 4m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f14f090dc8d049099a9469428ca40057 : RETRY_LIMIT in 3m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/368cbd2f6622446c9afdf03d13108fe8 : FAILURE in 3m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f299c689db84fe18f7cb9687b1edf90 : FAILURE in 30m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c0e22b67412249469e8aa41c440a3c4c : FAILURE in 23m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e90cb386d99a4f3eb0bea1b595232263 : FAILURE in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/557f9228eb884af2826f7b8858a524d0 : FAILURE in 2m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b505a3cba834f84a54842e5f2cb1622 : TIMED_OUT in 1h 30m 58s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fbbdcbfa3a740fc987eca03f778980d : FAILURE in 23m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/983082cf533e469cbbe261b0a7a4d8e2 : FAILURE in 17m 56s","accounts_in_message":[],"_revision_number":13},{"id":"79eea54421c6c620634bbccbc10b2a79bb16bb72","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-02 20:05:14.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/933379","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"c4676ca566962fb15b990c2463a53e1f7d5263b5","revisions":{"a0f28924c1c5b0395b4e66b93c6a5503796d36ef":{"kind":"REWORK","_number":1,"created":"2025-09-04 21:33:58.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/1"}}},"commit":{"parents":[{"commit":"27db680ae44290d8715f127777def65a76012fd2","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27db680ae44290d8715f127777def65a76012fd2"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:33:32.000000000","tz":-420},"subject":"sq: refactor _replicate_object to accept broker","message":"sq: refactor _replicate_object to accept broker\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0f28924c1c5b0395b4e66b93c6a5503796d36ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0f28924c1c5b0395b4e66b93c6a5503796d36ef"}]},"branch":"refs/heads/master"},"6356a7589f51fa779a4eea25d9f270b07f2e5a72":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-09-04 21:34:33.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/2"}}},"commit":{"parents":[{"commit":"27db680ae44290d8715f127777def65a76012fd2","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27db680ae44290d8715f127777def65a76012fd2"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-04 21:34:33.000000000","tz":0},"subject":"sq: refactor _replicate_object to accept broker, also adds path property to AccountBroker","message":"sq: refactor _replicate_object to accept broker, also\nadds path property to AccountBroker\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6356a7589f51fa779a4eea25d9f270b07f2e5a72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6356a7589f51fa779a4eea25d9f270b07f2e5a72"}]},"branch":"refs/heads/master","description":"Edit commit message"},"500eb2c7efd14353eda88d40df5f0c90954897a4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-09-04 21:36:30.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/3"}}},"commit":{"parents":[{"commit":"27db680ae44290d8715f127777def65a76012fd2","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27db680ae44290d8715f127777def65a76012fd2"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-04 21:36:30.000000000","tz":0},"subject":"sq: refactor _replicate_object to accept broker","message":"sq: refactor _replicate_object to accept broker\n\nThis patch also adds path property to AccountBroker\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/500eb2c7efd14353eda88d40df5f0c90954897a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/500eb2c7efd14353eda88d40df5f0c90954897a4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"aaf79662116fe580c8b553bde3420104f49811a0":{"kind":"REWORK","_number":4,"created":"2025-09-04 22:45:58.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/4"}}},"commit":{"parents":[{"commit":"27db680ae44290d8715f127777def65a76012fd2","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27db680ae44290d8715f127777def65a76012fd2"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 22:45:46.000000000","tz":-420},"subject":"sq: refactor _replicate_object to accept broker","message":"sq: refactor _replicate_object to accept broker\n\nThis patch also adds path property to AccountBroker\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aaf79662116fe580c8b553bde3420104f49811a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aaf79662116fe580c8b553bde3420104f49811a0"}]},"branch":"refs/heads/master"},"eadde0092d9d99ab241a2731df56d026132f9e19":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-09-04 22:46:38.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/5"}}},"commit":{"parents":[{"commit":"27db680ae44290d8715f127777def65a76012fd2","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27db680ae44290d8715f127777def65a76012fd2"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-04 22:46:38.000000000","tz":0},"subject":"sq: adds path property to AccountBroker","message":"sq: adds path property to AccountBroker\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eadde0092d9d99ab241a2731df56d026132f9e19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eadde0092d9d99ab241a2731df56d026132f9e19"}]},"branch":"refs/heads/master","description":"Edit commit message"},"66dee9ebb17a502d3605c04d4ad1ca2cdfa65bfa":{"kind":"REWORK","_number":6,"created":"2025-09-04 23:55:07.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/6"}}},"commit":{"parents":[{"commit":"27db680ae44290d8715f127777def65a76012fd2","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27db680ae44290d8715f127777def65a76012fd2"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 23:53:15.000000000","tz":-420},"subject":"sq: adds path property to AccountBroker , self.broker_logger to container/sharder","message":"sq: adds path property to AccountBroker , self.broker_logger to container/sharder\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66dee9ebb17a502d3605c04d4ad1ca2cdfa65bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66dee9ebb17a502d3605c04d4ad1ca2cdfa65bfa"}]},"branch":"refs/heads/master"},"ac344951f1fb89f9b30372056ac62ffc96a8b4f8":{"kind":"REWORK","_number":7,"created":"2025-09-16 20:04:10.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/7"}}},"commit":{"parents":[{"commit":"4be903c5763f71af495fd53e858e6b7b1c99557b","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4be903c5763f71af495fd53e858e6b7b1c99557b"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-16 20:03:04.000000000","tz":-420},"subject":"sq: adds path property to AccountBroker , self.broker_logger to container/sharder","message":"sq: adds path property to AccountBroker , self.broker_logger to container/sharder\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac344951f1fb89f9b30372056ac62ffc96a8b4f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac344951f1fb89f9b30372056ac62ffc96a8b4f8"}]},"branch":"refs/heads/master"},"7139c9e576800d7c2acb82a2c6d0d188071721d6":{"kind":"REWORK","_number":8,"created":"2025-09-16 21:04:14.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/8"}}},"commit":{"parents":[{"commit":"4be903c5763f71af495fd53e858e6b7b1c99557b","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4be903c5763f71af495fd53e858e6b7b1c99557b"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-16 21:03:57.000000000","tz":-420},"subject":"sq: adds path property to AccountBroker , self.broker_logger to container/sharder","message":"sq: adds path property to AccountBroker , self.broker_logger to container/sharder\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7139c9e576800d7c2acb82a2c6d0d188071721d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7139c9e576800d7c2acb82a2c6d0d188071721d6"}]},"branch":"refs/heads/master"},"0c8955f1ac58d07cdc15b05eb12e972b90ded8cf":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2025-09-16 21:43:15.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/9"}}},"commit":{"parents":[{"commit":"4be903c5763f71af495fd53e858e6b7b1c99557b","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4be903c5763f71af495fd53e858e6b7b1c99557b"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-16 21:43:15.000000000","tz":0},"subject":"sq: self.broker_logger to container/sharder","message":"sq: self.broker_logger to container/sharder\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c8955f1ac58d07cdc15b05eb12e972b90ded8cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c8955f1ac58d07cdc15b05eb12e972b90ded8cf"}]},"branch":"refs/heads/master","description":"Edit commit message"},"90a4cf6fa8fe1a8c3637d77f331289f2030b961c":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-09-23 15:51:58.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/10"}}},"commit":{"parents":[{"commit":"cc4358f47bce7a9d5d772e438b3ae7977e442c2c","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc4358f47bce7a9d5d772e438b3ae7977e442c2c"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-23 15:51:58.000000000","tz":0},"subject":"sq: self.broker_logger to container/sharder","message":"sq: self.broker_logger to container/sharder\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90a4cf6fa8fe1a8c3637d77f331289f2030b961c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90a4cf6fa8fe1a8c3637d77f331289f2030b961c"}]},"branch":"refs/heads/master","description":"Rebase"},"ca835f2c7bcd75b0e2e5c76b2d30bf33acce2360":{"kind":"NO_CHANGE","_number":11,"created":"2025-09-23 16:04:21.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/11"}}},"commit":{"parents":[{"commit":"303e17f18236551f7832deb64b66e17478e3ced5","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/303e17f18236551f7832deb64b66e17478e3ced5"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-23 16:04:21.000000000","tz":0},"subject":"sq: self.broker_logger to container/sharder","message":"sq: self.broker_logger to container/sharder\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca835f2c7bcd75b0e2e5c76b2d30bf33acce2360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca835f2c7bcd75b0e2e5c76b2d30bf33acce2360"}]},"branch":"refs/heads/master","description":"Rebase"},"a0dd1fe08f371b2bdbd3934791dd3c28c4697da2":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-09-28 22:26:24.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/12"}}},"commit":{"parents":[{"commit":"e07ce376c74bc5bd85d50ad1359bbb72f9e82d19","subject":"common.db_replicator: include container path in logs, not just db path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e07ce376c74bc5bd85d50ad1359bbb72f9e82d19"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-28 22:26:24.000000000","tz":0},"subject":"sq: self.broker_logger to container/sharder","message":"sq: self.broker_logger to container/sharder\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0dd1fe08f371b2bdbd3934791dd3c28c4697da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0dd1fe08f371b2bdbd3934791dd3c28c4697da2"}]},"branch":"refs/heads/master","description":"Rebase"},"c4676ca566962fb15b990c2463a53e1f7d5263b5":{"kind":"REWORK","_number":13,"created":"2025-09-30 00:00:42.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/959609/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/959609/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/959609/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/959609/13"}}},"commit":{"parents":[{"commit":"777cc77329b3e38f3632b7d4e6172875c77973b1","subject":"common.db_replicator: log container, db paths consistently","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/777cc77329b3e38f3632b7d4e6172875c77973b1"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-09-04 21:17:32.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-09-30 00:00:42.000000000","tz":0},"subject":"sq: self.broker_logger to container/sharder","message":"sq: self.broker_logger to container/sharder\n\nChange-Id: I19d21704c5808502cc64ffaec24d1765fd0e6fa8\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4676ca566962fb15b990c2463a53e1f7d5263b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4676ca566962fb15b990c2463a53e1f7d5263b5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
