)]}'
{"id":"openstack%2Fswift~959851","triplet_id":"openstack%2Fswift~feature%2Fmpu~I8450859546034646288b08c0126f676c736060ec","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-custom-async-updates","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-09-10 10:26:44.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8450859546034646288b08c0126f676c736060ec","subject":"mpu: add systags to container DB object table","status":"ABANDONED","created":"2025-09-05 13:45:55.000000000","updated":"2025-09-10 10:26:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2fdfb194d741e318b44e3fbe15a8f8f4797efbcf","_number":959851,"virtual_id_number":959851,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-05 15:27:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-05 15:27:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3186acb4bb81adfbef8d879edb1327fbbd0600e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-05 13:45:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9bad0bfaa1243f6798dac12a486dab741950d35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 15:27:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a97d233271c5498ea0d2a7cd3c46ace8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b9ba6ee3da24ad6bb101974fa8c40a8 : SUCCESS in 9m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8175053fe2e4a5cb65ef9b00c7469ab : TIMED_OUT in 1h 31m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6435356f022b4b188cdc73b56e1ec7d9 : TIMED_OUT in 1h 31m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56b7679475484ffebe68cc37bbc029fe : TIMED_OUT in 1h 31m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3bdd21a409b24118b713ae83bd5f044f : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d40fca2e24884c94b02b8c210fd1448d : FAILURE in 24m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/26f018febe624e5c87d4fcd94301f572 : FAILURE in 21m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/db072db8cf824e799993d29ac27ab834 : FAILURE in 28m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/21f3017578314c50a880447f62c6afaf : FAILURE in 13m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8e180e7214e444dcb8c4d158f204b76e : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/eaa2a24406f54a8aa1b25290bb0523fd : SUCCESS in 17m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/36cbf506c1d14075b24f470e1b13bca4 : SUCCESS in 17m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6b4249460ffc47f98386f90a44aa4e9e : SUCCESS in 8m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e9714d90b1124676bd828746f83f65b9 : FAILURE in 18m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f64151c2df0843fa81d216a91ebc465a : SUCCESS in 12m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5ab9e7c56bc46d49fc6cce9edd761b9 : FAILURE in 53m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7654cf23b8544547974de166486df450 : FAILURE in 26m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41e31b76941f41b9be3cc4c0dd5bf593 : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7828a4204fb4066bed4a75f416aa49d : SUCCESS in 20m 23s","accounts_in_message":[],"_revision_number":1},{"id":"246b8030d2ff677e61488a25fc70c152563cd53e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 16:22:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adff8fb43ecf4074b1d6c396ecbbba42\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c75668d92374dd3b8e2915a54c465c7 : FAILURE in 1h 18m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2fdfb194d741e318b44e3fbe15a8f8f4797efbcf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-10 10:26:44.000000000","message":"Abandoned\n\nreplaced by 960356: mpu: add systags column to container DB object table |","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"255eb241f848872e625c825fbc6a5ec4a445ff95","revisions":{"255eb241f848872e625c825fbc6a5ec4a445ff95":{"kind":"REWORK","_number":1,"created":"2025-09-05 13:45:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/959851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/959851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/959851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/959851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/959851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/959851/1"}}},"commit":{"parents":[{"commit":"f289d0fd5fd1ac226805560a91b9aa5fb50808fb","subject":"WIP: use version history to detect orphans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f289d0fd5fd1ac226805560a91b9aa5fb50808fb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-09-04 15:08:11.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-09-05 12:41:18.000000000","tz":60},"subject":"mpu: add systags to container DB object table","message":"mpu: add systags to container DB object table\n\nObject update requests with an x-object-sysmeta-systags header will\nhave the header value stored in a new \u0027systags\u0027 column in the\ncontainer DB object table.\n\nCurrently, the only means for a middleware to forward this header with\na container update is via a custom object update.\n\nCurrently, the systags column value is only exposed by the\nContainerBroker interface. The systags field is not included in\nlistings returned for container server GET requests.\n\nChange-Id: I8450859546034646288b08c0126f676c736060ec\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/255eb241f848872e625c825fbc6a5ec4a445ff95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/255eb241f848872e625c825fbc6a5ec4a445ff95"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
