)]}'
{"id":"openstack%2Fswift~959896","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I7a07c10e62d7dc0db657c8d041c043f676c16cdc","project":"openstack/swift","branch":"feature/threaded","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-09-08 16:26:51.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I7a07c10e62d7dc0db657c8d041c043f676c16cdc","subject":"Run account-server under gunicorn by default","status":"ABANDONED","created":"2025-09-05 18:55:55.000000000","updated":"2026-07-06 12:17:10.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7874846784b9e5dbb553b764c47d37e4469b3fb4","_number":959896,"virtual_id_number":959896,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-05 19:58:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-05 19:58:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e2724b4ad11f0465d8643b89476752c5bc5884e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-05 18:55:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca771488f069f0137bd582dadff8270903ccc4e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 19:58:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f4785bd2c06484dbd1e6d0496f814da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53333dc9abc4f28af5a60f29cedee06 : SUCCESS in 10m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7763c3714fe44ecfb917f659a5ec6e98 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5483f59a8ab4c6b9dc89cee58a4b340 : SUCCESS in 10m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68ea3daecca7489cb6859056b6b20f19 : SUCCESS in 21m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/241f371376a24783914270a89edf5d2a : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/88c37cb242c74814a95de9350c7473bc : FAILURE in 26m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a78eafb2a8d4448ab78cd9ca3223f1af : FAILURE in 9m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f1d7a5628fe24360b8d2e365c027add6 : FAILURE in 20m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e39c5be38ff0451eac3619a70a640325 : FAILURE in 11m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d4fe32a4e47d4bdaa877103300d344ee : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/67a07bfa4bb7472c9f210b31c2794f91 : SUCCESS in 12m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/928348be063149b0ba5f97e40e11022a : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e22f57dd6d9c432c8cd3e415cb8e230b : SUCCESS in 6m 08s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d271c5a504744ecbb21318bc3de71d97 : FAILURE in 4m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/80b05c01eb88460fac7e40c1a7331d1c : FAILURE in 18m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1271baab857e491aa6e41914f389b3f0 : SUCCESS in 57m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/553491b697d545cfb8f8a17e8e1464e6 : FAILURE in 21m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c99be5698e844e729a4cd53817d156db : SUCCESS in 3m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/19471036132b4bc2901a09b4a2caf151 : SUCCESS in 19m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0f74d7e607d5a45f78bb719b7903ba3c3c443ccd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 20:17:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/909690c0c5704ff8bca3609d954aaf56\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/116e81e9b6da44918551f5ec8e24c20d : FAILURE in 1h 06m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f94d6e799667c90ab07dc0a925c7bac0979c3da5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-08 16:26:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7874846784b9e5dbb553b764c47d37e4469b3fb4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-06 12:17:10.000000000","message":"Abandoned\n\nSuperseeded by proposed commits","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c675e89d54198f988bef11fa87bd93e6156eade6","revisions":{"c675e89d54198f988bef11fa87bd93e6156eade6":{"kind":"REWORK","_number":1,"created":"2025-09-05 18:55:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/96/959896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/959896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/959896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/959896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/959896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/959896/1"}}},"commit":{"parents":[{"commit":"99c7a8b2c5ef104f0b17c736a9915e918fb10589","subject":"WIP: Add dummy Timeout exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c7a8b2c5ef104f0b17c736a9915e918fb10589"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-09-05 18:55:35.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-09-05 18:55:35.000000000","tz":-420},"subject":"Run account-server under gunicorn by default","message":"Run account-server under gunicorn by default\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7a07c10e62d7dc0db657c8d041c043f676c16cdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c675e89d54198f988bef11fa87bd93e6156eade6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c675e89d54198f988bef11fa87bd93e6156eade6"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
