)]}'
{"id":"openstack%2Fswift~960414","triplet_id":"openstack%2Fswift~master~I2e50ae7695396b44f90462922526dccb44436e45","project":"openstack/swift","branch":"master","topic":"traceback-in-prod","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-09-12 14:38:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2e50ae7695396b44f90462922526dccb44436e45","subject":"sq: test for error in ppi rename","status":"ABANDONED","created":"2025-09-10 19:50:45.000000000","updated":"2025-09-12 14:40:09.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b08a59e76ff3676395fb98b937eacb75f6413d9d","_number":960414,"virtual_id_number":960414,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-10 20:50:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-10 20:41:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-10 20:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-11 03:11:19.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"}],"messages":[{"id":"2ddb2cefd6a98308b3e838f2f45641134ced962c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-10 19:50:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4028e366afd07f16c3123826a39469b0fb01341e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 20:41:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f372fea59253428e9be9128e813166b1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0936fb0fea724101a9edd4bc6919c9d6 : SUCCESS in 49m 30s","accounts_in_message":[],"_revision_number":1},{"id":"628bf312ca5817e39ea5245f75ef504c5708db34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 20:50:12.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71ef76c0f0594a3da2c259ae58e55d37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1bbcb1c278942db8ec833372fd6a916 : SUCCESS in 10m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e4036d11ac44f3296360e4726c39a89 : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0caaf5c6661c44b2a76396393297e3ee : SUCCESS in 10m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f5af67c879a44a6ac697b49786f9efa : SUCCESS in 10m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5ac521566662486eae6c4fb54fc00687 : SUCCESS in 7m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/96592b6cdd6142c08d67c4272b61eb27 : SUCCESS in 10m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5a6e3421c1954396b211465013470340 : SUCCESS in 10m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7b2d4f4c79a44a699e2b0b8094cacf0c : SUCCESS in 17m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/696fe1ad0bc041c19ca2cdc2cb5cf2c5 : SUCCESS in 27m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9282d15f7ef64e3c8cb4b631727a3f46 : SUCCESS in 7m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d6735cbb01064a049e9fb9039bd58de1 : SUCCESS in 14m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ac8747369fe243ecb20e6822c74abcf7 : SUCCESS in 8m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3004cd4aac0414b9f51c7655f0d0bbb : SUCCESS in 49m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/68fd3557b6964ce2aa534933faea98ce : SUCCESS in 54m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2609bd2dbd724915a7b0db5c7e2b13b7 : SUCCESS in 18m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c2e589723d44cfb86ba067f096e6b25 : FAILURE in 3m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b92961ea03314f41ae2b0e8e75c9a129 : SUCCESS in 11m 34s","accounts_in_message":[],"_revision_number":1},{"id":"29cd47ce73c37112b1944895c440789615910703","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-09-11 03:11:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c2e68a5cda1e897ab79784e7e18fb48ef90350b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-12 14:38:34.000000000","message":"Abandoned\n\nsquashed 960405: diskfile: Fix UnboundLocalError during part power increase | https://review.opendev.org/c/openstack/swift/+/960405","accounts_in_message":[],"_revision_number":1},{"id":"b08a59e76ff3676395fb98b937eacb75f6413d9d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-12 14:40:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b811556cbc0254a12e0f187b3e83e6c23442fbf","revisions":{"2b811556cbc0254a12e0f187b3e83e6c23442fbf":{"kind":"REWORK","_number":1,"created":"2025-09-10 19:50:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/960414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/960414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/960414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/960414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/960414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/960414/1"}}},"commit":{"parents":[{"commit":"4a8e8cf9c9d11dbb3ee3ac3285209a1c1dc11c0c","subject":"diskfile: Fix UnboundLocalError during part power increase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a8e8cf9c9d11dbb3ee3ac3285209a1c1dc11c0c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-10 19:50:16.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-10 19:50:36.000000000","tz":-300},"subject":"sq: test for error in ppi rename","message":"sq: test for error in ppi rename\n\nChange-Id: I2e50ae7695396b44f90462922526dccb44436e45\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b811556cbc0254a12e0f187b3e83e6c23442fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b811556cbc0254a12e0f187b3e83e6c23442fbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
