)]}'
{"id":"openstack%2Fswift~960453","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I66127aff3cdd20b6fa2dd756c4b75f3ef0a15288","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-03-06 11:14:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I66127aff3cdd20b6fa2dd756c4b75f3ef0a15288","subject":"Remove eventlet from swift.common.http_protocol","status":"ABANDONED","created":"2025-09-11 10:39:32.000000000","updated":"2026-03-06 11:14:34.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d56d1b062c780310b106a55e2230682fd1b38e6c","_number":960453,"virtual_id_number":960453,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-27 12:12:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-11 15:02:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-10 14:09:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"}],"messages":[{"id":"70a93c366bb0e3f17640d4330a0e32cc8739cf6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-11 10:39:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b52b0b132a0c61d5cf795caa575c15f7e13742f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-11 11:34:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16fec3b70504ae42d0fbf13f11b350d2cd99da58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-11 11:59:16.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c6bb6c8424c48200e2c652527588c4259aa46b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 15:02:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6425e6c1a9264caa8bacff6ddf225145\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25253b02821c4f5780b8ef8893fcf398 : SUCCESS in 9m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1540de635eb4140bfb2bd922be445bc : SUCCESS in 20m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e6a856c5a6c4e20acfa050fd221a0dd : SUCCESS in 17m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/acc3b744633f48038f53db37a65fa45c : SUCCESS in 10m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2defec9e5bec4717bac542f744eb43b0 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b5e0c80105ea4405aba7a2e79a9f9d8a : TIMED_OUT in 1h 30m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ffd7391f3c9549c2998ac1079b086e6b : TIMED_OUT in 1h 30m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/63f1af30b2ce418bb7510c4c8837a057 : TIMED_OUT in 1h 30m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/deb6af26fbe142afba35233c34122680 : TIMED_OUT in 1h 30m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6a734f511cb647a6b52b194e3860815e : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b66a05de0523441a828d495b322c8577 : SUCCESS in 13m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/78f9b9d052b14439a80dd3e525e07095 : SUCCESS in 14m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e933e5fd70554a3388d7e642633bfae4 : SUCCESS in 6m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/efc07a0fd72b40ff93b8a55f3b163989 : SUCCESS in 12m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac44d8144ad14d32a01b8ad1e1372319 : SUCCESS in 11m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2b291e0d8244a3fae7d0afaa4a7c932 : SUCCESS in 53m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6083b3b0a0f94e5cbc4f6d34944d417f : TIMED_OUT in 40m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40f34c28504345738a6779f3dcfd7ab8 : SUCCESS in 3m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9f3b2724b4574cd290cc22515a2e0a0d : SUCCESS in 25m 16s","accounts_in_message":[],"_revision_number":3},{"id":"4ab283c9fc6f6898a5192c36e292ecb4b2272d67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 15:08:49.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a08daf26002e477997a60cdb0a0a8f8f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64adc14904bd4479a07d52f77e30483b : TIMED_OUT in 3h 00m 59s","accounts_in_message":[],"_revision_number":3},{"id":"36eb9c3946e0c1c34bfa10c0894ababe357b286a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-11 17:01:12.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ea374a318f332589fc8b04913577d01bddad007b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 17:59:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f18d56769b8487fa414220873c49a58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10db9ae015ec483a8c97a40090bc1a59 : SUCCESS in 11m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6cdd67caa9f4f47b3a906ec170a0e1f : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd2088b33eaf49db9255c5403fc3becb : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a47073c23214d788a9ce0af1f355980 : SUCCESS in 20m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a19a932cf3a84ef9bb7c9e450fc4e679 : SUCCESS in 9m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0330b1638eb543ae9741c954097dc4d7 : SUCCESS in 18m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a7329c7f12dc4c1fa87dfa9a8f0f11f0 : SUCCESS in 26m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9ba2b219a94946f696ed488150f5d1bd : SUCCESS in 19m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/961561c924784be7b49530cdc7498dfe : SUCCESS in 30m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dde561ba348142c2a53a4c3b4724f8d7 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9f026a332fc149a1b5c9c526787fc344 : SUCCESS in 14m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/24783b6399b649cb827db20254e7787f : SUCCESS in 16m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a8f4a6bbca864ec38f9f5b690be3ae17 : SUCCESS in 8m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4b9506b67cf246ffbb6cc2f8432fef99 : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f153f061ddd143f3832ed927026bd6ce : SUCCESS in 8m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4485eff565c4d7dbd2ba9eef36782f5 : SUCCESS in 54m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c3a0ccbaa914ed0bd4c690c5f00f242 : SUCCESS in 19m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7f0651da7d4f179ce23937ee9d705d : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2322063af8ca4faf9f05388a555c8827 : SUCCESS in 18m 02s","accounts_in_message":[],"_revision_number":4},{"id":"11022ffd73e44e80503c1463c4b269eb616e7de6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 18:16:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77194793d9a844c790fc3a628ea0dfde\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a0093c9cbfc24796a41e326a31250fda : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":4},{"id":"36d403408a05877b0965c992cea3080489d19697","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-12 06:56:54.000000000","message":"Topic eventlet-removal-gerrit-functests removed","accounts_in_message":[],"_revision_number":4},{"id":"71702cdc0f1a770e1f138b1cc42953459d1ed61c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-09-12 06:57:00.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":4},{"id":"9ee05f10c905bfc5831307d5269c550ad0975ded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-10-27 11:17:19.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"88bec56eda0a41ee90a97fb263428dcaa9454de0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 12:08:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c42c9554ba2d4280b5e21dd6832483c4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6c95b8108a194a9b8e3ad9cd436d0b4d : SUCCESS in 49m 33s","accounts_in_message":[],"_revision_number":5},{"id":"80a0d9a91f69c153559eaf4685333428c7435bfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 12:12:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/281824bf0ec745e7b847f6210a4e32a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544cd8a5f2ec47c8a5a7b53ca00e2a4b : SUCCESS in 10m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae51dfd129414ad3ad320478ee7d66c9 : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/598811abf7b545b38d95a1e064bdce4a : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee13b501e5984739b37278e282b4dd9a : SUCCESS in 20m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/096b83ad9e484785bf7e066ed76d2e3c : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/558b529f95f94807a83c858c353255c6 : SUCCESS in 23m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/682a2892cb0147bfb06c85b398a554a8 : SUCCESS in 19m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9fd073ba3ff744cba4512ef544e64e25 : SUCCESS in 28m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e1335040f013450c9d213f038ec0bec8 : SUCCESS in 28m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/84616d3c907b475a81ef8bbec98af63e : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/806751e1766a45b69658fa8064fda390 : SUCCESS in 14m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/68092845bbd84e2cbf588f0ff01b8c0b : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c45df042252c4205b8171d784994f07b : SUCCESS in 4m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0090a7a172e4428a8e3e22af377ef9db : SUCCESS in 7m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/16d5ec8163034be2af89ead04fb1d474 : SUCCESS in 4m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6f28842024f44babf48a31085754839 : SUCCESS in 34m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/876a90f70be340adb2153a23506f9a85 : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c18f450efd144dd83811f5b8a5bf7e0 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d328fbc785f440f49265b0e9dd31f8e8 : SUCCESS in 20m 42s","accounts_in_message":[],"_revision_number":5},{"id":"1a3f7e657c4bfc4464a746fad98a9c18c0dfadb5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-02-10 14:09:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d56d1b062c780310b106a55e2230682fd1b38e6c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-06 11:14:34.000000000","message":"Abandoned\n\nProposing new patches shortly.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8b28da83b486553f1eb77b7a5f619f2030670775","revisions":{"05a8fda5f6836b911d7b3cbc8e7837e7ccea5e4f":{"kind":"REWORK","_number":1,"created":"2025-09-11 10:39:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/53/960453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/960453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/960453/1"}}},"commit":{"parents":[{"commit":"1f86f662d6da5959064da2dea052c0136bf25e6f","subject":"Fix account server functional tests if run partially","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f86f662d6da5959064da2dea052c0136bf25e6f"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-11 10:36:40.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-11 10:36:40.000000000","tz":120},"subject":"WIP: Remove eventlet from swift.common.http_protocol","message":"WIP: Remove eventlet from swift.common.http_protocol\n\nThis commit removes eventlet from swift.common.http_protocol by using\nhttp.server.BaseHTTPRequestHandler as HttpProtocol (which is the same\nbase class as used by eventlet itself).\n\nThis is a WIP; most unit tests rely on eventlet.wsgi and this needs to\nbe implemented without using eventlet to eventually run unit tests\nsuccessfully. A subset of unittest can be run using the following\ncommand:\n\nDISABLE_EVENTLET\u003dtrue pytest \\\ntest/unit/common/test_http_protocol.py::TestSwiftHttpProtocol\n\nChange-Id: I66127aff3cdd20b6fa2dd756c4b75f3ef0a15288\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05a8fda5f6836b911d7b3cbc8e7837e7ccea5e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05a8fda5f6836b911d7b3cbc8e7837e7ccea5e4f"}]},"branch":"refs/heads/feature/threaded"},"5608b2ecea005b510934b897783a5e14320da07f":{"kind":"REWORK","_number":2,"created":"2025-09-11 11:34:56.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/53/960453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/960453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/960453/2"}}},"commit":{"parents":[{"commit":"1f86f662d6da5959064da2dea052c0136bf25e6f","subject":"Fix account server functional tests if run partially","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f86f662d6da5959064da2dea052c0136bf25e6f"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-11 10:36:40.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-11 11:34:29.000000000","tz":120},"subject":"WIP: Remove eventlet from swift.common.http_protocol","message":"WIP: Remove eventlet from swift.common.http_protocol\n\nThis commit removes eventlet from swift.common.http_protocol by using\nhttp.server.BaseHTTPRequestHandler as HttpProtocol (which is the same\nbase class as used by eventlet itself).\n\nThis is a WIP; most unit tests rely on eventlet.wsgi and this needs to\nbe implemented without using eventlet to eventually run unit tests\nsuccessfully. A subset of unittest can be run using the following\ncommand:\n\nDISABLE_EVENTLET\u003dtrue pytest \\\ntest/unit/common/test_http_protocol.py::TestSwiftHttpProtocol\n\nChange-Id: I66127aff3cdd20b6fa2dd756c4b75f3ef0a15288\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5608b2ecea005b510934b897783a5e14320da07f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5608b2ecea005b510934b897783a5e14320da07f"}]},"branch":"refs/heads/feature/threaded"},"61c98f9ba23bf66ec8e595d56a35f3cfe2057560":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-11 11:59:16.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/53/960453/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/960453/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/960453/3"}}},"commit":{"parents":[{"commit":"8798d04299cff228b8a801097aa4ad960c499b46","subject":"Fix account server functional tests if run partially","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8798d04299cff228b8a801097aa4ad960c499b46"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-11 10:36:40.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-11 11:52:26.000000000","tz":120},"subject":"WIP: Remove eventlet from swift.common.http_protocol","message":"WIP: Remove eventlet from swift.common.http_protocol\n\nThis commit removes eventlet from swift.common.http_protocol by using\nhttp.server.BaseHTTPRequestHandler as HttpProtocol (which is the same\nbase class as used by eventlet itself).\n\nThis is a WIP; most unit tests rely on eventlet.wsgi and this needs to\nbe implemented without using eventlet to eventually run unit tests\nsuccessfully. A subset of unittest can be run using the following\ncommand:\n\nDISABLE_EVENTLET\u003dtrue pytest \\\ntest/unit/common/test_http_protocol.py::TestSwiftHttpProtocol\n\nChange-Id: I66127aff3cdd20b6fa2dd756c4b75f3ef0a15288\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61c98f9ba23bf66ec8e595d56a35f3cfe2057560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61c98f9ba23bf66ec8e595d56a35f3cfe2057560"}]},"branch":"refs/heads/feature/threaded"},"9eb5d53accb7f3dbede2551e0db7f1c51402cafc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-09-11 17:01:12.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/53/960453/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/960453/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/960453/4"}}},"commit":{"parents":[{"commit":"082a77c73948d8668e150a9a6dc0f3f04a78f26e","subject":"Fix account server functional tests if run partially","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/082a77c73948d8668e150a9a6dc0f3f04a78f26e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-11 10:36:40.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-11 15:47:28.000000000","tz":120},"subject":"WIP: Remove eventlet from swift.common.http_protocol","message":"WIP: Remove eventlet from swift.common.http_protocol\n\nThis commit removes eventlet from swift.common.http_protocol by using\nhttp.server.BaseHTTPRequestHandler as HttpProtocol (which is the same\nbase class as used by eventlet itself).\n\nThis is a WIP; most unit tests rely on eventlet.wsgi and this needs to\nbe implemented without using eventlet to eventually run unit tests\nsuccessfully. A subset of unittest can be run using the following\ncommand:\n\nDISABLE_EVENTLET\u003dtrue pytest \\\ntest/unit/common/test_http_protocol.py::TestSwiftHttpProtocol\n\nChange-Id: I66127aff3cdd20b6fa2dd756c4b75f3ef0a15288\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eb5d53accb7f3dbede2551e0db7f1c51402cafc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eb5d53accb7f3dbede2551e0db7f1c51402cafc"}]},"branch":"refs/heads/feature/threaded"},"8b28da83b486553f1eb77b7a5f619f2030670775":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-10-27 11:17:19.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/53/960453/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/960453/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/960453/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/960453/5"}}},"commit":{"parents":[{"commit":"e507a61f49da85280e2cb360ca3f123c20020b52","subject":"Fix account server functional tests if run partially","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e507a61f49da85280e2cb360ca3f123c20020b52"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-11 10:36:40.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-27 11:15:58.000000000","tz":60},"subject":"Remove eventlet from swift.common.http_protocol","message":"Remove eventlet from swift.common.http_protocol\n\nThis commit removes eventlet from swift.common.http_protocol by using\nhttp.server.BaseHTTPRequestHandler as HttpProtocol (which is the same\nbase class as used by eventlet itself).\n\nThis is a WIP; most unit tests rely on eventlet.wsgi and this needs to\nbe implemented without using eventlet to eventually run unit tests\nsuccessfully. A subset of unittest can be run using the following\ncommand:\n\nDISABLE_EVENTLET\u003dtrue pytest \\\ntest/unit/common/test_http_protocol.py::TestSwiftHttpProtocol\n\nChange-Id: I66127aff3cdd20b6fa2dd756c4b75f3ef0a15288\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b28da83b486553f1eb77b7a5f619f2030670775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b28da83b486553f1eb77b7a5f619f2030670775"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
