)]}'
{"id":"openstack%2Fswift~961273","triplet_id":"openstack%2Fswift~master~I4bb6cc238d4e567e3edc6c15a58d4a5f9a41e273","project":"openstack/swift","branch":"master","topic":"doc-link-tweaks","attention_set":{},"removed_from_attention_set":{"38368":{"account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"last_update":"2025-09-16 15:09:01.000000000","reason":"Change was submitted"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-09-16 15:09:01.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-09-16 14:27:20.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-09-16 04:39:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I4bb6cc238d4e567e3edc6c15a58d4a5f9a41e273","subject":"cleaning up and fixing some links","status":"MERGED","created":"2025-09-15 17:51:04.000000000","updated":"2025-09-16 15:10:35.000000000","submitted":"2025-09-16 15:09:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"961273-doc-link-tweaks","meta_rev_id":"8b97cf21636246bbf5c74d5d521701a3bc2da891","_number":961273,"virtual_id_number":961273,"owner":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-16 15:09:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"recommended":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"all":[{"value":1,"date":"2025-09-15 23:26:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":2,"date":"2025-09-16 14:27:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-16 04:39:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":1,"date":"2025-09-16 14:27:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-15 17:59:26.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2025-09-15 17:59:26.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-09-15 17:59:26.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-09-15 18:31:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3fd231bf93ffbbaae72bc856cde68dceb8300183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-09-15 17:51:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d313e020551031df771d511c070c49ffacdbcd8c","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-09-15 17:59:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"df23abd3652d6fd914725af64f9a9f9a383511bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-09-15 17:59:33.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"be9241d1e385486ee55cd37748b96c8e25449d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 18:31:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a25eb688fc2475bb8296559493573e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5a54ee9aef4af9a6be9105784095ba : SUCCESS in 11m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c3fae4b1f9a4bd88d20b9969c41fc5e : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ac4cfb8ca5d94396847bb13c249cf8c8 : SUCCESS in 25m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/af2c71dd00ff46c89193a2fabb2359a7 : SUCCESS in 17m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fb168bd3f5b745b58a962f1cb9d137bc : SUCCESS in 27m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/69772e261ce44227b51caa5f5a26a798 : SUCCESS in 26m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1268c2974f714f5ea66a0cb69561f2ab : SUCCESS in 8m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/027ac1ddb2ad4872b1ddcd744984dde0 : SUCCESS in 24m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/735a5df168e54523955d5eea5cbb78f0 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"aefdb782cbb93550732f1606eaca976ef125c022","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-15 19:47:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"95fdfdbabefb6514d4665406ea9137feda8f4ad5","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-15 20:00:07.000000000","message":"Patch Set 2: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cf7eaa1dc76bb2629bc94f7080f1b6bbbe8511fd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-15 21:19:11.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"912be7723000bd7ee9919c7dee355c622b622c51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-09-15 23:18:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"828f00d3eeffea657d12d56185c429e3f43c91fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-09-15 23:22:37.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"e47ea21e46f1f3c81266497e53bb70bcd722179a","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-09-15 23:26:24.000000000","message":"Patch Set 4: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7f91bf6e947372b5fd15a0127ec27d44c992c496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 23:55:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41bb4de6f1e3485aa1fb80da226c0fa9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d66aacff6747d8a7d16b58391b43b2 : SUCCESS in 9m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bbc0dc78f4d743728625bc0881400519 : SUCCESS in 6m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7c0d6d9a52ad4cc98e38fe7d1a19dd6c : SUCCESS in 23m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/46973f69cfac4cb59bba99ce2ac3b947 : SUCCESS in 20m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b059e1ecbb1b4a4691168474c0a862cf : SUCCESS in 28m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fd394ab5e96947ab9a39c3f4efa966da : SUCCESS in 19m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9932a093dda949dfba70fb69d0b3f5db : SUCCESS in 6m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0991cff5b874413dbff09eb56095f5d0 : SUCCESS in 27m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f430191eef914bd6a0d2b067fdde2e17 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":4},{"id":"380a49e8f91c18b8b73557e915d63d287d5cfd62","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-16 04:39:54.000000000","message":"Patch Set 4: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6c700e94b4258d0bfd878ea1aa311fc223552a6e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-16 14:27:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0d880f5e4182032d728664da6efca109352c1bef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 14:27:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"af67bf6c66fce8956a0ca5054f5708a94bf85e67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:09:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1695fd831b8946e5a47451cd5c50e208\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/695255094b9941dc891a7837f0df7dfd : SUCCESS in 10m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/da5f4b2d39614e0e914b414fc368ca18 : SUCCESS in 6m 45s (non-voting)\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca1943c606764d1e8a17baf0c1455b10 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e20c3294ea044fca86c0687c3a0dbe4a : SUCCESS in 7m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7ec6fa1322f46a1af6e62e64c919a18 : SUCCESS in 20m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d95c613922140d99617f865693bef1a : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":4},{"id":"dc25614059b7ef2a33d9b60ba5131dd92a8d905e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:09:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8b97cf21636246bbf5c74d5d521701a3bc2da891","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:10:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b154ca9646b4543ae6288db53db3d03\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f976753336741a29d40a40b766e736b : SUCCESS in 1m 02s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3b3c3c3ae80a43c5b28aebe613ce5d23 : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bd27fc6baf9bf981e7c1acfd05a72589c35ba875","revisions":{"11e87b991ea4e7d7572c0ddabea82a96990dd501":{"kind":"REWORK","_number":1,"created":"2025-09-15 17:51:04.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/73/961273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/961273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/961273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/961273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/961273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/961273/1"}}},"commit":{"parents":[{"commit":"82cb5a5d78b91c6af258cd8f06d30f69be90fa18","subject":"AUTHORS/CHANGELOG for 2.36.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82cb5a5d78b91c6af258cd8f06d30f69be90fa18"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-15 16:48:38.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-15 17:50:49.000000000","tz":-420},"subject":"cleaning up and fixing some links","message":"cleaning up and fixing some links\n\nChange-Id: I4bb6cc238d4e567e3edc6c15a58d4a5f9a41e273\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11e87b991ea4e7d7572c0ddabea82a96990dd501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11e87b991ea4e7d7572c0ddabea82a96990dd501"}]},"branch":"refs/heads/master"},"a9afabec303b2823cd086a0acad4ddc57e1063f1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-09-15 17:59:33.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/73/961273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/961273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/961273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/961273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/961273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/961273/2"}}},"commit":{"parents":[{"commit":"82cb5a5d78b91c6af258cd8f06d30f69be90fa18","subject":"AUTHORS/CHANGELOG for 2.36.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82cb5a5d78b91c6af258cd8f06d30f69be90fa18"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-15 16:48:38.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-15 17:59:33.000000000","tz":0},"subject":"cleaning up and fixing some links","message":"cleaning up and fixing some links\n\nVerified the links work properly by building the docs locally (sphinx-build -W -b html doc/source doc/build/html) and opening the links on the modified pages.\n\nChange-Id: I4bb6cc238d4e567e3edc6c15a58d4a5f9a41e273\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9afabec303b2823cd086a0acad4ddc57e1063f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9afabec303b2823cd086a0acad4ddc57e1063f1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9f0c2ef839e5334d98d933ee6ddfa024d870a39b":{"kind":"REWORK","_number":3,"created":"2025-09-15 23:18:01.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/73/961273/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/961273/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/961273/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/961273/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/961273/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/961273/3"}}},"commit":{"parents":[{"commit":"82cb5a5d78b91c6af258cd8f06d30f69be90fa18","subject":"AUTHORS/CHANGELOG for 2.36.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82cb5a5d78b91c6af258cd8f06d30f69be90fa18"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-15 16:48:38.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-15 23:13:42.000000000","tz":-420},"subject":"cleaning up and fixing some links","message":"cleaning up and fixing some links\n\nChange-Id: I4bb6cc238d4e567e3edc6c15a58d4a5f9a41e273\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f0c2ef839e5334d98d933ee6ddfa024d870a39b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f0c2ef839e5334d98d933ee6ddfa024d870a39b"}]},"branch":"refs/heads/master"},"bd27fc6baf9bf981e7c1acfd05a72589c35ba875":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-09-15 23:22:37.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/73/961273/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/961273/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/961273/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/961273/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/961273/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/961273/4"}}},"commit":{"parents":[{"commit":"82cb5a5d78b91c6af258cd8f06d30f69be90fa18","subject":"AUTHORS/CHANGELOG for 2.36.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82cb5a5d78b91c6af258cd8f06d30f69be90fa18"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-15 16:48:38.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-15 23:22:37.000000000","tz":0},"subject":"cleaning up and fixing some links","message":"cleaning up and fixing some links\n\nVerified every changed link works by\nbuilding and testing manually\n\nChange-Id: I4bb6cc238d4e567e3edc6c15a58d4a5f9a41e273\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd27fc6baf9bf981e7c1acfd05a72589c35ba875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd27fc6baf9bf981e7c1acfd05a72589c35ba875"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
