)]}'
{"id":"openstack%2Fswift~961323","triplet_id":"openstack%2Fswift~master~If9ad28cf84dd16b10e1682d2b8d18d5f4839f681","project":"openstack/swift","branch":"master","topic":"err-in-link-at","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-09-19 23:11:37.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-09-19 22:23:04.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"If9ad28cf84dd16b10e1682d2b8d18d5f4839f681","subject":"raise 500 on ec timestamp collision","status":"ABANDONED","created":"2025-09-15 23:00:55.000000000","updated":"2025-09-19 23:11:37.000000000","total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"2192c848c9e4849e7851b4c2893b8e5630c284bd","_number":961323,"virtual_id_number":961323,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-16 16:14:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2025-09-19 17:18:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":-1,"date":"2025-09-19 22:23:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":-1,"date":"2025-09-19 22:23:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-15 23:59:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-16 00:31:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-19 17:18:40.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"98cca7867c811a675f5557c2b5aa3aa0ea3a3c53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-15 23:00:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ffbf0c7377bed7e8042cb8d22b5965e05b1a701","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 23:59:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30b4b1d5f5544a26b5ef102e3f8f8b28\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/03d30ac8bb084cf882dd7d1a33b2bb7a : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":1},{"id":"76236297f8e677a7eb3ee48a2928c1249f1a4ac1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 00:31:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5827370948bf4cada5961ac210f33194\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6cc1bb19914073aaad920a6e617a5f : SUCCESS in 7m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/4bdc3c5e3bdf429cb3e890f35d8b9844 : SUCCESS in 41m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8b49ce2a5a347c69ace0bf6bf226a28 : SUCCESS in 1h 04m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8c57ab971fe249748dde2d795272ae21 : SUCCESS in 49m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/005480e26d744091bf6e10518722d06f : SUCCESS in 1h 04m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3d4bb6c4c064c65b4bcf6037b30acd3 : SUCCESS in 14m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df1d52a16f1a43efbb49916ab73b9029 : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed0805330a984009ba5610014e18cf96 : SUCCESS in 17m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ff93f0b00154ebcae30aca8ea0505e7 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/432452c8ed8541af80b0fd38d1613147 : SUCCESS in 23m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/26ae3b6b0c804f41adadd329ce91f029 : SUCCESS in 22m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cbf87c5224944cbfbad2e9bab05d1481 : SUCCESS in 29m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/431c5718438345a18735fd9364ab60f4 : SUCCESS in 30m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b8bff1dfa3b14c0a97f3c744cda23a51 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e3f23b6436a54f95b49bea97b395950a : SUCCESS in 14m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/529d70517d4d4dd79ebf32e11bf514f9 : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2e7c215b94a444ae821b1d44df96ad02 : SUCCESS in 6m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff5d870669124766b1ede1fb1a44d46f : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5d64e764018843f09a1abac71b5caad9 : SUCCESS in 11m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/81279c21bfc1451199b3e7b8675f895a : SUCCESS in 58m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a358fa7856d8415697bea8c2f83f525e : SUCCESS in 33m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0f46ba3e31e436c9fbfe85a5f24179c : SUCCESS in 57m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12d31be8599e4596960858ec02254edb : SUCCESS in 27m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2284d8249f348efba4fd63ebefbb3d3 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bbffe89a9d854396b2f2dc481fd3fc23 : SUCCESS in 15m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a3dffe41b4640abafdb0c882382e800 : SUCCESS in 56m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d04b86e05ef4fb38b9e6b4121a1a494 : SUCCESS in 52m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4756d690cb431f9e23e3d14066e8bb77c4191eb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-16 14:57:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"51a3bbfdd8d5c722b7e41f833fa813aed8e91781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 16:14:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc6e9fdff66247ddb24a19757a336377\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62bbf42fceac48bd951e17f038e08e72 : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/ee4862068be0494290b792d0bed85c0c : SUCCESS in 1h 11m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/70c06a501cd3450d858fd7b94ceefb0f : SUCCESS in 1h 03m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f55a49a941447e69da0f5583918a005 : SUCCESS in 51m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64463679b6c04233b6eb6fdb95accd39 : SUCCESS in 1h 00m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28a41e4515874a35943f697b19b91e2c : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9c016ed15654ba7826466270ce0375a : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f0f7e73852b46bab67929fe0d5c4cf3 : SUCCESS in 18m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2c68c8ee03a47188d5b49d6a548f683 : SUCCESS in 10m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/85ffdabfa6e343f1955bfd0b97a038bb : SUCCESS in 25m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b4be5f5faa1a414bb769b7ec376fc2e8 : SUCCESS in 17m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/37adb85dd0754082a341179dea49c1d2 : SUCCESS in 30m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9373b0f2e0474ecab3ba92b0e1dd092c : SUCCESS in 24m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5a9d36b1b019422aae2171bf8bde0e7d : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a704e7eace8742aea41d61c8863918af : SUCCESS in 16m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2edd06a2070c44baa4ea35aa5b8ab488 : SUCCESS in 15m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bd20f4320a8348fbaa12d5f9e07eee15 : SUCCESS in 9m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c11edcca9ba942188f51284bf715f816 : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cbdfeb4c04e645a38eaed4de2b41d4a1 : SUCCESS in 11m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ba7439144e1423db1b97e4368ec59fe : SUCCESS in 1h 02m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/723a0ef7e0dc41ca96d9c38747c6ce1f : SUCCESS in 52m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/44b12d3e4327478390e8786cb08a1443 : SUCCESS in 50m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7eab7592ac0a4587b78bddc0dc196b66 : SUCCESS in 34m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4b393b0d3744b10bc0e4b96591604af : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/511f9b12cfb247f8ad848501dbde3a88 : SUCCESS in 16m 33s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f37d09fe8664426dab71d8b4d12f27fe : SUCCESS in 37m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f12634659514429ab7537b53923b0448 : SUCCESS in 50m 47s","accounts_in_message":[],"_revision_number":2},{"id":"8412b79e03d13ff7d8094ae02fe43a709cd59b07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 16:43:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7830245c81524f448450fa896978878b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a729a3408cdc44149a041944b80e99c6 : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":2},{"id":"862081812a675b5293aa89bd1093289f4d62a7ac","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-19 17:18:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a96ec59e76f209bd9b8f0a0fed8096ac0ba8573e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-19 22:23:04.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2192c848c9e4849e7851b4c2893b8e5630c284bd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-19 23:11:37.000000000","message":"Abandoned\n\nsquashed 927327: object-server: raise 500 on ec timestamp collision | https://review.opendev.org/c/openstack/swift/+/927327","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef79b71af3da708757d905b79040184d77b2950b","revisions":{"04d46de2fcde39cb6505998e832d245d95e63695":{"kind":"REWORK","_number":1,"created":"2025-09-15 23:00:55.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/23/961323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/961323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/961323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/961323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/961323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/961323/1"}}},"commit":{"parents":[{"commit":"d06e6454beefa8389802d2211d1474bfc4aa1044","subject":"Test: timestamp collisons when uploading same EC object.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d06e6454beefa8389802d2211d1474bfc4aa1044"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-15 22:50:04.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-15 23:00:08.000000000","tz":-300},"subject":"raise 500 on ec timestamp collision","message":"raise 500 on ec timestamp collision\n\nBasically if trying to create a new non-durable EC object at the exact\nsame timestamp while another thread is trying to commit: just blow up\nthe second writer so it\u0027s never even non-durable and let the first one\ntry to finish.  If neither gets durable quorum they both return errors\nfrom the proxy.\n\nTODO: The object-server traceback is not very subtle.\n\nChange-Id: If9ad28cf84dd16b10e1682d2b8d18d5f4839f681\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04d46de2fcde39cb6505998e832d245d95e63695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04d46de2fcde39cb6505998e832d245d95e63695"}]},"branch":"refs/heads/master"},"ef79b71af3da708757d905b79040184d77b2950b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-16 14:57:06.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/23/961323/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/961323/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/961323/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/961323/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/961323/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/961323/2"}}},"commit":{"parents":[{"commit":"e5559f7ed00e330cee6aea02d4e1fbc8c33ba98c","subject":"Test: timestamp collisons when uploading same EC object.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5559f7ed00e330cee6aea02d4e1fbc8c33ba98c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-15 22:50:04.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-16 14:52:18.000000000","tz":-300},"subject":"raise 500 on ec timestamp collision","message":"raise 500 on ec timestamp collision\n\nBasically if trying to create a new non-durable EC object at the exact\nsame timestamp while another thread is trying to commit: just blow up\nthe second writer so it\u0027s never even non-durable and let the first one\ntry to finish.  If neither gets durable quorum they both return errors\nfrom the proxy.\n\nTODO: The object-server traceback is not very subtle.\n\nChange-Id: If9ad28cf84dd16b10e1682d2b8d18d5f4839f681\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef79b71af3da708757d905b79040184d77b2950b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef79b71af3da708757d905b79040184d77b2950b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
