)]}'
{"id":"openstack%2Fswift~961370","triplet_id":"openstack%2Fswift~master~I0168ab113fdda60ed858ed0928356699399d4044","project":"openstack/swift","branch":"master","topic":"p-test-reconstructor-simplify","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-09-16 15:49:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0168ab113fdda60ed858ed0928356699399d4044","subject":"tests: simplify TestGlobalSetupObjectReconstructor setUp","status":"MERGED","created":"2025-09-16 13:43:08.000000000","updated":"2025-09-16 15:50:49.000000000","submitted":"2025-09-16 15:49:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961370-p-test-reconstructor-simplify","meta_rev_id":"ec0d2bf2586e0cb3c2eddf9db0cd5f4ca2c9262a","_number":961370,"virtual_id_number":961370,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-16 15:49:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2025-09-16 14:29:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2025-09-16 14:52:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-16 14:29:46.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-09-16 14:40:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"119ad27a3e9d7142772e9b1e5a943ba8fa1fe425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-16 13:43:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cc95131a12a5d674a22b5569f45ba4cff146073","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-16 13:57:47.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"57a6d5e35abe749ba2b9afb786415cd418c5fe45","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-16 14:29:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7dc8a8bb30bee0a819ce82f04fbf3dc69fe520b7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-16 14:32:39.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f546c5081d672a43597c9207137f98b6ba8359b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 14:40:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3b20baf31fe4044b083bb1eb739d560\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce64bd39d014b3a82c9c4d3f4673356 : SUCCESS in 11m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/870f6950d09747d3a039c5f96bfe6b55 : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bea403489a3d4e03893bdcf2a3418444 : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db4772d957b246a09be0757970430a29 : SUCCESS in 21m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/650ee1120320436396fe3d0e61f38c0e : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b736bb3e3ad6400cbff7c4eb6469e2d7 : SUCCESS in 32m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/078254f5ec0948fe99e3aa94612fb865 : SUCCESS in 20m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/518397d0318648b89cf2f659de9a6d37 : SUCCESS in 26m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1737a461db4a4119bfd01234f2ca02eb : SUCCESS in 31m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5e955e56fc3f46179d77f786ca112011 : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/18a75755b38e4282a9b7c96abe5eef58 : SUCCESS in 15m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b8b20216b39f4827ac2ee0e858f00be6 : SUCCESS in 15m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8bf0004cb05e4be09ba64132925afd9e : SUCCESS in 39m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b66a0d6c6fa4571bae4865213ada9cc : SUCCESS in 38m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d857356d3904bbc99f06f05f06b48cd : SUCCESS in 18m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0b1e07e6f548e9be6f7b76eb814ce8 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f7c49c3427c441d19d121086133dfa17 : SUCCESS in 18m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a97809092ccec8662a8928e478105c729f5bd533","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-16 14:52:09.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2708234d56b5cd49dca294f0558ecfe542c3b8c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 14:52:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e41485cf8ad325eb10bd051fcf8971948a11a441","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:16:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/047fb8c89b1b430289416fb3a3b7ef67\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d990e5094300487493ba41d173f5a622 : SUCCESS in 1h 15m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7e81d4580219af127c2779ddebbd7e227c5d5bea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:49:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97b5dbad5def4d70b49748ec6544f8f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e278a2035f945a1868de4fcf259ebcf : SUCCESS in 8m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1269fabd81b54d4898321b16cf0ba40b : SUCCESS in 9m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3259b6d75cad43ff91f498493d3f86da : SUCCESS in 18m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5ad9035bf9e34f5f8a545b4a49757a73 : SUCCESS in 21m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c887742244794480af0f0795bfa7a155 : SUCCESS in 23m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bcd693695f2c47d9be2d004e2f8a705f : SUCCESS in 32m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d63a495e6653413aac4fe4303a15fabc : SUCCESS in 11m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2d7d3851b7994795a713088c007ba364 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e49e0496ab33458b9b485ed283f97dd0 : SUCCESS in 14m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e4003b7efea04ff5929684ad4d8cf4a0 : SUCCESS in 7m 18s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f395aaf332cf45679749db99f5247db9 : SUCCESS in 8m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/66e2cee8c96443028b2059f50c29991d : SUCCESS in 46m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f412d1b6c1b402cbc0e13e86f7bfd52 : SUCCESS in 39m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c62e7311492b4184b14f7ddf710c975e : SUCCESS in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b424c9228cf4f9bbe475bede7325226 : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d0a83378f2114d34b2c7917250ff274b : SUCCESS in 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"0f0d52a867929ab1d446a8dfad7013981f49d547","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:49:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ec0d2bf2586e0cb3c2eddf9db0cd5f4ca2c9262a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:50:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c514cbfa3e28477f8183088012f9f914\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af117d24d05d4526aa6215b6232cac61 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f79e90104c42416192a180cff73b60a5 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c26c7b8edd464d1fcf1d219f7c3fb040914d9da3","revisions":{"c26c7b8edd464d1fcf1d219f7c3fb040914d9da3":{"kind":"REWORK","_number":1,"created":"2025-09-16 13:43:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/70/961370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/961370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/961370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/961370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/961370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/961370/1"}}},"commit":{"parents":[{"commit":"63eeb005bd4fa9e2696ddf8e1fc84936b86afed4","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63eeb005bd4fa9e2696ddf8e1fc84936b86afed4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-09-16 13:40:49.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-09-16 13:42:56.000000000","tz":60},"subject":"tests: simplify TestGlobalSetupObjectReconstructor setUp","message":"tests: simplify TestGlobalSetupObjectReconstructor setUp\n\nChange-Id: I0168ab113fdda60ed858ed0928356699399d4044\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c26c7b8edd464d1fcf1d219f7c3fb040914d9da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c26c7b8edd464d1fcf1d219f7c3fb040914d9da3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
