)]}'
{"id":"openstack%2Fswift~961761","triplet_id":"openstack%2Fswift~master~Iaae0d9fb7a1949d1aad9aa77b0daeb249fb471b5","project":"openstack/swift","branch":"master","topic":"stuck-relinker","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-10-13 11:08:54.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-10-13 13:40:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaae0d9fb7a1949d1aad9aa77b0daeb249fb471b5","subject":"relinker: allow clobber-hardlink-collision","status":"MERGED","created":"2025-09-18 23:43:14.000000000","updated":"2025-10-13 13:41:30.000000000","submitted":"2025-10-13 13:40:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":36,"unresolved_comment_count":16,"has_review_started":true,"submission_id":"961761-stuck-relinker","meta_rev_id":"6dd5b4f2c7ff433323f24f5468f614e3d02b5f34","_number":961761,"virtual_id_number":961761,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-13 13:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-10-13 11:08:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-10-13 11:08:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-19 00:37:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-19 00:48:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-10 12:22:57.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"321b082ae3616924ba17f4aafda1832597141b30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-18 23:43:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5d13b23e24bfc0b55593d1fbf562c1060eba54e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 00:37:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ee5fbdf477742b5bb6cea2c445d59ee\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e8fe7db12b404e13babf29a433a355cd : SUCCESS in 51m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4afe21af3aefd1c4422cbda53316f4a8d6a533bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 00:48:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b688b093bb984b8cb3b0c5e00b4d1980\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6391ab11023b482a990e8c803ae2ea8e : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/16ffcf8bdac441e0ad4564a36c555ba9 : SUCCESS in 44m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1abdbd78a71842e9b1970fcb1c40a45a : SUCCESS in 48m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3a4ad0c011bd4ef0ba296b4a6155ef67 : SUCCESS in 57m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/365bb7e753f1422191125d770e30c9af : SUCCESS in 49m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65fcd308a00840449914d1d03085b841 : SUCCESS in 16m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6621cc9beeb4acabf2498ac069aa324 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cb2ded3c37743ec9c5ea631294731c0 : SUCCESS in 19m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c49463858c6e47f7a7967e2aa708c022 : SUCCESS in 8m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c09e04d1ebcf4bfa91b8af085e4136e9 : SUCCESS in 23m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1fc791a8e5e948a39e40fc493454dc97 : SUCCESS in 20m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1c14763713e3459c98535129bb0ab022 : SUCCESS in 24m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b9563d87d44b43d38bdc1ab298e7bd70 : SUCCESS in 25m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/365ffece44c341b98c08f74f64a41b84 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/212ba087e5144314873f0be8aaffc3ff : SUCCESS in 15m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d6b58d2b28b84511b30fabfc74f77dce : SUCCESS in 16m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c708bf238ade4c2f90a6a931f983d589 : SUCCESS in 9m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/583e7aa082004797be2230096eca8777 : SUCCESS in 10m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/153b58013a5941b5a9a8602d8bde6f55 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e7315eb092b4be5adfad8310b738ae0 : SUCCESS in 42m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25554467db8648b5a0f88803a7916082 : SUCCESS in 55m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0b86e7b239943edb0a364da70c6ab3e : SUCCESS in 30m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37d7da467d154b3e86adb4f3fdca7b42 : SUCCESS in 26m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8888cecd1bc04b28bc8ba39ba544ceb2 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8bf2b088a7c04f1b847f7c09703d8b6a : SUCCESS in 17m 03s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8b616057630434ea35d5ceeb5cfd458 : SUCCESS in 48m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/019df62d5a1f43c0a3dd5b2443f63694 : SUCCESS in 53m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5b3fcdffe8ed2e718f100264f7a77ebed2d26961","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-19 20:52:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d7acfdf04ab6fb6c638b69898a6a021fd9cfa6e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 21:46:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/803574103d1a4605b2b14a2d5033bc97\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dac10a7d5f434d74a4afc847dfd71b19 : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c170aeb4c196f8fa68abf7000dbda4b7369d74ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 22:02:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b351ae40c38e4708b5cbc98f4d61e121\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0558334406684d1b9ad09531cef211bf : SUCCESS in 11m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/f960cfdb0d93497cae50d61a8ea38e37 : SUCCESS in 41m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/02c4714d8d644910814ee1bb955e1fba : SUCCESS in 46m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f45a5a903b6450c8c53db1b95270c1b : SUCCESS in 48m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63710f4d82054066ae46bf2212b2f072 : SUCCESS in 1h 05m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df1c79dc655947fab273f554c8e6d94d : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9bb05cff182f4520a6e1d3cf67048ff6 : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7de14d5aecf64e55998672df6ba10572 : SUCCESS in 19m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c1c639f3c5148b8ac6c31488780f47c : SUCCESS in 9m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7e7768583d3940a89af21685c375a570 : SUCCESS in 23m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1cf58413dc114cc7ac0abaa816f1de43 : SUCCESS in 19m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/90b5a6ad676b4180808dfdeedbf09300 : SUCCESS in 29m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/25bd21823d2d4c9c84d009d81c732d17 : SUCCESS in 23m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b1d556a664674d0fbdb222013e093f01 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a649337b32874b40b80fbfa57c4e303d : SUCCESS in 15m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/34aa4e9c4b9d459e8e65451a387b943d : SUCCESS in 15m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bc6e078f57ae45718504005be1cd1519 : SUCCESS in 6m 33s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/433e67b6cabd4795838677d0036a4be2 : SUCCESS in 14m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b1e17940c3de4f81bd506047283009b0 : SUCCESS in 8m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/26d079a9a98c48bcafc86d8c1d5e3a8d : SUCCESS in 58m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a6b27593c9e45a7833f4efe0f7f618b : SUCCESS in 48m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0c863993699449b292fac16eb089c19a : SUCCESS in 56m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/094471362d2f4264a5b4982b173fe7f3 : SUCCESS in 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66fa122b068443eda84594de076737fd : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2e1b89436534d4bb0e8fcfe68db9028 : SUCCESS in 18m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a3d423770264feaa1b13739acaefc3f : SUCCESS in 34m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/990e5d1b45124f92a930751651081ad7 : SUCCESS in 45m 28s","accounts_in_message":[],"_revision_number":2},{"id":"f2923e8218a03bb85d6880206264d92da1b70bed","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-10 12:22:57.000000000","message":"Patch Set 2: Code-Review+1\n\n(24 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9ec72607795c900327b9256e1e0986046518e6ef","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-10-10 15:32:14.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1890f3523985e8e606780af089c40737f78c0834","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-13 11:03:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2cf0875b3ed362b23320cd08f970cbcc7d705da0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-13 11:08:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f198e43d80b598a4d5402f7534dc9654ddce7fe3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 12:12:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fcc21abce404908bd7d6dfbc5299ff3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/481a07b8d1c6401fa503ca389628adec : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":3},{"id":"c0526437b0b9038eb6d7c71345e7f2f386376e73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 12:26:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f575c046f04b4f978a1bac42c17d7a88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b49a377876c74c2d8abe1e1ba517f9f5 : SUCCESS in 11m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/10535e19b932418587c9981eb181dd99 : SUCCESS in 1h 01m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/678f386354c04c59949026b1a5912b78 : SUCCESS in 58m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/94ae162828174f56a35f111657ecf629 : SUCCESS in 40m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5776ddfe98714eecbf8dd38542e00b8d : SUCCESS in 1h 04m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b11c82218fe34148b720e6244bfb9e87 : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7446d3b4a8a45e9823954380ed30001 : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b7e745ce7924e589397fa6da224e01c : SUCCESS in 10m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1d9b6ec79409462b8261195b6a0306e9 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cd1f94d7b99041649569a7a88a1418b3 : SUCCESS in 18m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/276f65f7f59141f4a5d5c72ea3c77757 : SUCCESS in 9m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7ffdcf1ab31d402eac14eaa941be7b5b : SUCCESS in 29m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8f11a78d1fc7406e9cf554b7377e4c99 : SUCCESS in 20m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fa741c3c926b4512bac0ef64a650c105 : SUCCESS in 13m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/db288994c40144f28a182dedcc259dc1 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dc9b230978ae4c348547fe3a4689a1ba : SUCCESS in 15m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/367259f3368a4243a9bc17c5638d8fe6 : SUCCESS in 5m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d71ce8ec9e2a4433ab29e0bc7100e4b1 : SUCCESS in 8m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e116da17bd149e8b68927c4b593557d : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/01aa5526a8914b029b05bf64a24f8857 : SUCCESS in 56m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68a14c63bdb84e20a9ac49b5dd2f09ab : SUCCESS in 56m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4246aa757bd94a79b6462cafd3d953e2 : SUCCESS in 48m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4a43bb61df541cc83414a2bd50c5003 : SUCCESS in 25m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d22a855c60f84da5b489258821afdb26 : SUCCESS in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c75ef1a500394353894fc7a5e6e33714 : SUCCESS in 25m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a05d540819d4418a237c760b361da11 : SUCCESS in 39m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/250c0de7b8464f5eb8ad37b2216b6136 : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":3},{"id":"b61fad8bd4f2d925793f47e5e5b8372f80bb18e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 12:27:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"596d864746b482dbdb57ad43eff7550e6b877161","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 13:40:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebe8355b91b94c36b5ddcf0fa6420d15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5fa875141ab4ddc824570d57763af98 : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/f702630d2d8f4c91b5cfe301549aca7b : SUCCESS in 50m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/496813e332b24a97b5f6bc53c767f972 : SUCCESS in 31m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8a968bc474e346929b8e407973f97adc : SUCCESS in 1h 07m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4e6f5f54a3d45418b8d1625dccf42ca : SUCCESS in 1h 02m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/761210a1efd84a31bd7b8d4daeb1aa8f : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8e6c92a8f0104b3b8b0f4910e0ee0c8d : SUCCESS in 10m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8481c891208244b1810424bd901dcf1f : SUCCESS in 23m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7810610ec9914982a10576a5c181d1bf : SUCCESS in 13m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cda4dfb045ce4645800adb87fa708ffd : SUCCESS in 13m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/92ce9c72c5ed47c5ad0adb05dc8dc7c4 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9c80281447d44438a2367f539833ebc4 : SUCCESS in 14m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/68b1c5f3ed2448d1bd72eee2bbdd8d16 : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/26a38a5c03fa4aebab7cc1f1c768a84e : SUCCESS in 8m 18s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a559830423bc431daff0ef332e0a33c3 : SUCCESS in 9m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab07c92038a24a8680b1f4448f82cfb4 : SUCCESS in 42m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df4e6ae00d76401d92071ebe405f7d81 : SUCCESS in 35m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8cc25d3c92b84a49825f4a9b9ab7a256 : SUCCESS in 54m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bf391d86d8a42f3a5abfc72f28e0316 : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f154eafa1ba648058deaebe4d1b130e3 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b069c8fecdf644d190c3fa0b40c18f82 : SUCCESS in 17m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56602a932634470ea0e999b1c5e494df : SUCCESS in 46m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb4468bcac354d16a550d84e724d0fae : SUCCESS in 30m 46s","accounts_in_message":[],"_revision_number":3},{"id":"81a3129d4d9c02a61b58e7e4dabc8158909179f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 13:40:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6dd5b4f2c7ff433323f24f5468f614e3d02b5f34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 13:41:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a79312fffdc40f794e399a0f0285ad8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bb327c179264fa6b13d112ccf8406c3 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e6f9502cdd1544df81d86f95283e8fce : SUCCESS in 1m 18s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"be62933d006524390c5ce2dd5c832ff016e02edb","revisions":{"3c87e44bf633891c050f87cab2aacf4be4f94d65":{"kind":"REWORK","_number":1,"created":"2025-09-18 23:43:14.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/61/961761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/961761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/961761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/961761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/961761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/961761/1"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-18 23:42:57.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-18 23:43:08.000000000","tz":-300},"subject":"wip: relinker: allow clobber-hardlink-collision","message":"wip: relinker: allow clobber-hardlink-collision\n\nChange-Id: Iaae0d9fb7a1949d1aad9aa77b0daeb249fb471b5\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c87e44bf633891c050f87cab2aacf4be4f94d65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c87e44bf633891c050f87cab2aacf4be4f94d65"}]},"branch":"refs/heads/master"},"1980d36ca98c5a78c202962a93f6b7d877ba7d54":{"kind":"REWORK","_number":2,"created":"2025-09-19 20:52:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/61/961761/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/961761/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/961761/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/961761/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/961761/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/961761/2"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-18 23:42:57.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-19 20:49:53.000000000","tz":-300},"subject":"relinker: allow clobber-hardlink-collision","message":"relinker: allow clobber-hardlink-collision\n\nThe relinker has already been robust to hardlink collisions on\ntombstones for some time; this change allows ops to optionally\n(non-default) enable a similar handling of other files when relinking\nthe old\u003d\u003enew partdir.\n\nIf your cluster is having a bunch of these kinds of collisions and after\nspot checking you determine the data is in fact duplicate copies the\nsame data - you\u0027d much rather have the option for the relinker to\nprogramatically handle them non-destructively than forcing ops to rm a\nbunch of files manually just get out of a PPI.\n\nOnce the PPI is over and you reconstrcutors are running again, after\nsome validation you can probably clean out your quarantine dirs.\n\nDrive-by: log unknown relink errors at error level to match expected\nnon-zero return code\n\nChange-Id: Iaae0d9fb7a1949d1aad9aa77b0daeb249fb471b5\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1980d36ca98c5a78c202962a93f6b7d877ba7d54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1980d36ca98c5a78c202962a93f6b7d877ba7d54"}]},"branch":"refs/heads/master"},"be62933d006524390c5ce2dd5c832ff016e02edb":{"kind":"REWORK","_number":3,"created":"2025-10-13 11:03:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/61/961761/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/961761/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/961761/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/961761/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/961761/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/961761/3"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-18 23:42:57.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-10-13 11:03:30.000000000","tz":60},"subject":"relinker: allow clobber-hardlink-collision","message":"relinker: allow clobber-hardlink-collision\n\nThe relinker has already been robust to hardlink collisions on\ntombstones for some time; this change allows ops to optionally\n(non-default) enable a similar handling of other files when relinking\nthe old\u003d\u003enew partdir.\n\nIf your cluster is having a bunch of these kinds of collisions and after\nspot checking you determine the data is in fact duplicate copies the\nsame data - you\u0027d much rather have the option for the relinker to\nprogramatically handle them non-destructively than forcing ops to rm a\nbunch of files manually just get out of a PPI.\n\nOnce the PPI is over and you reconstrcutors are running again, after\nsome validation you can probably clean out your quarantine dirs.\n\nDrive-by: log unknown relink errors at error level to match expected\nnon-zero return code\n\nCloses-Bug: #2127779\nChange-Id: Iaae0d9fb7a1949d1aad9aa77b0daeb249fb471b5\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be62933d006524390c5ce2dd5c832ff016e02edb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be62933d006524390c5ce2dd5c832ff016e02edb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
