)]}'
{"id":"openstack%2Fswift~962042","triplet_id":"openstack%2Fswift~master~Ia75feb9b4cd903fabb6c082691b81dffd29ac6ff","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ia75feb9b4cd903fabb6c082691b81dffd29ac6ff","subject":"sq: try and add py37 as a bindep profile","status":"ABANDONED","created":"2025-09-23 00:54:10.000000000","updated":"2025-09-23 02:47:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f904fdc1c5270805304b5a2c3d2919af11396e86","_number":962042,"virtual_id_number":962042,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-23 02:00:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-23 01:47:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-23 02:00:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2503655c50558e57697d666268320d891418fff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-09-23 00:54:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc48ce52769dff0fbf4e8a82a4e030fb625362a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 01:47:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e6d0ecd831e46c79cc0e5a706aca275\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5315532f6ce7490a9b6bafbfb08ce9f5 : SUCCESS in 51m 10s","accounts_in_message":[],"_revision_number":1},{"id":"274930262976daa7aa4fd2eb50af9655e9ddb09e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 02:00:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90537668c1c6488d938c84ffd0e95df5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46f5be6abc50494fa79c36a5ae544771 : SUCCESS in 5m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/d7cac05a8d294b52ae7dbea48a97a047 : SUCCESS in 28m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/201139426dfc4384be921e4069f3acab : SUCCESS in 30m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d2a691eda1fa4f4087cd52846d9a954f : SUCCESS in 33m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27d05ef229204713b97ad1bccce74bb3 : SUCCESS in 1h 00m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9822fbe4ff0b49a69671ca165e48dcab : SUCCESS in 14m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71c1a66953f24a6d8997ce3dd479618a : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cda33066e534d0085687adb9e8ac5df : SUCCESS in 10m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c7b131eb7424f3c8f1178c4ad9ea354 : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1f3794823c0a4a9fb769783bff05cb5f : SUCCESS in 25m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9e5e877766454234b592757621021a45 : SUCCESS in 20m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/420342330c454ad081af3f0ba299180f : SUCCESS in 10m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0d0ce4e264434de2b6b6dec6d53caa89 : SUCCESS in 29m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/390b4231a5eb403aa2e48055159fe188 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/943cb5c1cb2648d8a0177205c9d7c7a6 : SUCCESS in 13m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bd5cd4a5f0d94ac4aaaeda2ef4e750b8 : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ecc38142e5d84bc1a2e72716691f2547 : SUCCESS in 14m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76803983fa484d179c461cc1fd096032 : SUCCESS in 14m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/891a9e5922f34e6184508aaad97ef2a0 : SUCCESS in 8m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/af769c211ab04bc49fa2ddad6c70dd1f : SUCCESS in 54m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bbbc029a61934732b3ebc7b844e795b4 : SUCCESS in 52m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/634e63cb7cdc42b3a9a192d058c37b32 : SUCCESS in 56m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd146f648d724da0b232891742a6e511 : SUCCESS in 27m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6923464104a4b89a96c86888f6f82eb : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b525c5a57534bc09402f5a8c0ca46e6 : SUCCESS in 24m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e802a89644244875a423d0e4da993373 : SUCCESS in 53m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d29f530231134de580f1a63ac07b0f51 : SUCCESS in 1h 01m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f904fdc1c5270805304b5a2c3d2919af11396e86","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-09-23 02:47:48.000000000","message":"Abandoned\n\nsqaushed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b47e524254add0423b265623d00d119e4c646be","revisions":{"5b47e524254add0423b265623d00d119e4c646be":{"kind":"REWORK","_number":1,"created":"2025-09-23 00:54:10.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/42/962042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/962042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/962042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/962042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/962042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/962042/1"}}},"commit":{"parents":[{"commit":"8d015d4ceb3d0db4e5302c0ed84d01f8fb0fb6be","subject":"Pull out a bunch of liberasurecode requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d015d4ceb3d0db4e5302c0ed84d01f8fb0fb6be"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-09-23 00:53:00.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-09-23 00:53:00.000000000","tz":600},"subject":"sq: try and add py37 as a bindep profile","message":"sq: try and add py37 as a bindep profile\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: Ia75feb9b4cd903fabb6c082691b81dffd29ac6ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b47e524254add0423b265623d00d119e4c646be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b47e524254add0423b265623d00d119e4c646be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
