)]}'
{"id":"openstack%2Fswift~962205","triplet_id":"openstack%2Fswift~master~I6b10c1f1eca2ddd08d2239f26e3924b42fb35245","project":"openstack/swift","branch":"master","topic":"rebase-relink-err3","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-04-09 12:47:38.000000000","reason":"Change was abandoned"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-09-25 14:25:20.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I6b10c1f1eca2ddd08d2239f26e3924b42fb35245","subject":"WIP: include explicit uid in .data file name","status":"ABANDONED","created":"2025-09-24 17:16:41.000000000","updated":"2026-04-09 12:47:38.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"673d3b3b03367a8606baeb033f46de720af72aee","_number":962205,"virtual_id_number":962205,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-24 18:52:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2025-09-25 14:25:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-24 18:52:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-25 05:45:25.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"ba164d0614b57ed142a640d2f22397e5d3594fa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-24 17:16:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ffb32c9c9d923183e2f36c33e5cd8a81eee9de4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 18:52:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7474a7201c144eb3be55d2d51d6f4238\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd3e784fe6e049ccb20319d4c37676f6 : SUCCESS in 6m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/9b26eee51f99434b9eaecbc2e66192c7 : SUCCESS in 33m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9b68fa01b261404c83cff77e15a9868b : SUCCESS in 36m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75690da1da6e4fc9b52416a5086effbb : SUCCESS in 33m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/545fec73ce2c45178dbefabd3f550ce0 : SUCCESS in 49m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e93b5896a4b4318877e71693952a215 : SUCCESS in 10m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fb639d15fdf4099a0e51c0375969931 : SUCCESS in 12m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3601fa3d596147b8a1c630dd6e8748ec : SUCCESS in 13m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b66a886f28046b8bfd104134f3ee70b : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b094ca6ec1054c22bd97e339411f9b20 : TIMED_OUT in 1h 30m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/20ce60888e7843879aaee3f2849c1afe : TIMED_OUT in 1h 30m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d22a6c0286e24b7a8727cb0fc079bc6f : TIMED_OUT in 1h 30m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/46190d5af3db46719d3086e52fba1bca : TIMED_OUT in 1h 30m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/486180f22df843e89d5a679cc9858d5a : SUCCESS in 9m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f0d62260a72945d6b212b2afb91e6c8d : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/888cfe0ac79a44c09fb55c5606571977 : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a2770287c3e14e228a7f65ffeca990ce : SUCCESS in 7m 48s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/de3d530795444c668d975e48a0f30412 : SUCCESS in 8m 16s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cc55c69c053b4c84b7179685e6cbb8f2 : SUCCESS in 7m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b166d21a274a4fd4889d057ccb210021 : FAILURE in 38m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1828da98dbf149d192f37620de6f86fd : SUCCESS in 36m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b660616f7cb04888a3523db3819ea9d2 : SUCCESS in 26m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de6cc132502c482da0ab090b837bd24e : TIMED_OUT in 40m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de0d153491514830bc7fda79643c4029 : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/47ac708bd65c4e0385c4f4609aec173b : SUCCESS in 16m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/781d0e4de245483fb705f20f1e814aee : SUCCESS in 33m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cd210e424cf45349a7d160ecfe3c37f : SUCCESS in 29m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8bc180651b5ea35c5b76d2440f49e03c8c81ae63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 20:54:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d424f980f0d4a1fbf40fbbfb7e5b19c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ed66a6fc655a49a8ae4880d5823abe37 : TIMED_OUT in 3h 01m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8ecb27f33216e4f483a394c7bcbccc216f8d216a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-09-25 05:45:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdc25b8195a0371bb67eb927c538ff801cc16c25","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-25 14:25:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"673d3b3b03367a8606baeb033f46de720af72aee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-09 12:47:38.000000000","message":"Abandoned\n\nsuperseded by v2 timestamps","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d26abe318fececb1fd3c3121d78ad96354a0a564","revisions":{"d26abe318fececb1fd3c3121d78ad96354a0a564":{"kind":"REWORK","_number":1,"created":"2025-09-24 17:16:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/962205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962205/1"}}},"commit":{"parents":[{"commit":"ca839dbed05c77a0861642021e7064b49bd34bce","subject":"WIP sq: make FragZipper configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca839dbed05c77a0861642021e7064b49bd34bce"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-09-24 17:05:02.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-09-24 17:05:02.000000000","tz":60},"subject":"WIP: include explicit uid in .data file name","message":"WIP: include explicit uid in .data file name\n\nTODO:\n * So far I have just hacked diskfile to support an additional\n   filename part.\n * I\u0027ve not even started to explore the changes required\n   for replicator/reconstructor. ssync will need support.\n * test_diskfile.py needs updating (and maybe other unit tests)\n * make proxy send unique values\n * fix diskfile purge\n\nFunc tests pass! Surely some probe tests will fail.\n\nRationale:\n\nTimestamp alone is not sufficient to guarantee the uniqueness of\ndiskfile \u0027.data\u0027 file names w.r.t. client requests. This patch adds\nsupport for the proxy to send an additional x-object-sysmeta-uid\nheader to the object server which gets included in the .data\nfilename. The proxy should ensure the uid is unique per client\nrequest. The request transaction id may be a good candidate.\n\nThe diskfile \u0027.data\u0027 file name format becomes:\n  \u003ctimestamp\u003e#\u003cuid\u003e.data\n\nThe idea is that the newest \u0027.data\u0027 file is still determined by\nwhichever filename sorts highest.\n\nChange-Id: I6b10c1f1eca2ddd08d2239f26e3924b42fb35245\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d26abe318fececb1fd3c3121d78ad96354a0a564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d26abe318fececb1fd3c3121d78ad96354a0a564"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
