)]}'
{"id":"openstack%2Fswift~962305","triplet_id":"openstack%2Fswift~master~I4ad1e22fd011652d80269a76353e5fc14db249f5","project":"openstack/swift","branch":"master","topic":"cohanaja-cycle-time-recon","attention_set":{},"removed_from_attention_set":{"38368":{"account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"last_update":"2025-12-04 23:03:33.000000000","reason":"Change was submitted"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-12-04 19:57:41.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-10-11 03:24:35.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-12-04 23:03:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4ad1e22fd011652d80269a76353e5fc14db249f5","subject":"Enhances sharding_in_progress metrics in recon log","status":"MERGED","created":"2025-09-25 19:07:08.000000000","updated":"2025-12-04 23:04:22.000000000","submitted":"2025-12-04 23:03:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":134,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"962305-cohanaja-cycle-time-recon","meta_rev_id":"bfcdbce11da908439cc5d8d2a7be4a04af685456","_number":962305,"virtual_id_number":962305,"owner":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-04 23:03:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":0,"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":2,"date":"2025-12-04 19:57:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":0,"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":1,"date":"2025-12-04 19:57:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-25 19:08:33.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2025-09-25 19:08:33.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-09-25 20:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-09 16:17:49.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"165268c79f74ed9a7525039438188490ecfb5113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-09-25 19:07:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8686f8235fc6e57b3dc3b8e6183fbc77ab6cf315","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-09-25 19:08:33.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a3bb988b6e1a7cb0eed024f6e12d050cfd2c24cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 20:19:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4db2a977b5ab4a639d46051aa49698b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e0e58edea044c8083e65a6a8346a520 : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/461ca5ff07764388b1123028094e9b3d : SUCCESS in 1h 05m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/24d8e7634d1041c697eb1127b99ad651 : SUCCESS in 56m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aa68ddf329b645c093f25ba8b940f44f : SUCCESS in 42m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/836d52649e654c1b8dd14822dc303c41 : SUCCESS in 1h 06m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/332ffd05d9874bc5b6e5f5675c008d5a : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36ae213b94fe4101ac54282753467c2a : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/059aa10839db4ba8b4d95c1774c736f6 : SUCCESS in 15m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c6eb3426d1a148cc853d9659fb2e441f : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f043727895d943f2b67b8e26b477a09c : FAILURE in 21m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f1db5afeebae4a0c95c6f7531e6966bc : FAILURE in 9m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c0e835f1f373417eaa92fc409f52c009 : FAILURE in 27m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6ee91b5b30624c6b80e625f100edcd6f : FAILURE in 11m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c9f7661904de4b509e2e7bfaa42fb3ae : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d09e35d9de12483495f614e2a578a562 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bbd98f665f294fc79bdd4cfcbe9d03b9 : SUCCESS in 16m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/009c57e162c840dead4c9c2c59f86a5e : SUCCESS in 6m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/98ea1cf34f644200a2e6614d8176f42b : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/047acdba59c14571887be0868548b513 : SUCCESS in 8m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d96aadded3ab40a187d44aacdf0608ee : SUCCESS in 47m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/71feeeb6da1a4bf2abffc14e84f4d9d2 : SUCCESS in 22m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/56ca6eea89314bfe9ca45f87e004dfee : SUCCESS in 48m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d516df427b854f0c931a1172a4295599 : FAILURE in 26m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fd3f33a3af340b6b29008a3d27f7075 : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bb0602263f614f219144d3fb8c0f9d52 : SUCCESS in 17m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41fdad43a5774601a7e7050fdb4c3340 : SUCCESS in 33m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bae4957077641b1af7d14c660fe4e45 : SUCCESS in 53m 04s","accounts_in_message":[],"_revision_number":1},{"id":"cdc7245b6492dddd3d9464e961d5b7b5822e62bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 20:50:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a52e43c797e54aeba60dbbb2b1893977\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c5a54d748b04919b301aea8bc10d206 : FAILURE in 1h 08m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c99c6b30e0a256565228ff0ee2f14697d2d55a46","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-09-26 05:26:32.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f492767b4473e1ceab0aad47ae6e92458e4e2c5a","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-09-26 16:47:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"629c0db60ee7d42615f78c5661f0e06ad9b70541","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-09-29 16:38:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"067c534d0862f45ed6bddad2ca609ec18bf4d6fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-09-29 18:43:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"beea7333ac9339e9ca5d6d8adf12ba93bbdcf02a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-09-29 19:05:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"751959f941f7bddcc9211702d015d84edcdb3afb","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-09-29 19:08:49.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"85f25995dded23a9b5a02c7570e02b55ed0ee45c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 19:59:01.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a4f78cd449c403f87e9cfcd15f8ccb2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b82900b4c8d648eeaac0f339047b4088 : FAILURE in 51m 30s","accounts_in_message":[],"_revision_number":3},{"id":"d2fe4c0ec58ed48b5ced2a1b5bde1d3c5accf174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 20:09:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c11a585245d4378a2707ecd522df915\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29fb77132d7747379529a27c7cb665bd : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/d5446c5d937d48f89b96ac9759028786 : SUCCESS in 1h 00m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a7e05edfd084f88a3d886ea18175c3c : SUCCESS in 31m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0c34b0062da1411d8babb256ea458ebf : SUCCESS in 32m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21b994002f5e48caa37bc46fd9640f4e : SUCCESS in 59m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1e06b5b8dd143f98fb1dc242fb983dc : SUCCESS in 11m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe4b68faaa9345eeb941a87b770122a0 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b59577695d8e4b669717fb985690df1c : SUCCESS in 21m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fcec6ab2087348ac8f3d13bda7d6d81c : SUCCESS in 9m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/56ad1476c0314ac18cc5b95bbaf92961 : FAILURE in 10m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1f278cb64d4f41638176d5431618d5d6 : FAILURE in 9m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0652376645554d73be608e157130c2e0 : FAILURE in 27m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4a3918c27fe34c1d827579cf14930a9f : FAILURE in 30m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cab25748f0ee45e68fb752ecb062d759 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c84cad7bc72f4f349437e4550d4eb744 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/85b9561b93f844b7bac04e6bbeccb1fc : SUCCESS in 9m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee87969cfb0d44fc99cf6972c17f761e : SUCCESS in 7m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d38c2ee4e5234404aad7b6361a9a60d1 : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f04a142c1bc6487683dc1e01b377b5ed : SUCCESS in 4m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/24d307fed8ed49b28e108cb9ca68fbc3 : SUCCESS in 58m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df4dd102c1c94be9b84fbe1f76001734 : SUCCESS in 51m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab6d77e7d450463caa7240792bed6acd : SUCCESS in 47m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4af7c28dc584bc58be5544d1c34d152 : FAILURE in 26m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e309b52403e3434091db5d3bc94ec538 : SUCCESS in 2m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11a82a60b89d4d559ef37af24dda1163 : SUCCESS in 25m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/237c5c2f580345388d6e0294e7b0fb67 : SUCCESS in 48m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41a1a3a80caa46309763f91d3f64e6fe : SUCCESS in 27m 43s","accounts_in_message":[],"_revision_number":3},{"id":"b80bb93258a535c9f47ca98ff2e5cb5d33a0c25f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-02 18:36:06.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"de56dd2b7bf072b2ecdd4509771f685c9b84de40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-02 18:36:39.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e6bc066a1541c9d711991bdaa533bf4e02beb4ec","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-02 18:39:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b1bae1a2f510a78d021e81e6c3057c17fd63102e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 19:49:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4b7b2111db0471e89d5a5ebb8456d7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89c79a3f1fb442f48edf1894bd43e56d : SUCCESS in 5m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/9e5c6788778345dba507f2862002c642 : SUCCESS in 1h 06m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd106209861b48459a82ef03e9cc54ce : SUCCESS in 1h 00m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8fbb40242ade484d8a86316914bf2ab5 : SUCCESS in 33m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ae2ffe9e23a48f5b46635b269340259 : SUCCESS in 50m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4081e7dc5e746a7ba0eee2cb0be801e : SUCCESS in 12m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07f0de37166f466e90d5a24042e26061 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e22075fa9ed14a51aa5b3a33448fdde1 : SUCCESS in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9e2d700b721c4d3d8c9278213359ef62 : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/038e9746b2f945cd965492a7964744f0 : FAILURE in 24m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ca2a13994384b479c59658418cb8e9b : FAILURE in 16m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/243b44e0c3934bdaa5c70a2f3e1e899a : FAILURE in 28m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/04d14d11554e440db4d81caa6329334f : FAILURE in 30m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d791234c434a43fab86840bbc994dd27 : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/25b3a1569dfa4963a7a8268b273d4911 : SUCCESS in 12m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/20f91ee5139f40a69a73258affa01671 : SUCCESS in 16m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6db35d57ce4e4365b763789350f8cf19 : SUCCESS in 6m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/64e591d063e34e749d5a3193bf45638e : SUCCESS in 14m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1aa95a1153394da7adc6dd12fc261363 : SUCCESS in 8m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7125cd9f72784c2a8b83853911303e58 : FAILURE in 59m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8dfe703620e64770b5167d79ce73146b : SUCCESS in 49m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba5ab88561a746dbbbdc49da19dd7721 : SUCCESS in 51m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/678b5554833842979a12692ca3e7aafc : FAILURE in 28m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c42df07216c54ba491ce3e3869757e2b : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2ca01ce4686c4ccba77a89d548a88492 : SUCCESS in 18m 58s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07c4393e3ac94c99813a2f27a0fb5f16 : SUCCESS in 57m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5ce9f678f26410ca4ea4c8a4ea4085f : SUCCESS in 47m 54s","accounts_in_message":[],"_revision_number":5},{"id":"fb7d31ab6c54b33c8271f3c55dc9cdc7c5b5499d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-02 21:49:38.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"67739f163c0fbd28cdf341ef28d2dba0506a4914","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-02 21:50:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6126599ba95e9102e41bbc9eab504ce3825ed338","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-02 21:57:08.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e913d4a7e7ef563899df4bc0a5e96201eb5ad7c7","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-02 22:01:12.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"572475f46344a80a467cf7e320cc5facbc249dea","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-02 22:45:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"48549f79e4698c18582cd476d973bdfc53583629","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 23:11:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/600ccd3ec4a94860b6366c49be4ed4b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/572728486e8345059d117beb6b3d7934 : SUCCESS in 10m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/3983668a52bd483fa1e75591ae216a90 : SUCCESS in 49m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25f97a8bdce84c2ebd0b5a8cae9a6575 : SUCCESS in 1h 00m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6d0924695ee645209d6d64b7f60d342a : SUCCESS in 39m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfb90e0194e24c26b641e892aeddb6ac : SUCCESS in 1h 03m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63f129cabe784908a192cf76920a556e : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61acf046a02642c79faa3c839c15fd32 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0655340f07fe42eab73906acfb733c4e : SUCCESS in 20m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e2a6b3f83634325bd66ff676bf365ac : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4f8011a1027f449c933ad88248ca8e99 : SUCCESS in 25m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6773ef1bfec14662ab423c67a09e16a2 : SUCCESS in 21m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/25ad4cfe95854b159fbf1843cc899080 : SUCCESS in 33m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0797c54952344136a11616b9ee2bc879 : SUCCESS in 30m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/47cdbe524350412aac1acffbf4f8a106 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4dac648e531543a199006ad3e0c18a38 : SUCCESS in 13m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3e474ac239e44d92a401c20bb02cf6b7 : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/41b5fc360d994f2eaca13f32269e99f8 : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61ffab531b3e4ef3857f1b3e6a7de613 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5922cd9a240b467f8bb9236beb740458 : SUCCESS in 8m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/68e19bf020e54eaaba4c47e7e0f67c25 : FAILURE in 1h 11m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b900f02d4464095ace689f6504c2da4 : SUCCESS in 48m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4bb9cfe7bafd4eeeb9f797f001a8212a : SUCCESS in 38m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0e5259af2b3494b8c5df9c4f2f61f0c : SUCCESS in 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db061d6a76464ab98cc53bd2f3ca1ac9 : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5dc767fdaf4447d6a9c1d3411be7f22f : SUCCESS in 18m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77ed66e4c96149ad807d24ebc3f7f4e8 : SUCCESS in 45m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddf4d5279be345188b33602fac04fb87 : SUCCESS in 28m 52s","accounts_in_message":[],"_revision_number":7},{"id":"3b07c1e8c67f8d1de6eadf3c0982256475550e77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-02 23:26:13.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"bd6f3a4bf88150b6d97a086c725c611040fee708","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-02 23:33:55.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9e236160fb8990d39cd1b6a663159c1924c8a75e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 00:25:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e7bcc727f414a96b137def6da135268\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b73a6cb0a784a8289e78481da425443 : SUCCESS in 6m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/21fbb58d26b5412aaabcd6f4712c87af : SUCCESS in 35m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ad71cdaf55b4738b63be255998eee05 : SUCCESS in 48m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c963522de9f84e959749ee4ce8786b23 : SUCCESS in 39m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6eb6eecac3b745d1a1cf3b9acec51a66 : SUCCESS in 46m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d02391f7a5d8475c835d9f8d0accf058 : SUCCESS in 14m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38c3583b69a244aea0db74a15591febf : SUCCESS in 17m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57c333bc2a484d92814a6beb137cbe24 : SUCCESS in 13m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9ce6f20ba0094fb2a1d7c6dcbf99b834 : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e69a200c6ef541d09cc9d5b8d8b5f7a9 : SUCCESS in 10m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0d878846ed6b49bd853363cfe2e84b94 : SUCCESS in 10m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5651469131364cdb9225a8cee964dd6b : SUCCESS in 28m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4475ccda1f414ed0b04a3b4b39d5ed26 : SUCCESS in 32m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b23e6f622f3f4870b82abc8919f52aa9 : SUCCESS in 9m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/49068553942e46b2bca216c84201450c : SUCCESS in 8m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b122337091f34b34a5bb3b5c7bbed041 : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cf25f24f0c904b4382f786950f145261 : SUCCESS in 4m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/62723f0f52d445c5aba4577af34e79b3 : SUCCESS in 10m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9588a46d8ea743a4b0f637ce61daeabf : SUCCESS in 6m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/392b2bbf67c144d9afb5c7d7cf722517 : FAILURE in 39m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b21bc50a5cce4f60a6e9846fc257b526 : SUCCESS in 54m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3429f58f7a2e4a5286da86b32c5febde : SUCCESS in 35m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdd1c915b77a475589e03eff673c8405 : SUCCESS in 21m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63c9117dce4948898767b51ff7b33ce0 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d907be242c904684817ac8d8a93e07d5 : SUCCESS in 14m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c20cd959f43d421095cd4ba00cd8d6f9 : SUCCESS in 37m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb3f1ca41f8c4a75b18de7e83c85606f : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":8},{"id":"896ae08dd8cf1470f53353c0b801bb5d4f6a8c68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 00:35:52.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bef9d0d9c6b49cbb69c2819bd61f4bb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6daec570e394465ab7ef370bfc108618 : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":8},{"id":"1089eab2ce5b3082c23710d4ed8ca1e93e0f01f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-03 16:42:39.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"29a8beb7aa96f58d8b2f2d046815a57abb4511cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 17:40:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c689f5c0fc5e4928b1b90ed7499c85c8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a793d0c5b0b4e70ac2d90c303583fc6 : SUCCESS in 53m 52s","accounts_in_message":[],"_revision_number":9},{"id":"6a2a1541cdd21d365c2a8a4c79871536212cdcdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 17:51:34.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bda9a8399b5042e09991bc53d9fb3022\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1939672490764cbb8efb15e579b389e8 : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/526145d1bf2b4b61b5a59d1c64637ae2 : SUCCESS in 54m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/03001fc9aeed4bff8d73fda3c8a2cbc3 : SUCCESS in 1h 02m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6cc840b20dc14408a10a1670e0d2e936 : SUCCESS in 44m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10662bcbd958415ea73d7d9ff7d0f4de : SUCCESS in 44m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f59080ca28234d35a0636d8691771bc2 : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0e96c39f3df4ffea90747ad83f85aa9 : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/848d222763714510a8f03ad1cd1cce2f : SUCCESS in 23m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a3c99779926841ca84388c9a02fd6b7d : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f2123ab828af436ea4005ca83e361e85 : SUCCESS in 24m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bd70c4be89bf4236906076d15bde528a : SUCCESS in 20m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5e3418985cd849f2b28f038c8bc05783 : SUCCESS in 10m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/14c216e2c8084ee59b845fc3ae599b7c : SUCCESS in 29m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0a240eef0d984bd3bca394f03941d496 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/233829b453ea4472b278ecdea8ade39c : SUCCESS in 12m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0a54b39a7fbd4eaf8c3a900fbdd47f54 : SUCCESS in 13m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7aa54ba0bfc54b72a157df9e9e7d3880 : SUCCESS in 6m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61be0dfd188046aba7925bfdac24e7f5 : SUCCESS in 10m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/73995bec5a214dff86f2c7aeda855854 : SUCCESS in 7m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e7ba0b76b06482585b0c0f8bf718781 : SUCCESS in 1h 01m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f51533b67c2e417e951970022c622f65 : SUCCESS in 21m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb6488626f4547c290d4c8e45dc0344d : SUCCESS in 34m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe7518438eb6404e8d830bf64d19a751 : SUCCESS in 21m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e35cbfe3fd9f437f9662377191ef9bae : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/14c3d9f5f7014b74ad1fefb2f3e53e46 : SUCCESS in 17m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa1a96f446b54a8c809cb6b1e3cdcabc : SUCCESS in 57m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d3f9b94d2fc4020bc93a48f8e7107fe : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":9},{"id":"faec4db9858ba2a296970d34663f36622f8b5067","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-03 20:36:54.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0dd2ae0f9f44259cefb8989ec7c6e6598ff450bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-06 22:33:54.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"9d43a75107b20f57d6b00aa864461f4781ca1870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 23:48:01.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e6aa506fa9f45fbb7b6cec488061f60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39503a55fcac4cad9fd586f3bdcf855d : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/4b628938a49546b4aeb4d18ee426bf3d : SUCCESS in 35m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/24545233fa674b0c9556da41d5fdd295 : SUCCESS in 1h 09m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9cf07b1bbd0a49c18bc9735c0a574cc3 : SUCCESS in 49m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1dc978bf9c24b2e88692156c5f3e69c : SUCCESS in 1h 03m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac95a628c3d64ccf99644df7450d5743 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37383a5282b14793a80775c9ca6c1851 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d752e23f15e43f9bb0108388431467b : SUCCESS in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de719d4ad3b34b4b920ab96d1d05b464 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/586a73b6d5da4b3bb339f4006c92d1c6 : FAILURE in 18m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7dd1888d19bc43fa8cb3f6881bbe8cd2 : FAILURE in 25m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d9a33f356f9b43048673d20b6194b979 : FAILURE in 12m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/88995884cb4349cca8e76be47ee1ce67 : FAILURE in 24m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/aee11521b7e9451e95102e09a6ca3dbd : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/740f2e42917042e88b126a1da60bf7f2 : SUCCESS in 9m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0e7ea4b220c548d091b0bdb57fd8b212 : SUCCESS in 13m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3f04cbe40cec450590ced705846789f5 : SUCCESS in 5m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9f98664a89d84017833bc20cd9898154 : SUCCESS in 9m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/79aee53ed2fa46bab985491271c3a571 : SUCCESS in 8m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbdcc6d80a6441efa1bbb9ce02c226c0 : FAILURE in 1h 04m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/63390c01898848a39b5d03f247cd8172 : SUCCESS in 49m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3950dd6f8d0e4bb3a715bcc5f958d5ca : SUCCESS in 37m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8138c46553ea48d3b25b83e787f0025d : FAILURE in 32m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ec42bf297b8478385f8da7c6c33aa0f : SUCCESS in 3m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39cb3a019e064a1bb613d262aafbd01d : SUCCESS in 8m 45s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1e26c81e96c47ebb6b5ff9838cee691 : SUCCESS in 37m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55a21432f55f4c9297e6c8c3714e6c40 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":10},{"id":"5b806951a5abff600e8e11652bcb8dacc46720d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 00:18:38.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47b0d98fa53940c68b51014f367b541c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eb277c3604ad4d4ba2d39d3ce1f20b19 : FAILURE in 1h 43m 01s","accounts_in_message":[],"_revision_number":10},{"id":"82340b6a65c90275148bbde56dee125dec8dcd38","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-07 16:46:02.000000000","message":"Patch Set 10: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"01968b804bea013d16e53b7a59f3f415da317839","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-07 18:10:41.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"9c1801f2e73260a4e8c27bbda3261c4c78b89646","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-07 18:16:00.000000000","message":"Patch Set 11:\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"f1d752c553710c4ef936bf0aca63e0e0efc2d018","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-07 18:17:13.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"b0262ff367cf7ea74f6a6d43dad3a6640d874c35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 19:10:58.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c2679c05582407cad330e0b206ae92b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/195831e7b8a14b15bdb3b015e2508d81 : FAILURE in 34m 02s","accounts_in_message":[],"_revision_number":11},{"id":"c8946fd185a2557d54e564a3abcb5a203015a555","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 19:23:42.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/353b55f4e5944ddebcd69a493a14a4bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a742939f984157ba54255c291d78ac : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/0f4d4eb2b6694f5e98f269c95e033082 : SUCCESS in 1h 02m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ed1266649ba4f42b7638937a4af50c1 : FAILURE in 45m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0237f489086149e49d367d36e760e88b : SUCCESS in 54m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1aa2cb12b4ae4e83be4604d47d1327da : SUCCESS in 1h 07m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7f26427076640d79042c2daa16ea117 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d64bf033025f457d993d436f36571a70 : SUCCESS in 21m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc84623d4fd54ea9b350593cd3b1b3b2 : SUCCESS in 10m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3d5b757df5664dcf9a912e70aa2d11a2 : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/579c7096ffd14d6b91dc713dc1808bcc : FAILURE in 25m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b91a10e87ee84dc1ad07eaeb47b01c84 : FAILURE in 21m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/499f334c63e54c1d89dc53da21f78263 : FAILURE in 10m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a444a81f715b493ea3ed9abcb4281476 : FAILURE in 11m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f4a0593e2625455aa7345abb47306161 : SUCCESS in 12m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0fa8b8f61608445784970b14d209b238 : SUCCESS in 14m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/96b750e9dc174d1b9cc3e20d588aaa0b : SUCCESS in 8m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e025e3d97c941d380466445885dc8be : SUCCESS in 5m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/494fdffdd0074eed9ca22985c2d8675e : SUCCESS in 14m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/463313ec3c384965bf93017a4791b667 : SUCCESS in 7m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f97f0cbc24214d21a63572e60ae7ad18 : FAILURE in 47m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dba9b628ab134d5998edae8af0e78cd5 : SUCCESS in 47m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5243ca83e1f440a08992df30bbcfaa36 : SUCCESS in 39m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0114bee076145999b5602876b973a4f : FAILURE in 28m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1020fa7b4efe443195dc3be4babc1c19 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1e2465f933b49858b1a1faf3438fa9e : SUCCESS in 15m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8851c6218610478385d5eaab1f881f6a : SUCCESS in 1h 01m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3071314ac5f401fa8fffaa9f531fd03 : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":11},{"id":"8a55562800846082ee59c728342b61de05f2a682","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-07 20:36:43.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"a79b76302ab67efa0dba3fb780d2f42d5ef71cde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-08 15:14:38.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"6efc3714c7f7d89d17832a80f5475a0c43710eaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 16:23:47.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c32c1f33dee14dfd95f68579863b59ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed37a413114f4e73b3c1a67ec75498a8 : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/919ba2bb8cef43d68e76c26ffb6945b4 : SUCCESS in 54m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94f5abf272c4487c82c5290ca9cc71b9 : SUCCESS in 58m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/619251ec9660410d936b467964345dde : SUCCESS in 54m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/079cf9b02054413cb845dbea5ce0f2b7 : SUCCESS in 1h 03m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f9ef74a39d5404a92972f00ed3a9242 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/90d8e88eb870422bbca5fb73bbb755f9 : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/110ac48e6f9943be8de3f3e63c318204 : SUCCESS in 20m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c09a216e31714eb4b2168e62eae74635 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/86ca966328c74a19bc929c25fee5fb85 : SUCCESS in 22m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bc9ba5bc38ff4bf1ad1b2fbf30f88b41 : SUCCESS in 19m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ea197c2be2f544e6ae8afd3d5036ba59 : SUCCESS in 29m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a8927e40625a45c9bd5db1b87777019c : SUCCESS in 29m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fc0fff484421484699bc595b72039dd0 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/28e6da52c91441409d723b1dacfd5dd0 : SUCCESS in 13m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4a7a5629585a409e94614b9197d42fed : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/15342bd8da5846928a804e973a71b7ec : SUCCESS in 6m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/82084bc7150c401899d2a52b60cc6e4a : SUCCESS in 11m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d99bc4c274ad454c830ca0c5254c569b : SUCCESS in 8m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/19050d28537e4ca0888e76d4cb71804a : SUCCESS in 34m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2cf5b70c82c64daba53f3c092d02b7b9 : SUCCESS in 56m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9bb3e345e88d4f0a90b9dfd44c7f99af : SUCCESS in 37m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10ad608bb7e34b6e92bedc0bec2dc2eb : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca9e4a8ee56c4b1cb9e8702df29f6b0f : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a826534c62f6456887e020386f84b384 : SUCCESS in 21m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c4d5dffa297403dbba09228354aaf6c : SUCCESS in 34m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/632e63476e7a44149499f91b811c9447 : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":12},{"id":"73b9c57f48b9f23cefcdcc299a84e284cd6cb43b","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-08 16:36:00.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"fcd2dbb05666111e53d7f075f90a0a985d5ad858","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 16:59:28.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/925be94e171640798b29911a9848e9df\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d00b02f5dbb94813a5296d9de53fcc5e : SUCCESS in 1h 15m 01s","accounts_in_message":[],"_revision_number":12},{"id":"f1cd5a31e6981703f79a3de94d97335f7af611e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-08 18:18:07.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"8f0c82876ad89abaada3cd647e4a5140aabc7549","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-08 18:20:34.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"0e65b21e6c708a08cdbeccceb5cc2645849019e4","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-08 18:21:01.000000000","message":"Patch Set 14:\n\n(4 comments)","accounts_in_message":[],"_revision_number":14},{"id":"f2d7fc468b84224627237a3a3b5439c8af5e8e6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 19:31:52.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4979ef7bfbd44e3d97ef181045e71ffa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b950e46af676472985d618e77a190723 : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/87eef8dc371d4526867f30227475150b : SUCCESS in 56m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/400220eae4674cd58d0b80080d6caedc : SUCCESS in 56m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b2789f1c2b0945829ba91dda0040e096 : SUCCESS in 50m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e5acaea1d94434ca00eda6fed11c1b4 : SUCCESS in 58m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8e7c4c570b9442b9116613caa0326c0 : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f5e530d04c247f1ad48a8458aa9c2b6 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd32d8ad0e8043d0b8b27351a9ae1777 : SUCCESS in 23m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9455dca6cf344a139e2017704f075a11 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/65585a442442472088ae9e46975149ea : FAILURE in 23m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d430615d390f4c6aa13c209f216311e8 : FAILURE in 22m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f965bab7ff7e450e8e4b5b898c7443a5 : FAILURE in 28m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e81391c9ac3644b8b51e6bd3a24cf4ac : FAILURE in 20m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/59350a78d7ab41569e60d8267626e970 : SUCCESS in 12m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ee817aa427d443cbb9ec2b1219fc03a7 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1f25b54e22b4489cbad82489ea73dfee : SUCCESS in 8m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/beff66c925b94b40ac55754e2d8f08a7 : SUCCESS in 9m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2cf913dbaf914125a72b90a9c9e2d046 : SUCCESS in 16m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2755b29c803b4718a42b5bf5c96c0027 : SUCCESS in 5m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/875e44483e094099891cb4208418a876 : FAILURE in 43m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c6cd050ecb5440d9ac2f462ff01622b4 : SUCCESS in 1h 07m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b13c2649ef84b5f88f7a7e10447a44f : SUCCESS in 22m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ff84c1abf32437883cf68d1d590bcbb : FAILURE in 27m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082486389e7b4a9eb8c4e208319bbe88 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d8a27f5cda694078a8dde16f8b3128bd : SUCCESS in 18m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/367169fee96f40d4a0b69ae846dcca7b : SUCCESS in 46m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e7463407c1c49579e09da20f5096822 : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":14},{"id":"dcd0a0f54690c81f63b7c5d0b590ee96d7532b94","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-08 20:05:12.000000000","message":"Patch Set 14:\n\n(8 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e51fac7f579926566974028953425df4308a6986","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-08 21:04:25.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"e06a0733be0b2bfa5165e6a3c378dbcfcee3e88e","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-08 21:05:29.000000000","message":"Patch Set 14:\n\n(7 comments)","accounts_in_message":[],"_revision_number":14},{"id":"203e1ca4b5cd046be8de8de66dfacb4250d30fdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 22:18:33.000000000","message":"Patch Set 15: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d55136230fac43fd85cff3986bc0c1fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68ddee3552ba47ccaf06d10aa07fee88 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/01e91d0edc6d48cfab75174f89ba47c5 : SUCCESS in 1h 06m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/131547f1e41849578f7031a20466cb49 : SUCCESS in 59m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0727580d1f25401381cf1615ac17d026 : SUCCESS in 50m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74e930312a70464fb01be86ac3d533df : SUCCESS in 32m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef7f3c72983c464592b60f28f8073181 : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d44b4dcba614a358112c14f8a85034e : SUCCESS in 12m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03c1c045cd9044309b124c6194ca79fc : SUCCESS in 18m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e7c9494d6173457789d1a1a171708e93 : SUCCESS in 4m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aef9a6fa779d4b64932b94669289bf29 : SUCCESS in 28m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a844c0b5681c4735b4554cb2a6b31fb2 : SUCCESS in 22m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ff85d641b9f8431d9ea186af7e5a5468 : SUCCESS in 26m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bc5616ca86d4404f8b0d77e6c896ad15 : SUCCESS in 25m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/747bc12989144db1b29d890b8ba54ecc : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ab9a730309cc4d42b38216f6fd54d4e1 : SUCCESS in 15m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4cf52ca9c07e45cdae8921867cc07f95 : SUCCESS in 15m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/86b0e5cf84b04f45b71134a9e12558ca : SUCCESS in 6m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/72d1f4ed215648d1b809fe39e242b71f : SUCCESS in 15m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/df6ea6b570964cda95492010a30a62a6 : SUCCESS in 8m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c11e5d931a842ce8e5d550dff413b5a : SUCCESS in 55m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ecd7e413053445029717075e8301ecca : SUCCESS in 56m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/13bc7e582bfb42c7b689f199bde31916 : SUCCESS in 53m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/138dfcffacb44c258a2384fafc088371 : SUCCESS in 29m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b675995b6a444c968ad4ec533ab98b9c : FAILURE in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/51a3a5399e1c41849bfe302700c6c00d : SUCCESS in 23m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/796c5b4cecf646ad9a7486871de7fcdd : SUCCESS in 1h 00m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4dfbfe7eb264ea3900597a9fac37c63 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":15},{"id":"26662c9adc7576cf30dde7a1964c5a0db534f321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-08 22:38:16.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"bbea940cd0c54afd015575b40277023418fbcb7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 23:44:59.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db74348f5e3748f9a9264c0375f5b9cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ddef14ee004bc68baf3e14ffe7c827 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/ac06b8e99f97445dbfbb1ee2d45f19a6 : SUCCESS in 1h 00m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f86124d564af4b00adbf0060e1f3f976 : SUCCESS in 57m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9340843bb6c14b85acc9ee73eaed09e3 : SUCCESS in 40m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/770c70c850bb4947b87f1488df18815e : SUCCESS in 1h 01m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee21fe9f828a416a86633bb37c068301 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5126979ad2f412f8be075fc74c795ce : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51ce2fa3ff6f492fb3cd428aa8e00fca : SUCCESS in 18m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce312e4af79d4c75b29fa7819b397b35 : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/18e5a55bfd644d8280ff3d787aa72b0f : SUCCESS in 11m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0a73fb7bb08542209f515ef0c0e986c9 : SUCCESS in 11m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a730248f4cc14c73bc52b82ae21fd002 : SUCCESS in 27m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e77df527995c40cb98f48ec71a14ff64 : SUCCESS in 30m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0e5ef5a129d9490680a312b5d70f5f67 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3a00084c603b45e4b87c36d7de56f30d : SUCCESS in 15m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5cd39421462b417a9a8239ad26d76809 : SUCCESS in 13m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7eb2b045ffb4287ac3501195ed3de97 : SUCCESS in 8m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bf6f402dda4243f09dead9aaa8eeb7c6 : SUCCESS in 12m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/71481e478cea409584bb436ba83d052b : SUCCESS in 8m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e13ce986edd44a8c8d82b9595b25c1ef : SUCCESS in 1h 01m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e7023c831684ca4b827f69cd2786608 : SUCCESS in 55m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea9a85375efc4a51b9e333c661182c01 : SUCCESS in 52m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e44cc989f15247de843b657d9107596c : SUCCESS in 22m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5f77442fa1b4845a9b21f13fcc852db : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cdcbe3ef1d349d689d8d20d53d1ee1d : SUCCESS in 15m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a82b5f9387a8448d8304bd9a8935bbff : SUCCESS in 34m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/349f245d004f4965a97b898cb9992ac9 : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":16},{"id":"05a70e0a2a6619f4290a53d629cc76c7b3b6587e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 00:36:50.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2081eae8b41a496a9d11ace23d9c5021\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b1a003af84724531aea158795c871c48 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":16},{"id":"82c4be6ad12e1f51cf4c7a7a4817bf5e4dd7dd53","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-09 16:17:49.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"d9ccef0a44e8dd065c20991b77348db915ee993e","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-10 03:28:56.000000000","message":"Patch Set 16:\n\n(11 comments)","accounts_in_message":[],"_revision_number":16},{"id":"c1b640669f509b3711af83945c1c4f557bfa09ed","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-10 15:57:14.000000000","message":"Patch Set 16: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"fa35b0408557d7060da1c7df095c0b51f4a35279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-10 17:17:03.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"9e429f4bf023d48b922e9de20ebcbb4576a0046d","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-10 17:18:14.000000000","message":"Patch Set 16: Code-Review+1\n\n(9 comments)","accounts_in_message":[],"_revision_number":16},{"id":"f227386366be995d2011335887a3ae8d4155f7e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 18:06:15.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ee387f07c8946b5b9bdcdb42cc1f13e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8b5b2f7591034638a98ea31276bd8f4e : SUCCESS in 31m 17s","accounts_in_message":[],"_revision_number":17},{"id":"fdc7688df72bfd2283cea2d3e15b4dadeb8b6e1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 18:24:32.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f95ee7be6d824c38a7e7b352a23b2070\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb412b7ee4234fba89a5a419fe247d25 : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/54ed032455994505b2dfa2b73cb46714 : SUCCESS in 1h 00m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/24a5dbfb026d4bfd81641fd7689cfb6d : SUCCESS in 47m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/745618b3e1174d7cbff552682ce05775 : SUCCESS in 1h 02m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e9c452c6fc24fb499c353abfebcf4d6 : SUCCESS in 1h 02m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a5a4e6bb7c4f4342aeff63f2efe0e99c : SUCCESS in 15m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b810a80404247548e83ea0f8a0307fb : SUCCESS in 21m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d959d719f014c838decb28165115c3e : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b4b8830475d74d249582ab8cd6a4c628 : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d39f452d5412468eba65393c8755f993 : SUCCESS in 17m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4917c23f5e9448f7ab6f0269afb9e11f : SUCCESS in 16m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3cb4ed9c5bdc486baf3c79d9d684d289 : SUCCESS in 29m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5228f6613f3b4a46bd865b80f9fd7745 : SUCCESS in 23m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5ccf2038bd054acdb87f0ea9a1426313 : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2baaf3ee6074424d9fe940eb848777c0 : SUCCESS in 14m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bd4e8e7c3cdf46b081e4ffa773999529 : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2547a90d96c44b88925a107aac3adeb : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1530b012836b4e5a8c88292d31762678 : SUCCESS in 8m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a3d6e07b3c824bcb9aa35d30a3934676 : SUCCESS in 6m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e81ebc99e494f81b3f8dbc27a287130 : SUCCESS in 37m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8340a5978bf749ecb691d4b3f90c2d1b : SUCCESS in 48m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/94c0a4094f27495eaf3b3a3548b036ec : SUCCESS in 51m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2f02b4be44249cea7f1542ecf8dd788 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e51566a043a24e0aa4ffd98afc809f66 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/50fc5286596f4204ba0e099ad3e83004 : SUCCESS in 19m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4acf8bc31f01436b993f6139811eb3c6 : SUCCESS in 1h 00m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c8eb5ee119742e5851f6d1b59eb292b : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":17},{"id":"254c9071998c517e7b9d57dfc82daec13cfbeb79","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-11 03:24:35.000000000","message":"Patch Set 17:\n\n(4 comments)","accounts_in_message":[],"_revision_number":17},{"id":"cdbb11b90ecba394e3eeec55df0f48163e5153ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-11 03:25:56.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"e3191f8888ad8f7e9c6f733d0baec105faa0dde1","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-11 03:27:47.000000000","message":"Patch Set 18: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"7d1adce8ec14078ab491460269aea5659e555a86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 04:19:25.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc6285fe3f1e4772831e6e4963cb63dc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d6efa35e395b482ea27b1ccc77fee0aa : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":18},{"id":"7d2781bc3a44d9a9bb434a9c8be9955711a825c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 04:37:07.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35cec20171cf4807a198b5e048d8a4e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99ce1bd0e9e4eaa8cf11d77c816b516 : SUCCESS in 6m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/9355f54cb6bd48c29b353ef4958fb5bb : SUCCESS in 59m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ecfeeb5da3ef42f086ca8e64341594bf : SUCCESS in 1h 04m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f0e56adae2b9492fb7ecc55ee3951676 : SUCCESS in 34m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ad1a2ad067b4ecdbc89c989f2bd71f1 : SUCCESS in 36m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec342a116f214c22ba626435f9a413bc : SUCCESS in 10m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9e960f8a3cb4782b2ea93fd94b8b9c6 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34daf762c56f4a1991996f339af062d8 : SUCCESS in 12m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d7cb71e75e234c59afd6d63cdedcbf60 : SUCCESS in 8m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c4b0cd1d20424106ba2985b457e61b80 : SUCCESS in 23m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8f08cafd49a041299f0ed0b359f9876e : SUCCESS in 8m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/28bf4acba336467da7eb07b199751f1f : SUCCESS in 15m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c592ae0afed242d5a358486e61db63ae : SUCCESS in 11m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fa681ab4e84f433ca60edb4067601df9 : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/03c4a466c93e4b6cbddb1e110ebe5a90 : SUCCESS in 15m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d4e0f9bd94c842b08abd3733fd31d0e2 : SUCCESS in 9m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7a85a29f74dc42908b1b2fc9fdb2eebd : SUCCESS in 6m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0bdf897323a1429e97a6fe685fbedd58 : SUCCESS in 13m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b1e3063002d74e35a0cbad5bb34be006 : SUCCESS in 5m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f454dea088ec4239b40f3ffbd6d958c8 : SUCCESS in 57m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4dd2599d10cc43cdbb4d7d40209e42d8 : SUCCESS in 35m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b331fbe35f96420d9c8b2da35d6fc786 : SUCCESS in 27m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5991eabb7c734a07b1c115029fba198e : SUCCESS in 27m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea1d64a7e9a0411f9bb7ae59940941d8 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/42a83cc4782140299c4391d88a314cd9 : SUCCESS in 17m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9644322717042f0a8bb422b3732628a : SUCCESS in 49m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d51a6f37e16b436f9bcd8401ff8ec4c1 : SUCCESS in 29m 27s","accounts_in_message":[],"_revision_number":18},{"id":"ac1c9680d8db1fc81ed41c73031d10e94ea260fd","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-13 15:42:04.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"3fdd2b40167fba94939522e31d391b044271634d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-13 16:52:13.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"94696ba10467de47d6534294b2bcb1c87150d0a1","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-13 17:26:23.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"86a9624fa2e633fef8b0179679a0e571584d23f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 17:55:43.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aec62b49824418d9712b9404f11c1c9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/96bc95d05f6d4ae19d1a0adfd4653fcf : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":19},{"id":"9a61ced672423ab0ca16333127ede93ac8147248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 18:04:03.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6e03411fc884504ba8147b103828e77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f614c8649664d98a25eb536d1a3df86 : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/edd7d8644d384904af34752cc40da20a : SUCCESS in 28m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7582b45ffb754136ac4125114baa87f8 : SUCCESS in 1h 03m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0d0c4587d4d34da880ac094b4c5343b9 : SUCCESS in 28m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3741ce67c104948bbee97c564d839a7 : SUCCESS in 1h 06m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63ea66ffc1404929b8a7afd864a67736 : SUCCESS in 20m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6336f060788145ad9930c20f28f6a701 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/529084eff7bb4e8a8bfe5277889440cb : SUCCESS in 20m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a2ce0d5545ce4f028940905f659e078d : SUCCESS in 4m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2b02c4d8aa7b43ed8f8932ca1a1d7c93 : SUCCESS in 27m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7afc9537e33f4300b925daf099f8f32a : SUCCESS in 21m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f7aad4ea336b4d0c9be7bff18d8f8f0d : SUCCESS in 34m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5e79f8477a684268861a4b28e7ab08d1 : SUCCESS in 30m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/979aa174a02243ee9a48dc0c5595a1a7 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3b6a6fa54fd64fdfb5cc7c3265742855 : SUCCESS in 9m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/00e349442e054420a3617819258bc045 : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03eeeccde8ed490083cf5751c006535c : SUCCESS in 7m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6e177f094ba44d638ea6e03e29f6e50d : SUCCESS in 9m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/31030bb3337d49af827a5fe260178b8a : SUCCESS in 8m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/50aa1261bd494d16b75c2024d643dbac : SUCCESS in 37m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e53b1e24052640f88f8e93ef3706cd8c : SUCCESS in 57m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/defb03b06ebc40309bc49525a8e0fcbb : SUCCESS in 38m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9378b1cfe35b4b1583bce915736fb18c : SUCCESS in 32m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb5e7d4dd6e04217b6246fa306cc6247 : SUCCESS in 3m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb710b3345524985ad56171be814e99b : SUCCESS in 17m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7373e416c5441428aae6e39163895a1 : SUCCESS in 46m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9967ff4e857a496f92578c8a2f360305 : SUCCESS in 46m 54s","accounts_in_message":[],"_revision_number":19},{"id":"c99e25185d780b7ebef8a12d01ddc20c68dbdadd","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-13 18:28:16.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"bccbdf2ce5cee5c0b8c5cd3638effd91de466c15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-13 20:16:12.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"8c9f2fa628f4c2aee58e7747da986b1221f27165","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-13 20:16:20.000000000","message":"Patch Set 19: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"70dd6685cf7201c1b4bff782f731383b8d86fbbc","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-13 20:24:41.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"e48aeae9acfe928de57b3314283f755ba4bafc6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 21:04:19.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dec13c5487184e8187875906e4cd4801\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94dea8d12da34e269d14448a10bd7ebd : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":20},{"id":"5032e5535a3c07e119f6636fff67cb7454756204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 21:21:06.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/188eac981a28473b9a5875cb7be52996\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d8c08b878af43009e4e6fc5a6409a81 : SUCCESS in 9m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/d38d9bb9ed7c4d4ab3aee29e6638a439 : FAILURE in 25m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cb77c3af6ad143aeaa8597bfea6fbec5 : SUCCESS in 45m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a14c764dff6e423cb0b6930f054ba88b : SUCCESS in 55m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3073a7c8cf7d4110b2a364e504d0f0e4 : SUCCESS in 43m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8ff9667757249d1a3d333cf6733e650 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aeab40c88b7e4ad9999a6764dafd8df1 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e8154368b2f4bfe83105be3656326f9 : SUCCESS in 10m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9bb613bf761041c78ed91ff38d31a726 : SUCCESS in 6m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/79c2fb30c1ef46c9b1df983f71971f77 : SUCCESS in 27m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/61add9e27e3d4d35a5629b1a13bf0cd1 : SUCCESS in 21m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ac0c3d9bedf64c99afdbb7f8f63c729c : SUCCESS in 20m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1105641b860e42b08cf7e588ef032211 : SUCCESS in 31m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4e1031913ee14f7395497aa0369157b2 : SUCCESS in 12m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/307193453f454e0aba7d4062f12b82fd : SUCCESS in 8m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b6954022b73546b68b322ad49ae34e20 : SUCCESS in 15m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dff6b3d9e8a74359b6f66fbd625f308a : SUCCESS in 5m 33s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3014100257634b9495adb4c20abf90b6 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1e9c73839880402b8dcddfafcd8258c6 : SUCCESS in 8m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/56123cba83e84c9d8fd2057b76feb732 : SUCCESS in 58m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/89ffab23eaa24112a30e85ef63ed5e9d : SUCCESS in 21m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9b5888a9c791419499c39e46a8d9c785 : SUCCESS in 32m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bd39bbcbcc94f44a41bb4bf4fcb8622 : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b223eae995048d48138f228397bdce0 : SUCCESS in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/27de26c69bd44c6db8c6b5dd067dad78 : SUCCESS in 19m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b6ad116960445dcb959ecec60499ed7 : SUCCESS in 48m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2a302585043413abbfdde87e71f107f : SUCCESS in 32m 41s","accounts_in_message":[],"_revision_number":20},{"id":"9bc59668d3dd7c2623fe9cad579a085fd5d2457e","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-13 23:32:46.000000000","message":"Patch Set 20:\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"9d4c406b1f7adfcad322cd7b585d646da19b4ab6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-13 23:37:44.000000000","message":"Uploaded patch set 21: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"762b3f755e6260d83bf135aae7e4e6fb76a62c02","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-13 23:38:56.000000000","message":"Patch Set 21: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"2988054ffadf3800b09f5da314989177ff0c794e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-13 23:52:12.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":22},{"id":"f0a0f2983b182ac7992b6175263ccb4c9cf3f235","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-13 23:52:41.000000000","message":"Patch Set 22: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"6c94eb218818a6b6f535283dcc3602a03f29e746","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-13 23:55:40.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"42b95dc1e201a39e9e7d1426551fdad4a13a9699","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 00:47:52.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41bbd930156447639f8a305c5e70206b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8a1e17314354cc7a2e5f15e90751d80 : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":22},{"id":"a4954c97a70482e8b38c20bb26a72fbe307a7fdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 01:05:29.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/581118092c21478ebd8d3920263f95dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f73b0169c2c43d39e258bf108da961f : SUCCESS in 10m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/5caca8494636425c90444a50cc173ee1 : SUCCESS in 1h 07m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/79b3b6d069ae4fdda8b73124f7b6cf7e : SUCCESS in 49m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa92a6e61c6f4de09c6aec4e60c5de5a : SUCCESS in 35m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/006532f88d0c4668b6c49a75d7a42ac1 : SUCCESS in 32m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0441726e7f8424683d3c61464966184 : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca9a83cb93a843ae855f812dfe90db84 : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8422e8234c94ddf9b3c9f6f27bbfc83 : SUCCESS in 20m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f4a45d4fa79040fc81865b69cd6b16bb : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d6375903d1904591b3e574aeddd86479 : SUCCESS in 24m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a7184d7e78ad4c17b97684dbb4e1347f : SUCCESS in 16m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/74df2fc006814437bc24d7a1b4ecd094 : SUCCESS in 19m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/401abb2332074609affdaa6e6e8cc0b3 : SUCCESS in 28m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/348cd64a4d04498fb96aedde7aa6a78b : SUCCESS in 11m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/78e79cde00cf43369fbe04b5a537aa1a : SUCCESS in 12m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b00a6cd4c7374480adc17872b06c4cfc : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1f235d8290244886a883703140237461 : SUCCESS in 6m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/866618ad5c484e568dd3f40a1630a78c : SUCCESS in 15m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2a8eef037aa4451ca56400eaa3318054 : SUCCESS in 7m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa245679048e49f6bf92b3faa8333b85 : SUCCESS in 53m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b30b58eeb073407fa792cb012ec938d5 : SUCCESS in 57m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0896c73c8e414ffabde57aa21e3f712b : SUCCESS in 22m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a4482b7c3de4e75bec21a7497f0955c : SUCCESS in 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0feea8ddd6248e1a7974914c1a0057b : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6c8c3ea395ca4910be9574a00fb4d8af : SUCCESS in 15m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65011ee3d5224020a62f609a6fcd78a2 : SUCCESS in 57m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bf08adf80674936b382690f1f949688 : SUCCESS in 1h 03m 14s","accounts_in_message":[],"_revision_number":22},{"id":"c7a100f680bc4c01b93d73bf4b64867077e9dfd5","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-28 00:31:34.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"2db0a6b4ae385e61570939c71c6b5551a34780fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-28 13:46:07.000000000","message":"Patch Set 23: Published edit on patch set 22.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"ab326393f1d283ad09f570350dda8e46b947116b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 13:46:14.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 962305,23","accounts_in_message":[],"_revision_number":23},{"id":"e80664f028631b875a225b95afbbd3a2f3f0ce8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 13:46:15.000000000","message":"Patch Set 23: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 962305,23","accounts_in_message":[],"_revision_number":23},{"id":"ff50f8f73aeb44ccd0a9127d3485b2018f7cc2ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-28 13:57:24.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"cd5eb8b03b6bcee7b5a9a011ca5378dccd6f0025","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 14:57:27.000000000","message":"Patch Set 24:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1250bdc3177f469bae2461e135d5a827\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e596a4f5ea1d482da4f0bc76cb0bdcb6 : FAILURE in 58m 12s","accounts_in_message":[],"_revision_number":24},{"id":"b56fdbef164d99830865aeb7c655ccfc99d67cf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-28 15:09:39.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"d16f146027ce704b70f011f54b074ffd4eaa6cce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 16:21:43.000000000","message":"Patch Set 25: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4f6c83823474f8e93f4bab8be2a88d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b17c376c1fe4db098689f4acb798851 : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/04a2d0fc2e394b1ebb0456d861575087 : SUCCESS in 28m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a3d2bec95ef040aeb11dea152d1701d8 : SUCCESS in 1h 01m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a852e8b0a98943f6afa1cc4d919de813 : SUCCESS in 56m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bc8f89975604add85dec6e5d619b74a : SUCCESS in 33m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1118c8125b094956ae609959e5c2e8bc : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1a1222f70e24180a1676250204f2885 : SUCCESS in 20m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e779ba8ac59542eca67e53309bfb8bf2 : SUCCESS in 19m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2032ddb1d6b745498eb69613f6705e1a : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7684bc058f794ff08bc409441590410f : FAILURE in 23m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ad2f516528394bbc863d1e61ef69e047 : FAILURE in 16m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/313c43499fd643bd83830ba08ed9f53f : FAILURE in 23m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/168995e1676f4818842954b30b046be4 : FAILURE in 33m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8efc44dcd0d74bfa8200efe51641ad5a : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f9e1026b91784943a78bd2a123e43f91 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bf78c109906148b89229696674807c23 : SUCCESS in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fa5fa77b97b74c5a819ef98696fc56bf : SUCCESS in 6m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9f0620adc1164f649764790013a0b2cf : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/57ebcd14ca9e475c828f4eefc122bb6c : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f57c4801be844bd9823ee699a122e28 : FAILURE in 59m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/40d69aaf5ccd4c6182d87202859bcf65 : SUCCESS in 51m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/412ef201290b4b1da36e605813088fe0 : SUCCESS in 54m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19da535c729c43b1b1500a60f58e564a : FAILURE in 27m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c278123844144149aa882c34ad84121 : FAILURE in 2m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d1aa2ea81bef4a119de9f4e5d52a1afb : SUCCESS in 21m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3795e835fcf46dd98b0184eed8688c5 : SUCCESS in 43m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b704bc32522430f994f8350f39e3aa7 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":25},{"id":"0d4cdf34d131a88511d3cdcbb803ffb5e6d4520f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 16:30:13.000000000","message":"Patch Set 25:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6019bfdca2942158b07cad6da9234e4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/585a97d1dabd4861b300578e8e27ab92 : FAILURE in 38m 37s","accounts_in_message":[],"_revision_number":25},{"id":"56cd87a1645838c14a611947abaa7b4abb767a58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-28 22:22:34.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"349a717a0c9f0687e8cb1b2c4a47ade848bb3342","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-28 22:31:31.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"ba788941248e8c8889116b1e37e09c32050d9593","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 23:06:00.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4179a1cb85664560af4943e530a038c2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4549dbfad7fe4868b6d84d45188acadb : SUCCESS in 28m 02s","accounts_in_message":[],"_revision_number":27},{"id":"0f9feb581c8c3835a75e6042525aa4b7ee9a5604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 23:55:13.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95a6824c08de4106afecd46f719f7adf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d986acd47aa43ad8b7b90517abc2f34 : SUCCESS in 10m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/a3d18348d113429eac930b528f7bc90c : SUCCESS in 58m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d760378f185e4ec182933ef5cec7fe2e : SUCCESS in 1h 00m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ed9b2d10cbcb455bb5ad52b6593c52a6 : SUCCESS in 59m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d65f3120b37448b58d74a4bd0cfa2eaa : SUCCESS in 1h 01m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a245f9e7e5042aaaa3dcfe8cd1119d6 : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba56f3de8cff44f3ab75576ae6ec653d : SUCCESS in 10m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ffa8fc171d64a04a19d9d1aaf0d7e0e : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/645d6452f87d4f00acc4372649a57c23 : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9403e216357045ff810e6032b8f21572 : SUCCESS in 22m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b4ca8244e06945d2a26d2861b0a8025a : SUCCESS in 15m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/695c0fc786a540898677c5f623e69eda : SUCCESS in 12m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/dc4e8e9321684c9fb58a9a61d4e53faf : SUCCESS in 11m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b6bb06a6e6904d2ebcade7ec66c7be74 : SUCCESS in 12m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3dcc140a61c84c0f900c592494dba218 : SUCCESS in 14m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5303ab7017034d25aead5d14a26e0c08 : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6ce21fd5e6424b698d9a9ceff181b991 : SUCCESS in 6m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d3cc7c3918024b7f98c33d7ac1da5c9d : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9a48dc630592453e8073aa05e4fad9e0 : SUCCESS in 7m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7be388dfaebc4b46981a6c26f3c59d24 : SUCCESS in 59m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5568a87e1643442fb6bb953618dcf29f : SUCCESS in 22m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7683888c594842a5a09ab5d562407e8d : SUCCESS in 21m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/343ca29841b54e59ba77024806354bf8 : SUCCESS in 26m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/823c9b0f0f3e466db25d26f2a6f217c5 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f759b4b39784cbea08afc4ef277aed7 : SUCCESS in 16m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5447be6bf474bb48e9b395220abfff0 : SUCCESS in 56m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54af429a04054a1a80bca311bc0f0464 : SUCCESS in 55m 50s","accounts_in_message":[],"_revision_number":27},{"id":"d4598392dbe0c57a5c6acf8cfe71d9904555a110","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-30 04:40:28.000000000","message":"Patch Set 27: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":27},{"id":"8f9b3cadea6648f02697d10fe666f764cfaa5d56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-30 18:11:40.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"99afb3966a560f4267b9b305c24229a34a6b923d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:27:19.000000000","message":"Patch Set 28:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe9f0e93a1f746c4934f3731f78f4c83\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/21039c35203147238a32e1d58e9b21f5 : FAILURE in 7m 57s","accounts_in_message":[],"_revision_number":28},{"id":"39f5225551e7f2584cd54a29684f2de03bf76b63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-30 18:28:05.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"abaf043c62eb588a7c296613f377d50c9ef1e8b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:58:30.000000000","message":"Patch Set 29:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a915ab1e2ba4fd3b4eb669cbfedd749\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fa8ea83dbd1e439bb4dca9266da3d939 : FAILURE in 11m 58s","accounts_in_message":[],"_revision_number":29},{"id":"37de2a5152ad0de61f02d0e222bff2a33cd36624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 19:31:29.000000000","message":"Patch Set 29: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/604e9e3ce8534183a97aa7fa63438fc2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c98e123461a495385adb4d564ccabcf : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/f3b1b71d67604532999cd3d1afe48bfa : SUCCESS in 1h 00m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aa8a46a9f6b04f1ea6fa5a9e364cbeab : SUCCESS in 35m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2f2c5d494fc54c95bf45bca1c9c38296 : SUCCESS in 35m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9416123c24254ef0b423715e72bb6a78 : SUCCESS in 58m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/296ac99742aa45e9ba2300b97ba42399 : SUCCESS in 14m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ec83add1c284e54aad4ccf62daf09a7 : SUCCESS in 15m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/071b34103ae64b5aa43b0a1e74eabead : SUCCESS in 18m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/74688438d8a84a0c8d001507187786b3 : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6ced1ece88a14c5189b36e1cb316dacf : FAILURE in 4m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0cc47db8c2da42dcbf5e709fd1784e0f : FAILURE in 4m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/428237c5f0774351ad87f2b9f6d1b216 : FAILURE in 10m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c4fd0c2e4f07440f987350246353134c : FAILURE in 4m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/41f84b95f021475e88b3c8746c440cbc : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/36d85d0521dd428eba50098ab1dbcb38 : SUCCESS in 11m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/87e2fac0fe3f47569a0c98f784570515 : SUCCESS in 13m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e3e82e9e5d0a48fbb979e7801f5c7597 : SUCCESS in 10m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61203ea95f914664ae8417866c88d880 : SUCCESS in 15m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3a7472a1fe3442b49ecc8860883fef2b : SUCCESS in 8m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c138a1fd3994d62b90aa0a0ce8499db : SUCCESS in 35m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/236a76c3ffda42299833be293998cc74 : SUCCESS in 21m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e87a5a6b4e8f4f86b9ea15e38a5300c3 : SUCCESS in 48m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98e351a82542465f97b84d2c6d9523c0 : FAILURE in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5549596d02db425d97994ddd04a08865 : FAILURE in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c889ff26516c4175a03947173b33402c : SUCCESS in 18m 59s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/341f2621f8b94fae8d4dde12f82fd90e : SUCCESS in 47m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9022652bbcaa45a7b311be56e02adef9 : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":29},{"id":"ebb7962fa08a9c6ecc4e582badec5f9fd7835dd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-30 21:40:00.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"ef499bcbe91738dbcb626a3ee7e17f91db2a8353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-30 21:43:22.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"514d9fac8b1edee4341be02823ef33e838aa50f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 22:37:37.000000000","message":"Patch Set 31:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12391a7b7098490daf890a8b5d286baf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de18a21288e943efa2380a3ae4069600 : FAILURE in 51m 43s","accounts_in_message":[],"_revision_number":31},{"id":"d9e20b259371751e38283fb33ecf6522a1ca34dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 22:46:02.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/240cf10e238e4004b1c151bc338d4309\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9423439dddfb405fab0768a66865a340 : SUCCESS in 10m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/bf7d5c0754b54b6c8b088c49f1799fcf : SUCCESS in 42m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f81c1e70e7584e9a815764dd1f21f7a3 : SUCCESS in 28m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/baccc7db8ad94bb1a24d9964c4d6fcbd : SUCCESS in 55m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef2aa2257600436881a41db45cf7f914 : SUCCESS in 46m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8de74f5e9f94a51b31c91d0f370c17d : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/773d3a36b72749ae9499857ab4e1ceda : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2bc93c1a44f5402280016b04fc3a8530 : SUCCESS in 19m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0426892574bb414d96c96c9dd3aa475a : SUCCESS in 6m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3a7c12029dbc4969bb47ba17a618e518 : FAILURE in 16m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/58b1eb022df54f2d937b41ec8761abce : FAILURE in 23m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1d75a3b35a044042b51522f5bccba79a : FAILURE in 17m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8b481932592b4086906d5e5505358095 : FAILURE in 29m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/92912b2d82134374a0adcbefd10d35e6 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/691233a4be43452e97a5fc1e178bd8e7 : SUCCESS in 16m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3d417187d62d4cefb764130c55ae3ca4 : SUCCESS in 14m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ba969f5dc96b4a1f96215fb3fc9bbaab : SUCCESS in 6m 48s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5a569cd0149d49db836d12b25d50a062 : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ca11f0076c0e4da798785d83ea4d3059 : SUCCESS in 7m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/616b26adfae04e04b24968c56cab366b : SUCCESS in 57m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/23f6122fc7f448609bd712be80dfecfc : SUCCESS in 51m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c205ddfbf10c4ed8819e0fb6bdfd1848 : SUCCESS in 49m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf77abc008534977897f71a5f7381dc3 : FAILURE in 30m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05aebc8f840b4aaeb4f393b7b9513f02 : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a5fed32ff894033bc80e777b3432c68 : SUCCESS in 18m 33s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b559a6f0453a47d9af499eb9d2d02b7d : SUCCESS in 48m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfae3862f9d4489b8294f5cf70f0c720 : SUCCESS in 48m 27s","accounts_in_message":[],"_revision_number":31},{"id":"f1f1da8a99ad65a596320ea730b2cc64ba785673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-31 00:57:21.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":32},{"id":"111da013829678b1c34a56c60b93f5c6196ce4c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-31 01:29:30.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"992851b698c2daea130ba2968abb8fb50f02cff5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 01:42:27.000000000","message":"Patch Set 33:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2be0245bfc3547c2b4414b56c4469ab9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23f231d423be4af4a85734c611fa8f69 : FAILURE in 11m 10s","accounts_in_message":[],"_revision_number":33},{"id":"acfedfc4bfd3b8fa2a58f132feec3c3724f73512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-31 02:12:58.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"91f8dd71212b128bd69787c5b5e86a8561df941e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 03:32:29.000000000","message":"Patch Set 34:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/491a793cf2c047ce8b24d82cc8525c76\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a8353f61e6340e3add74edf9a0678df : FAILURE in 35m 18s","accounts_in_message":[],"_revision_number":34},{"id":"a05a2b50439a06038056923d6d589c1c504b26eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 03:38:01.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37b82bf95a814231bb4b2238198f216a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc97df7947bb411da63a15da05a36776 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/8d8de9a82ca24a2aaa2e88ee7a6e38f2 : SUCCESS in 50m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eacfd78dfab04234bcb763dcbc10df06 : SUCCESS in 55m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/771162d09ae04d50aba91703ffbe5793 : SUCCESS in 1h 00m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f33f6eed115249dd8295810722c2c3d4 : SUCCESS in 1h 02m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48d4fafc54624423b740e07eb88cd9ea : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44e0f3ad96ea460b9ece8c7a0675d725 : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7336718fa6574d9799d63ca275f3d8e5 : SUCCESS in 20m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3b98076bb7a4c65a511eee5e2c1a26a : SUCCESS in 8m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/48b4b06ca7bd4c6aba2eb2d04af9bb74 : FAILURE in 24m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c1021396d5274400bc7473c7ce193fe7 : FAILURE in 10m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7c6f7671e16a4ed496f5be93f1344a67 : FAILURE in 28m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8f1759c67a1c40e49cebf17d00172d10 : FAILURE in 18m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/de80257d50ee451a9a692eb22c2a08a2 : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c6e62898422742859ecce4fcb85493db : SUCCESS in 15m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/facf196838284667b563456d0378a530 : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c13d57698fb430e97d84d7324864ff6 : SUCCESS in 4m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/44308b0c5fa74ce8ba8821dcf711b88b : SUCCESS in 10m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/97f1af7bd8f6476099a56fbf3f42858c : SUCCESS in 11m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e3a8e4389aa468ca5167fe5a42d877a : FAILURE in 33m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d0f9ad891c14f3f846b1ed4cfe3e0cd : SUCCESS in 45m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/387cc975cb1746dd8d9611af3b0b5708 : SUCCESS in 56m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a568b7af3c9742f180e272cfd629ebdc : FAILURE in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63d96782153c4dec985626ba518196b4 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30ab3c4002ba47c79acd14b4c09b39ec : SUCCESS in 15m 39s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e6397189bdc4beba060a6ea390f90ef : SUCCESS in 45m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/363358f813bd4a8f828d6472445a95e7 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":34},{"id":"cb815783d28c6df02a52d5c8293ceb50ebd8eead","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-31 15:44:55.000000000","message":"Patch Set 35: Published edit on patch set 34.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"bd60c91ea1ead0d0b41e46a57f681b2bc932ad4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 17:10:37.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21018120f3db4c74b871fac35f2df988\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8fd9ecc7ba944e795633efd86b69c48 : SUCCESS in 13m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/a84151ac90ac47868f212eef8fd10cf4 : SUCCESS in 38m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d2ef805839f427d8df7f874d3faca96 : SUCCESS in 1h 09m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2f04fe0996854faba8b634bb14e82c0e : SUCCESS in 32m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a6ffd3d01364a79aabb3e12ef251c80 : SUCCESS in 34m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a442373a4e8458bacf9237dd9c1662b : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/860c1b5d0cd64e0aa105a26f219a2ea6 : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80817448327a4b3fad150fa096e545a2 : SUCCESS in 11m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5ee423919cf44f01bdd4b7e3a0478ba8 : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6fd2868961c84982945d24b34a1bd103 : SUCCESS in 13m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/52b0ff70f12a4df3a149d4701700ea1c : SUCCESS in 24m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/078f433381c046749d006435aa6ff3f4 : SUCCESS in 11m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5803b208c07a4a9c9447d6488273d8ad : SUCCESS in 31m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/232a64d791154120868d9eb4776cd73e : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ba888f4e0f6a4c958d132b86fe19d305 : SUCCESS in 12m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/30b421c076694ea0851a505930ab5b25 : SUCCESS in 14m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4f6e18ece33a49b2bcb529b64b6131b9 : SUCCESS in 7m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/22d48d720c714d2b9b27a0fd49f095aa : SUCCESS in 16m 41s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1a59f01f054a4460bcd4f933e1f55f72 : SUCCESS in 5m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d4d7cf64d7447e09a0c4f07b69b04a7 : SUCCESS in 36m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef9a0fd6d014402b902cbc95549ec0e6 : SUCCESS in 51m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0f70344bf6d04f95af91effb0bafd297 : SUCCESS in 54m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbf595beccf147e68a5d8f884b2e7bbd : SUCCESS in 24m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65216635418a4c5fbf95bc0511f9ec64 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/556f1637ae29436393b10c2446cad3d1 : SUCCESS in 10m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b0045513c3d4a2996b004d2860f5ab0 : SUCCESS in 35m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c86a302a704e4222a70e60745eaf5035 : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":35},{"id":"7c3d561806905eadb5161c09ae2217d11f310463","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 17:52:44.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a0d2161a88340cd8d5b4f403942705b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5dae994d7512414b87d27a7686c723da : SUCCESS in 42m 11s","accounts_in_message":[],"_revision_number":35},{"id":"241bdcf6c5a939dc425b336b477f0a6bbe7c0c80","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-11-03 15:55:54.000000000","message":"Patch Set 35:\n\n(2 comments)","accounts_in_message":[],"_revision_number":35},{"id":"773fa0ef4a9ef9aa90c0af269c7d329625962e32","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-11-03 16:51:01.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"d21fe549fc36a36b3ed1bd374ed0649a12e71599","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-06 23:02:57.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"5405511c78851695bb41167a737b6b32a2bce875","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 23:37:00.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dec00fabddd74865879fe36606ebb818\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5172b6d84a5f489fbf6583cd28935c56 : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":36},{"id":"596fe4774e8126be4c3ac42eb955c7c946460731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 00:23:33.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dd6d644765f443ab21a13f1205cf7a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d959f7487f48dd9ff92070f34dce72 : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/bb661d06db054e5ba561f8a90c585b2a : SUCCESS in 55m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b7a29f9abbb47b49104deff99350f1c : SUCCESS in 1h 00m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fc9e6e4e30d341268eb1d843ff61c1af : SUCCESS in 57m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/847c792523da4ba583f937008c04336f : SUCCESS in 1h 14m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47c26532a64d4cf2847d04ec0da50b1f : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f375bb77008459e84012c06a772fb15 : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3dfc9e5dc6734082b10d72350d85367a : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8451d58c27674225846b5a20d4bbeb35 : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9d6116f87edb4c9da0397fa51cac6fef : SUCCESS in 24m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/55ce7596f6144b058a43e4958fb7465a : SUCCESS in 21m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4384be628b6241c0a8b0a965b849c85b : SUCCESS in 26m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a3127c32626246a1b4e3bebdd8224feb : SUCCESS in 19m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5f3e6d0134d340e784e064df9d844bf3 : SUCCESS in 12m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2e1bca60b6de468f8046d9ebd0e6ec40 : SUCCESS in 15m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7c8f38e8542c418987f25cf214817357 : SUCCESS in 14m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fb1d3e5be170454ab771b7adc07778e5 : SUCCESS in 4m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b5c3b6ecdbbd43efb0bbd87a9316a8d6 : SUCCESS in 13m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/99d900a0171543e8aa95c4e6f393ca11 : SUCCESS in 5m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e9a8e7c4aeb4fb6802a257ef2e85e61 : SUCCESS in 57m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b1743f1e1b854a36b13cb2b4b67df7f1 : SUCCESS in 39m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e0af043edb144afa4b03d0ff8235be4 : SUCCESS in 47m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89a33cb772384febb4724f21b8340c7b : SUCCESS in 31m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cb1ebac51614f6caf101ff4d2c06988 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6953374c24b5452f9c8e7ba6eb89c1d8 : SUCCESS in 18m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f60c9f98301441759ba0d1225e8f8bc1 : SUCCESS in 50m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0d87701fdb746f785cc4bcfbe2b4fe2 : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":36},{"id":"aac2216623d546babc6fe263ae68746a2724b641","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-11-14 22:07:42.000000000","message":"Patch Set 36:\n\n(5 comments)","accounts_in_message":[],"_revision_number":36},{"id":"f14dfd162c3f1855c29780e55e51f904412c8bdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-20 18:23:14.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":37},{"id":"640226f6fb860dbe97e294ecfb8b2f68b7d3bc18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 19:38:58.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5de15a3ed3f84785a66ffc51aeab6bb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b26956c2b44bd9bbd769b5553299b1 : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/b793f7a515ff4141818e0e603304fb90 : SUCCESS in 1h 02m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f120023dcf924001a6f423a9dd6aa9d6 : SUCCESS in 30m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e0f23edf7394e18baf459f8fbc60548 : SUCCESS in 53m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f30dc5423f514c7ea11f1b7174c7c4cd : SUCCESS in 1h 04m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d15edbacba44b7da634a068e89aa041 : SUCCESS in 12m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7afa764dea874be8b47e20c2bbc8a9aa : SUCCESS in 20m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7768e0066ea245c59b046b50e76de0f9 : SUCCESS in 21m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/143fc386491b4ac887d0e91deab0fced : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3ce6ef85b82942a5afc24ec8b8dfb77c : SUCCESS in 20m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cd5aa847a298425bb171a948bea84613 : SUCCESS in 15m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3038c80d355344c48d850ab023c0e3aa : SUCCESS in 28m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2e977555b6c6499d9c17c6494af51373 : SUCCESS in 23m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/34884d7914e74baead9be931ecf755fa : SUCCESS in 8m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/454b8d7d04a34a6b918f628ad5dd7b82 : SUCCESS in 8m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b91b479f4b8c4ede95d2b3ebc2b1ece5 : SUCCESS in 13m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ee91a456b0c48f18979e09a27f72a37 : SUCCESS in 6m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/865d0607719f4db48188f96d2cd43463 : SUCCESS in 13m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/030217b85e9e4ab8a105519251aeaa4e : SUCCESS in 9m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad35511f3d264a91a65278683b48cdc4 : SUCCESS in 34m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9dbf667345c341f59040675851e9fada : SUCCESS in 47m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b8f724b04274d3d9569a158f6023019 : SUCCESS in 47m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7365a033f6e648b68dd7dbbcaa45025c : SUCCESS in 30m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7148df77a974bf09c39e3f6baadaeef : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1e31f8d344347ec9db3a3287252f66a : SUCCESS in 9m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffaa42a2f1a24ebf9d4d05811a73bf3a : SUCCESS in 54m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/547e5076458742bebc96a6b48b9f5cef : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":37},{"id":"393a43cadb823e6f0eb8ee18e67078e2c1b18089","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 20:08:15.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32738711623d4a9f90e8cd798be49a8b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d059a6fd6dc049498276525e04553a5a : SUCCESS in 1h 41m 40s","accounts_in_message":[],"_revision_number":37},{"id":"0c5f68f6301ee8e8679dee85a1d6d6666ea42153","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-20 23:32:58.000000000","message":"Patch Set 37: Code-Review+1","accounts_in_message":[],"_revision_number":37},{"id":"c0d9e8dbf61056d8fc9057c2cf2fd4c0928dcde8","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-11-21 16:55:44.000000000","message":"Patch Set 37: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":37},{"id":"6b462658f3f90dadb2ce61ad92b7754396d2c3a3","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-11-21 18:20:20.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"46f83856ee4b19153fcbcfb10425b756e03713b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-22 04:24:49.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"22cfa42147a7597e0177f204835cb6e7eedf1210","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-22 04:38:07.000000000","message":"Patch Set 39: Published edit on patch set 38.","accounts_in_message":[],"_revision_number":39},{"id":"748702cc38f4689694e821c68f519ff60a669a3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 04:38:36.000000000","message":"Patch Set 39:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/438792f47885477cb5e3204e2d63bd0a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/70fc89660d2b41ea830b96fec2f6848f : NODE_FAILURE Node(set) request 684eeaba5c1848c3b3a8390fdd257aed failed in 0s","accounts_in_message":[],"_revision_number":39},{"id":"984de4b07842339df67a9a21707a4a44137a42eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-22 04:38:43.000000000","message":"Patch Set 40: Published edit on patch set 39.","accounts_in_message":[],"_revision_number":40},{"id":"72121120243f773ec234dc19d4d57a24509faa0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-22 04:39:08.000000000","message":"Patch Set 41: Commit message was updated.","accounts_in_message":[],"_revision_number":41},{"id":"87200bbfb95cd6e67a62331146bc22303d3e83e1","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-22 04:41:23.000000000","message":"Patch Set 41: Code-Review+1\n\n(9 comments)","accounts_in_message":[],"_revision_number":41},{"id":"6d8b9dfbd45258fec85369a85c6d2d85ad518a0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 05:45:26.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f34f27805fb041eeaf974d47365685f9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0549d618a54743f6936b8e376b61bfc1 : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":41},{"id":"92e24c66b677714b4f31829e6875d63b5d1f4ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 06:21:03.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e7453124b454e3cbf2db9aa637f9b40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41cf563337e4d1d848a5bdb7f13c035 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/92894016713e4f18a9708061b2d8642b : SUCCESS in 59m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b912ee5fe6144c64ad40770cab15c7c7 : SUCCESS in 55m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ff671ed61377483d96ce56776af7b808 : SUCCESS in 31m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e101a17bda8495682047a1cec2cf90c : SUCCESS in 1h 01m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6863064d68cd4aec9c0e8c35d1ad71ab : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a4a37ab80374dd290bfefbdb22771e1 : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa7de95219c84c49a2efe0a8e896699e : SUCCESS in 19m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4552e38baa7b4b508c46d1a242e06c70 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4cbeaf93aca4953992f464a7649fceb : SUCCESS in 24m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/50f48e86a4894af78946c411d5897016 : SUCCESS in 10m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fab02b92c4ec46a0a3cdf98e2ecdba13 : SUCCESS in 27m 30s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a3144c690ecb41ec9e4f693700173274 : SUCCESS in 18m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6372e5c8b74f43ec85713179eb77b6d2 : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4701113699db4384a6b3e94cedfef61b : SUCCESS in 8m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/63c2d37cff0d497d8b3a7dd9c4ac74fb : SUCCESS in 9m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/69910a6b72d1489dbc8f923aa9403deb : SUCCESS in 4m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a4c84a39ac8a4d59b39a897484220c55 : SUCCESS in 13m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0c151c4a4e664783827365acc4267bc3 : SUCCESS in 8m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fe0934007854496a3a8a64ebe3330f5 : SUCCESS in 37m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c22182e5318402eaa58a44362738360 : SUCCESS in 23m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/072131eb99894891be4ab53dc90762ca : SUCCESS in 55m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86400dc62f164718a628a0d6002c47c3 : SUCCESS in 23m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2632e18c9c7b4efd9cd63bcd91f365fc : SUCCESS in 2m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b714cb97813e4086acd62a83b31941de : SUCCESS in 19m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/686e08241e0243c78d25693e739c052f : SUCCESS in 31m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cea4ff4ee2f484082eab2aedbdb28a9 : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":41},{"id":"40489ec4ee018f034423987459f7e71b79817269","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-11-26 03:20:03.000000000","message":"Patch Set 41: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":41},{"id":"b96b8df6caa2b4beb5fc996569da0c1a28549f65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-26 21:20:18.000000000","message":"Patch Set 42: Patch Set 41 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":42},{"id":"6bc14808faf30eb915d59103095330a88f8fc15e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-26 21:43:06.000000000","message":"Uploaded patch set 43: New patch set was added with same tree, parent tree, and commit message as Patch Set 42.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":43},{"id":"9e943308fb60017a8017cf064c92abe0ea06ad93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 22:48:43.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/760ab2cec1dd40049b1eaefd996c4e1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6b38041fc743499aab51a49e12e533 : SUCCESS in 5m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/04c9a29e8e0e4a3a90bdc9de8dee21fc : SUCCESS in 48m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/04f87068c92440e6a6c9edf8d752fa10 : SUCCESS in 46m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/16317c53f9ad4500ae9e218dd2b7fef2 : SUCCESS in 42m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3820dab8c024a369e5280b0ed9f82d5 : SUCCESS in 1h 00m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b4fb2de12af45ac9c1874bc53dcffa4 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7519d06ac4842c9a989e86f15140ca4 : SUCCESS in 12m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14d0dae756a549f1bcb7309731f4b563 : SUCCESS in 22m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f02c1e38108345119dc9b891f3ed3e48 : SUCCESS in 8m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ae9abac1e64446fabaf0e48604908735 : SUCCESS in 28m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9375ac54693f4a9596690420fd435c72 : SUCCESS in 21m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ec22b895051443e287ec37529c45a5ce : SUCCESS in 20m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e504e739690544cba26d7aad749c82ef : SUCCESS in 13m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/da251d535b26470ab3512a222421e4b4 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4a16f075a91b4fa0bfe2357f2d5e5803 : SUCCESS in 14m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e666b55e8a704a8ca9f2d68b6f5450fe : SUCCESS in 9m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d7a11255ef8c4f4f8d27a6d1acaead4b : SUCCESS in 6m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3eec617e261d481d92f9d0f0fb58cccc : SUCCESS in 8m 16s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e26d22a4fb6240d491a6ee0317353508 : SUCCESS in 8m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c48ac119d5d44dffa71da166ea0223e3 : SUCCESS in 36m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/783578022358444bbabb6a95ae0ce5d3 : SUCCESS in 51m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd53b0eb1be940b9a3ab4b42923ae8d7 : SUCCESS in 49m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8f8c50bbe404cc9990caff27f201197 : SUCCESS in 29m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab27d560aa6549a1a005fff20b91a27f : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0548ea9aaf3460dacf28eb10373e82b : SUCCESS in 15m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b2aad9037b94c94a210aac3040e8ae2 : SUCCESS in 42m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91dd68218bd3456c9053f617a2cfb2e2 : SUCCESS in 32m 01s","accounts_in_message":[],"_revision_number":43},{"id":"941cef9fe1d630316558aa4e240088669c30eb73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 23:44:11.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1429f5235764d3884e01a70b858233b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fb7ff217ff97433481801839bfffdb72 : SUCCESS in 32m 51s","accounts_in_message":[],"_revision_number":43},{"id":"032f681be93b8ec03a408edf8bd6378ff7196da8","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-11-27 00:14:56.000000000","message":"Patch Set 43: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":43},{"id":"dff54e61cb68f2419e37b56fc7a2e330e9478b10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-12-04 19:55:30.000000000","message":"Uploaded patch set 44: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"6199050bbcde57fac27c7a805a8b8e9fde649904","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-12-04 19:57:41.000000000","message":"Patch Set 44: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":44},{"id":"463500761360a2ac374c6167089d4294bc61d1ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 20:53:39.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e63287e846264124b8ce30d0a4efdc33\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c38f50e10c4439383b8ab9b4a494ed5 : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":44},{"id":"f41301b8f14b9f5b1e9d02f7b88669e0899d93ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 21:09:47.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb73efd3350c4c9c95b8077c934371cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0eddbf557e8436b87635d6b68626632 : SUCCESS in 11m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/cb410a0c90df46b2a53852034b28eac3 : SUCCESS in 59m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d99cd1778c554f57bb3d73a7cc349281 : SUCCESS in 1h 02m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/438da6c4bd05405e8c6626f20f31f466 : SUCCESS in 33m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b41b1dfac7e4415db79f8686170cf90e : SUCCESS in 32m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e128fba83924deeaeaf8ff4718d73ec : SUCCESS in 10m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81b03458efd9422d949e18d50b9a2b88 : SUCCESS in 11m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7721d79e3e434373a9c4675e550d99e4 : SUCCESS in 11m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/098ef7e91772482186ff90d0c6b006f9 : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/13b3a1a40dcb47fc80068bd371332495 : SUCCESS in 24m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/95f5ad07c2a84b46ac68827f973caf69 : SUCCESS in 20m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e7531eafb6a0486e957b77c0100192a2 : SUCCESS in 33m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e248252105d8436ba1e903b4317241db : SUCCESS in 11m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fff39544c04c4dd8a54210082b78db34 : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e73e91eb03384d0a8d7fc474b48e9874 : SUCCESS in 8m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4dc8c7a8e73948c7a9f877e2c921e8d6 : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/43713f75bd03402195b3341f01175909 : SUCCESS in 6m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/32e939b8e0884146acdcd6c4b4402e30 : SUCCESS in 14m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/13d75786a12e4cab9bbcdb0fe19c7a67 : SUCCESS in 8m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b989bad44e2e4a17a34311a68769eda2 : SUCCESS in 55m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e037b93c6acc455f85c9976a03d3a61c : SUCCESS in 22m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b7f87e518df4e6faa3ca5141c894205 : SUCCESS in 21m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9034996806324bed96368cc2fa6634ae : SUCCESS in 28m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50a8a3bffb70426295a2e40d8740a966 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cdb42f6bda3343b99fcd2f258a13a7f9 : SUCCESS in 19m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79ee88de4b3246b9940e5172b80ec49e : SUCCESS in 56m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50b96a90b2794beeb662c57b4a562324 : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":44},{"id":"789fed2185a3e22df0e3add9a8248f3b54fc6881","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 21:10:16.000000000","message":"Patch Set 44: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":44},{"id":"d8083e42b7344dd4e67c7a0fe22ae5ecfbf6389c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 23:03:32.000000000","message":"Patch Set 44: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/210f2245cb324bd48fe322d256334651\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59100e163f5a4154a0dfcf283f9049e1 : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/fdb798de79374fe6b2cb072eba14c502 : SUCCESS in 1h 08m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f7573d6c95db48a6bee4f6f9ed0873f8 : SUCCESS in 1h 02m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/98b0b2b50db843c2bf44f78abfff874d : SUCCESS in 58m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65fef83a08ee4c09814f363861330ba2 : SUCCESS in 1h 00m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/82cc7c6844314fc89b28219037233fb3 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b2c1dfc9678d4483b97821990a6b3c98 : SUCCESS in 21m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2a348040afb94ae889ee05be15eb5105 : SUCCESS in 18m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6af9f3334b2e420f83dff3f5a6fd345e : SUCCESS in 29m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/40dd8d8400ad49789804ef13318548bd : SUCCESS in 12m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/999ba8a5e0264c1ab2e8f0c416254358 : SUCCESS in 8m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d766b49d1aef45ca8a6b6b599cf5fcb8 : SUCCESS in 12m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/95ef59823ab54fa490bb99dd60a83679 : SUCCESS in 8m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/148c135555ec4b9987c49751add20482 : SUCCESS in 5m 50s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/211e64bd79fb429593179365eb41fd99 : SUCCESS in 4m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c5e01c670924aac97f9e72fd6c3f736 : SUCCESS in 33m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/77e992cf55f54be681a31601aa2a399a : SUCCESS in 37m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9892467d97f5409c9fa2c3fe3143c472 : SUCCESS in 50m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c3f8dd2f35647ed948d38f1bd5d17c3 : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6667c41b2e79483fa3fa683ec93e621e : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5edba3a4d7a443e9b9e886f8ac933e80 : SUCCESS in 18m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc3a2fc70aee4416a05a207874725203 : SUCCESS in 49m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27e90fbf529e4a01ae828b77ef6a6161 : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":44},{"id":"b173b0a58e35db5aa8837523c2ded4397b744c10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 23:03:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":44},{"id":"bfcdbce11da908439cc5d8d2a7be4a04af685456","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 23:04:22.000000000","message":"Patch Set 44:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbe62c7869784f92938de8d19a259e04\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/600bd45bb9d14a81999e29082e6024f6 : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e844809714e445b8ac944fb025a46680 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":44}],"current_revision_number":44,"current_revision":"9babfefade23df1c9c3d562dcea8e21372516bb0","revisions":{"61dc95ae948f2aedcdded7c14fbf26f372aa2992":{"kind":"REWORK","_number":1,"created":"2025-09-25 19:07:08.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/1"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-25 19:06:59.000000000","tz":-420},"subject":"Adds cycle time tracking to sharder recon logs","message":"Adds cycle time tracking to sharder recon logs\n\nMetrics include a min/max/mean/median/p99 metric summary and info about the slowest/fastest 10 db files processed in a sharder cycle.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61dc95ae948f2aedcdded7c14fbf26f372aa2992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61dc95ae948f2aedcdded7c14fbf26f372aa2992"}]},"branch":"refs/heads/master"},"3901e40569b2481596cbb5403cec4eb1e0ba863d":{"kind":"REWORK","_number":2,"created":"2025-09-29 18:43:24.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/2"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-29 18:43:15.000000000","tz":-420},"subject":"Adds cycle time tracking to sharder recon logs","message":"Adds cycle time tracking to sharder recon logs\n\nMetrics include a min/max/mean/median/p99 metric summary and info about the slowest/fastest 10 db files processed in a sharder cycle.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3901e40569b2481596cbb5403cec4eb1e0ba863d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3901e40569b2481596cbb5403cec4eb1e0ba863d"}]},"branch":"refs/heads/master"},"e157d9c3933f0d5d857cf69a088923684e7230b7":{"kind":"REWORK","_number":3,"created":"2025-09-29 19:05:17.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/3"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-29 19:05:10.000000000","tz":-420},"subject":"Adds cycle time tracking to sharder recon logs","message":"Adds cycle time tracking to sharder recon logs\n\nMetrics include a min/max/mean/median/p99 metric summary and info about the slowest/fastest 10 db files processed in a sharder cycle.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e157d9c3933f0d5d857cf69a088923684e7230b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e157d9c3933f0d5d857cf69a088923684e7230b7"}]},"branch":"refs/heads/master"},"883cf7249279bb3073f92f4fc101a6eb8367e19f":{"kind":"REWORK","_number":4,"created":"2025-10-02 18:36:06.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/4"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-02 18:32:46.000000000","tz":-420},"subject":"Adds processing time tracking to sharder recon logs","message":"Adds processing time tracking to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/883cf7249279bb3073f92f4fc101a6eb8367e19f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/883cf7249279bb3073f92f4fc101a6eb8367e19f"}]},"branch":"refs/heads/master"},"5a9ef6d033045aadd2f04610c43525ba678c4b67":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-10-02 18:36:39.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/5"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-02 18:36:20.000000000","tz":-420},"subject":"Adds processing time metric to sharder recon logs","message":"Adds processing time metric to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a9ef6d033045aadd2f04610c43525ba678c4b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a9ef6d033045aadd2f04610c43525ba678c4b67"}]},"branch":"refs/heads/master"},"c6fc85702c748918c93c89d8b687503351a62a0b":{"kind":"REWORK","_number":6,"created":"2025-10-02 21:49:38.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/6"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-02 21:49:24.000000000","tz":-420},"subject":"Adds processing time metric to sharder recon logs","message":"Adds processing time metric to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6fc85702c748918c93c89d8b687503351a62a0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6fc85702c748918c93c89d8b687503351a62a0b"}]},"branch":"refs/heads/master"},"f098f0578214156ec13f3157948817c122c69ada":{"kind":"REWORK","_number":7,"created":"2025-10-02 21:50:40.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/7"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-02 21:50:33.000000000","tz":-420},"subject":"Adds processing time metric to sharder recon logs","message":"Adds processing time metric to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f098f0578214156ec13f3157948817c122c69ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f098f0578214156ec13f3157948817c122c69ada"}]},"branch":"refs/heads/master"},"935458bcbaae5cda1020f6874f6de5d91e331a4d":{"kind":"REWORK","_number":8,"created":"2025-10-02 23:26:13.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/8"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-02 23:26:05.000000000","tz":-420},"subject":"Adds processing time metric to sharder recon logs","message":"Adds processing time metric to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/935458bcbaae5cda1020f6874f6de5d91e331a4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/935458bcbaae5cda1020f6874f6de5d91e331a4d"}]},"branch":"refs/heads/master"},"4c3cb2b1939911027549419fd09cd4276728051b":{"kind":"REWORK","_number":9,"created":"2025-10-03 16:42:39.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/9"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-03 16:42:30.000000000","tz":-420},"subject":"Adds processing time metric to sharder recon logs","message":"Adds processing time metric to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c3cb2b1939911027549419fd09cd4276728051b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c3cb2b1939911027549419fd09cd4276728051b"}]},"branch":"refs/heads/master"},"02a51f22c7343c9832dd264b7cb2ab6316794824":{"kind":"REWORK","_number":10,"created":"2025-10-06 22:33:54.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/10"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-06 22:33:42.000000000","tz":-420},"subject":"Adds processing time metric to sharder recon logs","message":"Adds processing time metric to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02a51f22c7343c9832dd264b7cb2ab6316794824"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02a51f22c7343c9832dd264b7cb2ab6316794824"}]},"branch":"refs/heads/master"},"5965077cebe46a8d178708544ae10e1ab20dc4e3":{"kind":"REWORK","_number":11,"created":"2025-10-07 18:10:41.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/11"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-07 18:10:35.000000000","tz":-420},"subject":"Adds processing time metric to sharder recon logs","message":"Adds processing time metric to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5965077cebe46a8d178708544ae10e1ab20dc4e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5965077cebe46a8d178708544ae10e1ab20dc4e3"}]},"branch":"refs/heads/master"},"32369bedf2eb93e2fbb57c067d4a4cf06c143207":{"kind":"REWORK","_number":12,"created":"2025-10-08 15:14:38.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/12"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-08 15:14:31.000000000","tz":-420},"subject":"Adds processing time metric to sharder recon logs","message":"Adds processing time metric to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32369bedf2eb93e2fbb57c067d4a4cf06c143207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32369bedf2eb93e2fbb57c067d4a4cf06c143207"}]},"branch":"refs/heads/master"},"1c972d5b905295507c5c62b9d144756d582d25bd":{"kind":"REWORK","_number":13,"created":"2025-10-08 18:18:07.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/13"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-08 18:17:54.000000000","tz":-420},"subject":"Adds processing time metric to sharder recon logs","message":"Adds processing time metric to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c972d5b905295507c5c62b9d144756d582d25bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c972d5b905295507c5c62b9d144756d582d25bd"}]},"branch":"refs/heads/master"},"6bce68942cf419a058ad2b2a99fb0faf6616f500":{"kind":"REWORK","_number":14,"created":"2025-10-08 18:20:34.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/14"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-08 18:20:24.000000000","tz":-420},"subject":"Adds processing time metric to sharder recon logs","message":"Adds processing time metric to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bce68942cf419a058ad2b2a99fb0faf6616f500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bce68942cf419a058ad2b2a99fb0faf6616f500"}]},"branch":"refs/heads/master"},"d11fdde7055b7fe533d0469e7d792d0db34cd7d8":{"kind":"REWORK","_number":15,"created":"2025-10-08 21:04:25.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/15"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-08 21:04:02.000000000","tz":-420},"subject":"Adds processing time metric to sharder recon logs","message":"Adds processing time metric to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d11fdde7055b7fe533d0469e7d792d0db34cd7d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d11fdde7055b7fe533d0469e7d792d0db34cd7d8"}]},"branch":"refs/heads/master"},"a67a64d0de25e6c42ae7d957da2db2f66996b4e5":{"kind":"REWORK","_number":16,"created":"2025-10-08 22:38:16.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/16"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-08 22:38:04.000000000","tz":-420},"subject":"Adds processing time metric to sharder recon logs","message":"Adds processing time metric to sharder recon logs\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a67a64d0de25e6c42ae7d957da2db2f66996b4e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a67a64d0de25e6c42ae7d957da2db2f66996b4e5"}]},"branch":"refs/heads/master"},"6e6e5f81bb0e215234ffeaf94a84ea831ccfd643":{"kind":"REWORK","_number":17,"created":"2025-10-10 17:17:03.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/17"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-10 17:13:50.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\ntotal_replicate_calls: Number of replication operations during cleaving\ntotal_replicate_time: Cumulative time spent on replication\nprocessing_time: Time to process a single broker\ndb_sharding_total_elapsed: Total time since sharding started tombstones:\nnumber of tombstones in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e6e5f81bb0e215234ffeaf94a84ea831ccfd643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e6e5f81bb0e215234ffeaf94a84ea831ccfd643"}]},"branch":"refs/heads/master"},"117cd99c31fb220ebc02f1440bef15724b4a9ff3":{"kind":"REWORK","_number":18,"created":"2025-10-11 03:25:56.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/05/962305/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/18"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-10-10 21:45:43.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n                           cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n                           cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/117cd99c31fb220ebc02f1440bef15724b4a9ff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/117cd99c31fb220ebc02f1440bef15724b4a9ff3"}]},"branch":"refs/heads/master"},"f49bd133afd4d6d1bef370b0927a8c4744c63d9e":{"kind":"REWORK","_number":19,"created":"2025-10-13 16:52:13.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/19"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-13 16:52:03.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\ntotal_replicate_calls: Number of replication operations during cleaving\ntotal_replicate_time: Cumulative time spent on replication\nprocessing_time: Time to process a single broker\ndb_sharding_total_elapsed: Total time since sharding started tombstones:\nnumber of tombstones in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f49bd133afd4d6d1bef370b0927a8c4744c63d9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f49bd133afd4d6d1bef370b0927a8c4744c63d9e"}]},"branch":"refs/heads/master"},"e80a21041291b999d7cb3b8fccfc4ead4acda82f":{"kind":"REWORK","_number":20,"created":"2025-10-13 20:16:12.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/20"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-13 20:15:11.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\ntotal_replicate_calls: Number of replication operations during cleaving\ntotal_replicate_time: Cumulative time spent on replication\nprocessing_time: Time to process a single broker\ndb_sharding_total_elapsed: Total time since sharding started tombstones:\nnumber of tombstones in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e80a21041291b999d7cb3b8fccfc4ead4acda82f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e80a21041291b999d7cb3b8fccfc4ead4acda82f"}]},"branch":"refs/heads/master"},"5ed86abe4c157f9146d342cea19ebcb4f9b74c8c":{"kind":"NO_CODE_CHANGE","_number":21,"created":"2025-10-13 23:37:44.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/05/962305/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/21"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-10-13 23:33:35.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n                           cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n                          cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n                          timestamp in recon file  will keep increasing\n                          even after sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ed86abe4c157f9146d342cea19ebcb4f9b74c8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ed86abe4c157f9146d342cea19ebcb4f9b74c8c"}]},"branch":"refs/heads/master"},"2aa162f9d07d9ff72322b889aa9ca2a69e7e80aa":{"kind":"REWORK","_number":22,"created":"2025-10-13 23:52:12.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/05/962305/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/22"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-10-13 23:51:17.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2aa162f9d07d9ff72322b889aa9ca2a69e7e80aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2aa162f9d07d9ff72322b889aa9ca2a69e7e80aa"}]},"branch":"refs/heads/master"},"ec9df28eda3f45667d8e4b579334c96e39f0559e":{"kind":"REWORK","_number":23,"created":"2025-10-28 13:46:07.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/23"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-28 13:46:07.000000000","tz":0},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec9df28eda3f45667d8e4b579334c96e39f0559e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec9df28eda3f45667d8e4b579334c96e39f0559e"}]},"branch":"refs/heads/master"},"b2d36877d98948288df744b9e9e0adbc5c1602c9":{"kind":"REWORK","_number":24,"created":"2025-10-28 13:57:24.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/24"}}},"commit":{"parents":[{"commit":"8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8","subject":"Update resource type in labels.html","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-28 13:43:01.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2d36877d98948288df744b9e9e0adbc5c1602c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2d36877d98948288df744b9e9e0adbc5c1602c9"}]},"branch":"refs/heads/master"},"c58c841aebb980ffcc50263e0397d199a2f54b76":{"kind":"REWORK","_number":25,"created":"2025-10-28 15:09:39.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/25"}}},"commit":{"parents":[{"commit":"8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8","subject":"Update resource type in labels.html","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-28 15:09:31.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c58c841aebb980ffcc50263e0397d199a2f54b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c58c841aebb980ffcc50263e0397d199a2f54b76"}]},"branch":"refs/heads/master"},"c2e2ab14f96641c19d351bab47952587c130e6f3":{"kind":"REWORK","_number":26,"created":"2025-10-28 22:22:34.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/26"}}},"commit":{"parents":[{"commit":"8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8","subject":"Update resource type in labels.html","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-28 22:22:26.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2e2ab14f96641c19d351bab47952587c130e6f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2e2ab14f96641c19d351bab47952587c130e6f3"}]},"branch":"refs/heads/master"},"021d3673378b6a1beb5c0e3e0f0415fe42280ef9":{"kind":"REWORK","_number":27,"created":"2025-10-28 22:31:31.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/27"}}},"commit":{"parents":[{"commit":"8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8","subject":"Update resource type in labels.html","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-28 22:31:25.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/021d3673378b6a1beb5c0e3e0f0415fe42280ef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/021d3673378b6a1beb5c0e3e0f0415fe42280ef9"}]},"branch":"refs/heads/master"},"a18fce79d8077ef6ab54ab6989e75ea2823e14a5":{"kind":"REWORK","_number":28,"created":"2025-10-30 18:11:40.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/28"}}},"commit":{"parents":[{"commit":"8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8","subject":"Update resource type in labels.html","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-30 16:58:00.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a18fce79d8077ef6ab54ab6989e75ea2823e14a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a18fce79d8077ef6ab54ab6989e75ea2823e14a5"}]},"branch":"refs/heads/master"},"bd4386544f7f34e627d86ad365f33a0fc1869a89":{"kind":"REWORK","_number":29,"created":"2025-10-30 18:28:05.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/29"}}},"commit":{"parents":[{"commit":"8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8","subject":"Update resource type in labels.html","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-30 18:28:05.000000000","tz":0},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd4386544f7f34e627d86ad365f33a0fc1869a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd4386544f7f34e627d86ad365f33a0fc1869a89"}]},"branch":"refs/heads/master"},"b227fa29fe783b30f7c5f55c553938e24afb3a80":{"kind":"REWORK","_number":30,"created":"2025-10-30 21:40:00.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/30"}}},"commit":{"parents":[{"commit":"8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8","subject":"Update resource type in labels.html","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-30 21:28:51.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b227fa29fe783b30f7c5f55c553938e24afb3a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b227fa29fe783b30f7c5f55c553938e24afb3a80"}]},"branch":"refs/heads/master"},"c7a46741c988b78bba68c8101b9b8771ebaacd9b":{"kind":"REWORK","_number":31,"created":"2025-10-30 21:43:22.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/31"}}},"commit":{"parents":[{"commit":"8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8","subject":"Update resource type in labels.html","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-30 21:43:22.000000000","tz":0},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7a46741c988b78bba68c8101b9b8771ebaacd9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7a46741c988b78bba68c8101b9b8771ebaacd9b"}]},"branch":"refs/heads/master"},"ac515bded8ec618cade80fa3dc71f11123e79a09":{"kind":"REWORK","_number":32,"created":"2025-10-31 00:57:21.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/32"}}},"commit":{"parents":[{"commit":"8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8","subject":"Update resource type in labels.html","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-31 00:49:04.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac515bded8ec618cade80fa3dc71f11123e79a09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac515bded8ec618cade80fa3dc71f11123e79a09"}]},"branch":"refs/heads/master"},"5be5ee9b9f07a0bffca5f42707d1a97530182c6b":{"kind":"REWORK","_number":33,"created":"2025-10-31 01:29:30.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/33"}}},"commit":{"parents":[{"commit":"8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8","subject":"Update resource type in labels.html","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-31 01:24:14.000000000","tz":-420},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5be5ee9b9f07a0bffca5f42707d1a97530182c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5be5ee9b9f07a0bffca5f42707d1a97530182c6b"}]},"branch":"refs/heads/master"},"8f4cab321e1b1566a52e7952c88a1ca6dda08926":{"kind":"REWORK","_number":34,"created":"2025-10-31 02:12:58.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/34"}}},"commit":{"parents":[{"commit":"8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8","subject":"Update resource type in labels.html","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-31 02:12:58.000000000","tz":0},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f4cab321e1b1566a52e7952c88a1ca6dda08926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f4cab321e1b1566a52e7952c88a1ca6dda08926"}]},"branch":"refs/heads/master"},"565917f6126148c43814f4b5bd7a9615d20a14d3":{"kind":"REWORK","_number":35,"created":"2025-10-31 15:44:55.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/35"}}},"commit":{"parents":[{"commit":"8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8","subject":"Update resource type in labels.html","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cbe10552a66be4ba5b271c5bebecdc60d9a9cc8"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-10-31 15:42:41.000000000","tz":0},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/565917f6126148c43814f4b5bd7a9615d20a14d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/565917f6126148c43814f4b5bd7a9615d20a14d3"}]},"branch":"refs/heads/master"},"1a336b9be2d83bd9be65d718045bd6f9cae7e533":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2025-11-06 23:02:57.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/36"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-06 22:42:51.000000000","tz":-480},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a336b9be2d83bd9be65d718045bd6f9cae7e533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a336b9be2d83bd9be65d718045bd6f9cae7e533"}]},"branch":"refs/heads/master"},"886ac32c0e21a226ac6e0b7757540fadc2659a90":{"kind":"REWORK","_number":37,"created":"2025-11-20 18:23:14.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/37"}}},"commit":{"parents":[{"commit":"676e68fc0f563f8971cb97e0b862518fbb08d7c3","subject":"Merge \"trivial: tests: add failure context to flakey assertion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/676e68fc0f563f8971cb97e0b862518fbb08d7c3"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-20 17:40:36.000000000","tz":-480},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/886ac32c0e21a226ac6e0b7757540fadc2659a90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/886ac32c0e21a226ac6e0b7757540fadc2659a90"}]},"branch":"refs/heads/master"},"be0cdb01877849a9723a9ed4a2982a82c1a5d234":{"kind":"REWORK","_number":38,"created":"2025-11-22 04:24:49.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/38"}}},"commit":{"parents":[{"commit":"676e68fc0f563f8971cb97e0b862518fbb08d7c3","subject":"Merge \"trivial: tests: add failure context to flakey assertion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/676e68fc0f563f8971cb97e0b862518fbb08d7c3"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-22 01:52:26.000000000","tz":-480},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be0cdb01877849a9723a9ed4a2982a82c1a5d234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be0cdb01877849a9723a9ed4a2982a82c1a5d234"}]},"branch":"refs/heads/master"},"a12b8d25fd9e194db1670eaf3eb1177b66635eb1":{"kind":"REWORK","_number":39,"created":"2025-11-22 04:38:07.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/39"}}},"commit":{"parents":[{"commit":"676e68fc0f563f8971cb97e0b862518fbb08d7c3","subject":"Merge \"trivial: tests: add failure context to flakey assertion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/676e68fc0f563f8971cb97e0b862518fbb08d7c3"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-22 04:38:07.000000000","tz":0},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a12b8d25fd9e194db1670eaf3eb1177b66635eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a12b8d25fd9e194db1670eaf3eb1177b66635eb1"}]},"branch":"refs/heads/master"},"e7f4ebecee3322c777385cf17c77562b3403260a":{"kind":"REWORK","_number":40,"created":"2025-11-22 04:38:43.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/40"}}},"commit":{"parents":[{"commit":"676e68fc0f563f8971cb97e0b862518fbb08d7c3","subject":"Merge \"trivial: tests: add failure context to flakey assertion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/676e68fc0f563f8971cb97e0b862518fbb08d7c3"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-22 04:38:43.000000000","tz":0},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_calls: Number of replication operations during\n        cleaving after it completes.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    db_sharding_total_elapsed: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7f4ebecee3322c777385cf17c77562b3403260a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7f4ebecee3322c777385cf17c77562b3403260a"}]},"branch":"refs/heads/master"},"a596fb4ba103b93992553b5d8805773af6c6a3c6":{"kind":"NO_CODE_CHANGE","_number":41,"created":"2025-11-22 04:39:08.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/41"}}},"commit":{"parents":[{"commit":"676e68fc0f563f8971cb97e0b862518fbb08d7c3","subject":"Merge \"trivial: tests: add failure context to flakey assertion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/676e68fc0f563f8971cb97e0b862518fbb08d7c3"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-22 04:39:08.000000000","tz":0},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    total_sharding_time: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a596fb4ba103b93992553b5d8805773af6c6a3c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a596fb4ba103b93992553b5d8805773af6c6a3c6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0946729e3d820b8c60a856d09f6be98ee10a88d2":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2025-11-26 21:20:18.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/42"}}},"commit":{"parents":[{"commit":"f324fb4ad3f20b1c3fbfd838c782a1726e1bf7f7","subject":"Merge \"tests: pass Timestamps to date_header_format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f324fb4ad3f20b1c3fbfd838c782a1726e1bf7f7"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-26 21:20:18.000000000","tz":0},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    total_sharding_time: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0946729e3d820b8c60a856d09f6be98ee10a88d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0946729e3d820b8c60a856d09f6be98ee10a88d2"}]},"branch":"refs/heads/master","description":"Rebase"},"6ff6944debc2d5ba8aba57ec484617655a5a897c":{"kind":"NO_CHANGE","_number":43,"created":"2025-11-26 21:43:06.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/05/962305/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/43"}}},"commit":{"parents":[{"commit":"f324fb4ad3f20b1c3fbfd838c782a1726e1bf7f7","subject":"Merge \"tests: pass Timestamps to date_header_format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f324fb4ad3f20b1c3fbfd838c782a1726e1bf7f7"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-26 21:42:45.000000000","tz":-480},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    total_sharding_time: Total time since sharding started, this\n        timestamp in recon file  will keep increasing even after\n        sharding completes.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ff6944debc2d5ba8aba57ec484617655a5a897c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ff6944debc2d5ba8aba57ec484617655a5a897c"}]},"branch":"refs/heads/master"},"9babfefade23df1c9c3d562dcea8e21372516bb0":{"kind":"NO_CODE_CHANGE","_number":44,"created":"2025-12-04 19:55:30.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/05/962305/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/962305/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/962305/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/962305/44"}}},"commit":{"parents":[{"commit":"f324fb4ad3f20b1c3fbfd838c782a1726e1bf7f7","subject":"Merge \"tests: pass Timestamps to date_header_format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f324fb4ad3f20b1c3fbfd838c782a1726e1bf7f7"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-24 17:52:27.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-12-04 19:54:00.000000000","tz":-480},"subject":"Enhances sharding_in_progress metrics in recon log","message":"Enhances sharding_in_progress metrics in recon log\n\nThis adds some new time.time() logs around _process_broker that allow us\nto see the length of time processing a db took. The information gets\nadded to the in_progress_shard range for the corresponding db.\n\nAlso included are some changes to the CleavingContext to enable tracking\nthe number of replicate calls + time spent running those replicate\ncalls.\n\nUpgradeImpact: below new recon metrics have been added into\n``sharding_in_progress`` section of sharder recon.\n    total_replicate_time: Cumulative time spent on replication during\n        cleaving after it completes.\n    processing_time: Time spent to process a single sharding broker.\n    total_sharding_time: Total time sharder spent on sharding from\n                         start to finish.\n    tombstones: number of tombstones container has in total.\n\nChange-Id: I4ad1e22fd011652d80269a76353e5fc14db249f5\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9babfefade23df1c9c3d562dcea8e21372516bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9babfefade23df1c9c3d562dcea8e21372516bb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
