)]}'
{"id":"openstack%2Fswift~962315","triplet_id":"openstack%2Fswift~master~If56d8a527656eef1d74e24023574af57f74dda30","project":"openstack/swift","branch":"master","topic":"sq-coop-cache","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-09-28 20:13:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If56d8a527656eef1d74e24023574af57f74dda30","subject":"sq? some notes from review","status":"ABANDONED","created":"2025-09-25 21:58:56.000000000","updated":"2025-09-28 20:13:02.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2765853febfd44d3519c127f3cbc0028a7c177dd","_number":962315,"virtual_id_number":962315,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-25 23:00:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2025-09-28 20:12:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-25 22:58:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-25 23:00:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-28 20:12:38.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"de802ee84d68d2df62ab889ee066bb960980fcfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-25 21:58:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d56a17696eea9ea2a6b0a3e1b4c609e3810c75a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 22:58:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/249c88936c0b46a0afc3ea44df7757e8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0fc98b8a26c846759822e9d15e7e465c : FAILURE in 57m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ad29c0cde8c4ab6512ead3da3b156a6b49b9cf38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 23:00:56.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e025fca5e57494b90c2bfbe2f403590\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/493cd98096374e4f8fbb7be6a209a674 : SUCCESS in 5m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/def3d765f7eb4f288bca098acba87e24 : SUCCESS in 48m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/98ba3471e287438284eb8133540e7b46 : SUCCESS in 50m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9284f734f4fc4902bcbc49050c99469f : SUCCESS in 31m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c50487c41f694bbd983894d438002ff6 : SUCCESS in 31m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d998f0af45584bfe9087ce9b5dff97b1 : SUCCESS in 10m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4795cbc1ae441059b1e6746029e1251 : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3971b96117634d799c20906a4b6fbee2 : SUCCESS in 17m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/59143898168343ec8e57fc11b2a76a41 : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/212f0930351a4be49ef4eff297f3f7d1 : FAILURE in 16m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ab46cb4717d34091a868cc83e343cd24 : FAILURE in 16m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6f38f07eb851404ca6c3256052f5bc5d : FAILURE in 18m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fa15d249921848da903d02c39bb12371 : FAILURE in 11m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9e8cd53450aa4146b7146b5427e26cb0 : SUCCESS in 11m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b30ed92c765f42daa9393fef66272ce0 : SUCCESS in 13m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bce5c768e0774e2da253e520338b1bb1 : SUCCESS in 12m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ed53f412a4824284916fd6135bf8945a : SUCCESS in 5m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8dfc1392b3cc4f42b0d25dfac74d6d01 : SUCCESS in 14m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3177be8b4d7b4711bed0c3e50bc6a59d : SUCCESS in 7m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2d13ed58b974deba1bac1be0de7bbbd : SUCCESS in 46m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/be00607fb4684478841d5c4ac8d33663 : SUCCESS in 55m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f73b3f171297425cba04a6682342d5e6 : SUCCESS in 37m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da052801b70649528df4e172f8427795 : FAILURE in 21m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00a88b2b8ccc4c4ab865201a4a5062d7 : FAILURE in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a1c1a90e7f44daaa58ea03adb3bb81c : SUCCESS in 20m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c008c07dbd9e4499b0ac7d34778e99e1 : SUCCESS in 37m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/158cdf63196f4dbdb436b365f8065d33 : SUCCESS in 52m 36s","accounts_in_message":[],"_revision_number":1},{"id":"7ffd78e889753eb971c3e29a4414e0f79ba592a1","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-09-28 20:12:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2765853febfd44d3519c127f3cbc0028a7c177dd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-09-28 20:13:02.000000000","message":"Abandoned\n\nChanges will be squashed into https://review.opendev.org/c/openstack/swift/+/908969","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51cae607bd8e283f89ae2b10e313240a0cdde04c","revisions":{"51cae607bd8e283f89ae2b10e313240a0cdde04c":{"kind":"REWORK","_number":1,"created":"2025-09-25 21:58:56.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/15/962315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/962315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/962315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/962315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/962315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/962315/1"}}},"commit":{"parents":[{"commit":"e96c9149410bf1d3d08c152bcaeb9f9af89d67ef","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e96c9149410bf1d3d08c152bcaeb9f9af89d67ef"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-25 21:07:35.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-25 21:08:17.000000000","tz":-300},"subject":"sq? some notes from review","message":"sq? some notes from review\n\nChange-Id: If56d8a527656eef1d74e24023574af57f74dda30\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51cae607bd8e283f89ae2b10e313240a0cdde04c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51cae607bd8e283f89ae2b10e313240a0cdde04c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
